)]}'
{"id":"openstack%2Fkolla~950948","triplet_id":"openstack%2Fkolla~master~I562ce5e54752015863a20bf113bce74a69e02331","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-27 13:09:27.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-31 00:23:37.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-28 14:27:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-05-31 00:23:37.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-31 00:23:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I562ce5e54752015863a20bf113bce74a69e02331","subject":"[mariadb] Fix incremental backup using correct base dir","status":"MERGED","created":"2025-05-27 07:04:48.000000000","updated":"2025-05-31 00:24:31.000000000","submitted":"2025-05-31 00:23:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950948","meta_rev_id":"2de0106f10e2ff4ce3966e35e37bc2c5cd78c042","_number":950948,"virtual_id_number":950948,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-31 00:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-30 20:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-30 20:29:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 07:12:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-27 10:35:06.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-27 11:05:10.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-27 11:05:10.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-27 11:10:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 11:37:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b4402f114a192140029227d08f1470ceaae93a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 07:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2943c7aa27007905bb20a2ebec1c499ddac02ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 08:39:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"438e8ad52c612eb5e6df7944afd2720a7e7ef70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 08:53:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e63eaf8d180d178e27a6bf8d9c6ad4f196244484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 09:00:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec8cfef8de98b114a68586b966d53ff86be12232","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 09:23:17.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"51780ad15008493c0018be98c359de5cdfa65bca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-27 10:35:06.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"bacf132844fe78ec7c214e7fcc9aea9661dddf42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 11:10:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b89659693a44b459e363d1b08622723\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d3bb1c1eea0b43e9a8c76886e3494df0 : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/daf9c9fc1c6a4a09b5e3415f02299ad9 : FAILURE in 2h 08m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd953589975b4846a1b71082346e9bf7 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a15f89e6c5f479ab5b7fba4c598be21 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fb166dcade54f22a444dc3eb1a08160 : SUCCESS in 7m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e4ff926174d4a3085e27685f2d0a54e : SUCCESS in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c4a1e4e08ac810ceebd4e8d3f49929f9d2b83401","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 11:31:02.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"30c682309212fb96be8f83f252650790a565e6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 11:37:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/994604bff37f428aaa954a94e14ea3fb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/55344ad6e97046cd8bf7ab2ecd25fb3a : SUCCESS in 16m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2fb609ca298b4bca8bb7ccaaa47922a3 : SUCCESS in 25m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5468503478cf44a689bdabf212af3db6 : SUCCESS in 1h 06m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/19996f84cba143e185da240c00dcb365 : FAILURE in 1h 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ef654441ba4a1e86813218dccf39e7 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6da3a305e5494c27b897de40160373af : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4df14f99ff53481d9b9ab2ebc86b3442 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/416c9e4a982e4970b36e44ff48c53151 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed63498c610d48b6b3c9940e71a2ac58 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7eb4ba0c87d42aa832907f13cbe1e3e : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a4e5609829a4f42bce7e293b20ade0a : SUCCESS in 4m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24eaa4f3340d4cb68579b5e661ad9e11 : SUCCESS in 2m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e66c6e193a44325b432f31d32589158 : SUCCESS in 24m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f88638cd57ac4d89951f5e05f4807130 : SUCCESS in 24m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/00a4b626c9f141929ede3517f6741336 : SUCCESS in 1h 05m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cb279e0c321c46eeb6745beea848bfc2 : SUCCESS in 1h 23m 11s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a22cb47a109148428869e438db905b1d : FAILURE in 24m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3e112820edcc4c47be67e04f50e4ca5f : FAILURE in 2h 29m 54s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a7dba39ec4a2411cb62389838d0c7560 : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":4},{"id":"1a5c56d9958f056122e8f6ea8feb6d0c483cb623","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 11:50:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e20d6ca3c961a539ad0890f8e9a8b807e2e6c4f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:04:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c56f7f81fca44ba95d6ad8db2e0bfa3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2cfa4603899c46b6914520d82b176063 : SUCCESS in 37m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/da3488d7d79442988b3d2a90518cb5dc : SUCCESS in 1h 13m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5446d0d92bb2476497336c7104ac5460 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53ee2cf02fb740d3bfdf54732c4f7c82 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62cd19e75c0942d3a4fad736f7489c16 : SUCCESS in 8m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/94641cf558a7459dbcdd311c2709bb51 : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e7e04f28df9b2fdc9226c9e48f429ff6ca6a29c7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 13:09:27.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"138966cb8befa77e8b724f52e4b256eb2506003b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:17:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b0e5eb072c34314a1779003a09a4463\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/034ba723b669456b85df1b145112f73d : SUCCESS in 16m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/034c10a6727a442d958d51fbe509a63c : SUCCESS in 27m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/83f3fb804ebb45028a21f5d89ffbf3d9 : SUCCESS in 59m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9b8a89a654bd4dfd9eca926d16356d92 : FAILURE in 1h 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c649abdab8b545b6ad3b8a4bd0f1f48a : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/830f7a844b6a4f60afae2c07ee2048d0 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa4ee384e544414f8b668fb435bd0724 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bd37f8868674cd283981255b0f17792 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/051c8d7c37a14c2b91d8777c7a1268fb : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d24dde64fc40248a304cd8e765a974 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f64613c8b3fe4d579bd8c64c2fdd7944 : SUCCESS in 2m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e407a82bb5114e338a173446350832db : SUCCESS in 3m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/daebda3760b3434e8b4ae17baeb58e6e : SUCCESS in 24m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c2b004ec915d450e863c5cc0b490348d : SUCCESS in 26m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a29e09a9558b44dc8b66e009628fbc69 : SUCCESS in 1h 03m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/80266879562e4378b5166e60120654dd : SUCCESS in 1h 45m 04s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8bcfc46d76084be8960e220c11056bd7 : SUCCESS in 2h 02m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5ccb89b89c794c32a9229a6e96ae5965 : FAILURE in 2h 23m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/55828ccfd14a4e8aad753f25f76f37c9 : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":4},{"id":"a82890b712c1d26b3ded27b5941b96bf4d777a2b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-28 13:42:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"57ceaa4b787d0e8d4d3769a38632ee416cfbd15a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-28 14:27:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ec81fce2d3c5316c57d441d2754f3a08c5e6fc3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 05:04:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be90218342ed42c12e0c7c9a6d7190e53ebca0bc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 05:49:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e45e1d7c32ac0aaa4a43e99e69d3408fc010505","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 06:47:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13ab2ac47f224df69361578e470ebe4f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7226c05d042a45ecba9348c2994fda41 : SUCCESS in 34m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/04e071d6177a4a679ac3ff760ab3e49a : SUCCESS in 1h 40m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/712eee437bc84005b99f0ed1c0ca3355 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f19a09627cc547a8a1c07abf74118a15 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3e186f2055fd4c178fe5b01ba06670e0 : SUCCESS in 6m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8e83983c11b84531851e31288aee16fd : SUCCESS in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"075f8d2e046898254c765ba19feeece57b8a9a40","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-29 07:14:27.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52cd6d0b43f50108e7dec3d64631a78f0e5a60b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 08:14:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32027ac78b744b058823302f44e379ee\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c6df1c6b84d449d81ba10e9287b4ef1 : SUCCESS in 18m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/11a134cdca56415fa33ed96184cc2970 : SUCCESS in 30m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6f13425fec7d4cbebcb89079502fb7ee : SUCCESS in 1h 03m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/462a262bff274613b0c30fa677a8fea5 : FAILURE in 1h 33m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4308b1c17265462dbf5ddb0a86e67e5c : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/944e27c2a56842a2b2b933a09a53c7aa : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f44fee5347344d5da0ef79471b1136c6 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5519b6f03a634c2ebf5bffee3c8009b8 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0196f6cae8ed4f0ebb2c73337fab3ce2 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd26fbc982e4a538e1b15627c9e817a : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37cc136ca1de4da4b5ee4cd941995f60 : SUCCESS in 3m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4eee02f87606451c975c0df67ed6d0e6 : SUCCESS in 3m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d77518e21e664399800bb990e61d0b2f : FAILURE in 43m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9e291f5a64c74e08aa14d7d83c68587d : SUCCESS in 40m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/97086c0c9e7c4c3cb1705096d4df8401 : SUCCESS in 1h 14m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0b39e57951f343c0a9d82bda85d75a5a : SUCCESS in 1h 53m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f6629035e3964704a9d31a4f0679f365 : SUCCESS in 2h 21m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/648f4d355dcc4e9c8c9de3d7b15dd5b0 : FAILURE in 2h 49m 42s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/fd9d9950c16c439d8722c16595b99bb9 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":4},{"id":"3a55428a45c4294a33265420c95ad7a10ca20c16","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 12:17:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e102a5babb33ae75e73c91d5157ae22cb2081868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 14:12:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/875547bf03264b39972fc3becbf56cff\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/98b954adc1b44112897d67cb68314338 : SUCCESS in 15m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3eb2724813d74480b91c7958360e26be : SUCCESS in 28m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3d0e7868b1f44a3a92f3c2238cfe4e6b : SUCCESS in 1h 00m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ca8e4a4232ee444d94c5e9d48a652c2c : FAILURE in 1h 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c5be8ed1474587952fd09433e5c0fd : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf5ac3f014ef49acbd730155f1b2f7ce : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7650488d82424b31946c0a63c64847d3 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2a99e92e2e0451a9c081fa60d1d042f : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83c7f1b1a5074359875517ee80be118a : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c3a4fb89e5b46f79d2fabae4bf934cc : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4af98e90bedc4a81a8ac37fd83d6d748 : SUCCESS in 3m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1ef23d826504477bb663daa2cebaef83 : SUCCESS in 2m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a9de01c62d694b67a63b23c9b5e939da : FAILURE in 37m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a7610280cbd04e3dbbf9d0b9d3e3c9f6 : SUCCESS in 23m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0ddbc4514760442691581747deed5bd7 : SUCCESS in 57m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1cdf20be4a5d417c8bb2e8a72e8457f4 : SUCCESS in 1h 36m 37s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e6d8112f48544650843a116390d54d80 : SUCCESS in 1h 52m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c986febaae6546f18fdeb40d9d231b3f : FAILURE in 1h 24m 31s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/db07e9db7d0b4e698bbefb021ada5141 : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":4},{"id":"2cd64e6c39ddc04e0989d94d5137ca4bb5542ee3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 14:14:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a3eb632b3f4072e93d7202557026b5a633d4bec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 14:41:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d283921714b848299e364ce1ff659bc0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3f6bc77a9cd04e3fb0221158c0594957 : SUCCESS in 48m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/897cccf1bead43f7acab4b9d57df76f4 : SUCCESS in 2h 21m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da6702d5e0e641f1824ef494b916fde0 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ef8c4b5cd1941d983e4d6129e2f2326 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a23bb42c77814eb6bb0c642343b162ff : SUCCESS in 11m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cd089824c5984eb6895bcbf91a727512 : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c511c7e0393f7fe1bcac70619df00b52e089b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 17:56:00.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b392e25da7480f449d9873cadbdad9519ea2e16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 18:44:23.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"68dd514f15606866a20197d4f742434135599700","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 19:51:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aee6d1dc5fb24a04bc313f7566d2ed57\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7fe5708f65ec4e1a8ae11ddf273c6b75 : SUCCESS in 59m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/54acad2d095c4d7f82b9645ea345dc4e : SUCCESS in 1h 05m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1a4bc69f29942a6ad1a4ba7035bce3f : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e817bb859f404d6aa7ae52759b6b303a : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f133d069710488d91e3ab8b3fba5ff0 : SUCCESS in 5m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a02096557c7a4dbcb400f9fcf2322e6f : SUCCESS in 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4ee10d4a92d0333ee8883dbb9b0e07ecec1ad69f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-29 19:56:35.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"dddb640a002cfc5f9e66d63854da648121818e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 21:26:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1b85d759cf349759855eccfc86e1173\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3322ad4f8136488b98c216c2f2d365e8 : SUCCESS in 19m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6da0774298e441bdbe0a4e8f66503255 : SUCCESS in 28m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1f8835126474c9cae203daeb7ba76c7 : SUCCESS in 1h 00m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e386865c4df942aa902c1bcfc084b825 : FAILURE in 1h 42m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e09aca5069b4c1cb2711c129c297820 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28631e08cd3b43f180d22a258121dbb0 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42896c9cae5a41c293467d0af81e98f8 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9407db8a59d546e19ccce3596a5e5d0d : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a07b559e7001482280db82690c1438f0 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27cab75617214483a4f72c58a2795fc6 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49a9372b529c4422a582ee5e22b08ee8 : SUCCESS in 2m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ff5068517e32478f9e88f5c21f07e702 : SUCCESS in 2m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/69b9a5287d5c4e6691ff592aa150fba5 : SUCCESS in 20m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dce0aa6349b7421d81a5e268e9912954 : SUCCESS in 25m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/264a00667a3a46f088018653de1b4fb0 : SUCCESS in 1h 07m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c6676c8f92a54687bdf8352260378a95 : SUCCESS in 1h 32m 43s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/97e21d284c20438e81706da789a5f3d7 : SUCCESS in 1h 54m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5300457b4e854aabbb9d6be5b219e1e9 : FAILURE in 2h 35m 51s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/443b29d0d11744cc9cb52ee79f997143 : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":6},{"id":"58db385fc650a25209bb569c18fe6ab94cdf8ae8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 10:27:15.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c389a588e07cec58433f4a81e169744a06b03de7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 12:10:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d90bbf4f8d4f4c1ca4534c592e24cae8\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e0d70fa3ce364db689bf086ac2857a50 : SUCCESS in 1h 06m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/014643b0c95f4fffab1635d6e8a6b591 : SUCCESS in 1h 41m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e58fb0a640714ce7841f6a4340f1e307 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a7c675587284bc8a8c7822388e2f7e7 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9fa1dc7a3144423af94c80f1d4889e0 : SUCCESS in 9m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/10578cdf42c64358b0348c799506d95a : SUCCESS in 1h 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"33b4c68ae168b709230ac5d1fa5132b41f755157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:27:43.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fdfc89f37984f7e95c45c0cdd3259e5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/68a5e8073583431d86e85301b318fab8 : SUCCESS in 18m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/19b4be8a317645b1b99a742719a73afb : SUCCESS in 24m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a9c23826c3a648178e5cc256b79201fd : SUCCESS in 1h 10m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8cf49b5d9eb549b193a29564a4b8bda0 : FAILURE in 1h 33m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7eefd7d203244f69a13cd25e4b82e3d : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/142aeeea7b1c41d6b330d333cad9517e : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59ff8439039043b2bfa5cf4dca5ac45e : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be107b812d004f4ba157a4e7232bc94a : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a0944fa9eb4c59826804a939900046 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc2e055bd2e42278c1de7b7de419699 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b43570b8df04557be74f3da9e01dd6c : SUCCESS in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/49b53a6835b54b32bee8e17b6f5dae0a : SUCCESS in 2m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0db9011bfe640b086dd505d93c44ef7 : SUCCESS in 20m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a85fde7bed194d859d9eb99db4ae2af1 : SUCCESS in 25m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/982ac6abd877407c82055f452c78aad1 : SUCCESS in 1h 05m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1203e19722674ca99968255aafdc6b11 : SUCCESS in 1h 36m 41s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/18ef8f8eba354239885b4bfc7725d97a : SUCCESS in 1h 55m 23s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7d4dd2dc8acd44b1b43f664899f545b1 : FAILURE in 2h 21m 23s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b690a12f1d85404489074d3c41efb309 : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":6},{"id":"0bee15ced6617ab7433741c7861b586dc89814a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-30 16:07:58.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8a4754d4db17809ba2c939941b009ba168f1e588","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 17:21:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9892e1e1b9cb4642ac9e1b59aeb1873c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f55d9511b43b4461a3c110644d1c55ff : SUCCESS in 35m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dabe61decabc468f9a51184ac12c119d : SUCCESS in 1h 11m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f5d0dd195c4e4f2a944761eb3bdbea8a : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dfcf9b6ed56b44a4b2074a1ce39916e2 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f36006b68b0c4d2299b1c91fc0a32d91 : SUCCESS in 8m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8d595cf652f649c4acca6ef5b24d1e12 : SUCCESS in 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"17c9c4e17268ecc5fc081c16e6e95dfb1f1c7850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:23:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66c5b99fd26544f4bf659bed853ae155\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f977155f58d44f0db4a904c20f0c4e90 : SUCCESS in 18m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/411e3fea08bb4ff6ab814aec002e8055 : SUCCESS in 27m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3bb922a3c4424d8ab48a4258eb8078fd : SUCCESS in 50m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c334f1881201413c81ae6e898a70ea07 : FAILURE in 1h 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2e16bb9f9a4366a8832198e65e088a : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39831e35889f4dedac5973bc5716c455 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/977cb7ae5e11485a9ff13be0cd631a2c : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b4cdc78bd1541dea3d7e710f5ad077b : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864369a945174b28b14d9254b9e11f3b : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb2dc2c3e264e4b88e04494a44d51e9 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/599979860b4849edb0362681776b96cb : SUCCESS in 2m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/99e1df35d8f14eff934f380042035abb : SUCCESS in 2m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dc0ffd88d699469ea4b175e426fffabe : SUCCESS in 25m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/215b8da338d04288b0e6779d3777c66f : SUCCESS in 29m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4c8dbd72931d4dca94245f114a3214bf : SUCCESS in 57m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/925eafbdd40c4197b46740274fc36413 : SUCCESS in 1h 31m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7e16fac94e6e4ed19cdff351fc14bc4a : SUCCESS in 1h 56m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/58eb54463f7a4e08be34544bc334823b : FAILURE in 2h 13m 51s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6f812d23d0fa41b198ed391c4c062bcd : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":7},{"id":"249ff116aef847477358bca10b717d337c886870","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 20:29:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9e7a236819df1a935b3366535ea5567f2bedadef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:53:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de54be398d1144a7960e24189c32a101\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f8562045f62f4d75beb57bc6b7266ca8 : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8908866f93b74a1583c01b355e199400 : SUCCESS in 1h 21m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb3cfc5c2a93416288bf54bbae7cd081 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ec4ba7445464879a89eb77924f61369 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/874854d11c7246e69d6b2a339441e59e : SUCCESS in 5m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ed370fbf54b343e19f9096b9d12bd1c7 : SUCCESS in 1h 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1108e1e2fc7b5a5d95983b9a25e8920190c92755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:42:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0adb72d9164d549578fc005e43ed44\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba94336364854c2cb0f9b9e5bdb4ec89 : SUCCESS in 20m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/93e43e3ae72a47009915f286fabdc6d6 : SUCCESS in 27m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eb9cfbbabebf4e58ba9776d5817656a8 : SUCCESS in 57m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9304a40621424b5b98be7c43c7a75cc0 : FAILURE in 1h 29m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f17557c4b54bb48dbd8b1caafe006a : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50d7352d7831459b8b771113c47f6c7c : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5fe55db949c455cabf14fb609afb91d : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f0265b45f524c24ae0f47fc0e81aa1a : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64180e1e668e469ea09208d1fd4fb515 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6138a1685c4145b8986d4a936c02afd3 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2e8c943914941f895566e251ac9535b : SUCCESS in 3m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a983f5e3063413d85a09551bb7b37d0 : SUCCESS in 2m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8ee1f383c6b2413ba7c3edf27d031877 : SUCCESS in 22m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/eb77bf3c406b4961b4d3a4a22910e888 : SUCCESS in 24m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4df8ee9dd39a4c0282534d11a15d0951 : SUCCESS in 1h 01m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/93575024c4e24bfab4feb9d480d03b59 : SUCCESS in 1h 24m 24s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2d57e20593ec4e38a0070e2a8c06ef8b : SUCCESS in 2h 07m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/15d633994f81438a91ce767b5c993d0d : FAILURE in 1h 32m 53s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/770d320c613d4305b7e1cfb51c9794a5 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":7},{"id":"c53860327fbfb4e2c10835d61999dd2f9a143566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:43:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"54a09fdad5b12197ccbd036ba16c3630feafaed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 00:23:37.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18beec8d45f4a219cf5833acd8aab27\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/938ab0e519a14049ac4f0df56c184e90 : SUCCESS in 14m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bc9a038669894dc981178c3299cf24bb : SUCCESS in 21m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1c2f05d5ba6043ffb9dc9a27ef872dce : SUCCESS in 49m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6423e234f87049919f595015fdad6ddd : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/014489a3a569419781ccbbd37f390b06 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64e4f7728b214c84942d3800291870f5 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28534b1eb825439cb2cb4ffdbbeb703f : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f889214ce78478c9829499b7dcd4ebf : SUCCESS in 2m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a50ab6a0aa8449ac989e76234cd13f0e : SUCCESS in 2m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cdcf77785bae46e68724d67c9f2a71a8 : SUCCESS in 21m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e4e2eaf0405848e9bfa044781fa28e31 : SUCCESS in 24m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/965af8c161f54656a18cad87d3f2d518 : SUCCESS in 1h 07m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/be88428af74447e9933bb72408e22053 : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":7},{"id":"c1adcb75efb203f6486ae05cf546de69cb967d64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 00:23:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"2de0106f10e2ff4ce3966e35e37bc2c5cd78c042","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 00:24:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b33a960c2fca440394884e232ab14430\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c112e6021e48442ea24acf830358a064 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/484327eb4eb244acb014561523b64d9d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bcf37e5016ecf1bc50dea111285cad887fd471b3","revisions":{"a1f7797fe4b578a0aa0a7098566fa280a60047b9":{"kind":"REWORK","_number":1,"created":"2025-05-27 07:04:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/1"}}},"commit":{"parents":[{"commit":"ff373463c431635f4264490cf3a1442e3d9b38f2","subject":"CI: Drop centos9/rocky9 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff373463c431635f4264490cf3a1442e3d9b38f2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"subject":"Rework mariadb backup script","message":"Rework mariadb backup script\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1f7797fe4b578a0aa0a7098566fa280a60047b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1f7797fe4b578a0aa0a7098566fa280a60047b9"}]},"branch":"refs/heads/master"},"f32c3af1f1432784752280db9d95e3f86467bcb1":{"kind":"REWORK","_number":2,"created":"2025-05-27 08:39:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/2"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 08:38:55.000000000","tz":120},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f32c3af1f1432784752280db9d95e3f86467bcb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f32c3af1f1432784752280db9d95e3f86467bcb1"}]},"branch":"refs/heads/master"},"b5dd7889c8c24d5adedf921fce011121ffd76f01":{"kind":"REWORK","_number":3,"created":"2025-05-27 08:53:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/3"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 08:53:41.000000000","tz":120},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5dd7889c8c24d5adedf921fce011121ffd76f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5dd7889c8c24d5adedf921fce011121ffd76f01"}]},"branch":"refs/heads/master"},"a463507496352c9b927a4ad8cc836a55d2acf4b1":{"kind":"REWORK","_number":4,"created":"2025-05-27 09:00:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/4"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 08:59:52.000000000","tz":120},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a463507496352c9b927a4ad8cc836a55d2acf4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a463507496352c9b927a4ad8cc836a55d2acf4b1"}]},"branch":"refs/heads/master"},"4ce4b90787f8caffca03a6782abcfc57a3596088":{"kind":"NO_CHANGE","_number":5,"created":"2025-05-29 17:56:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/5"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 06:09:28.000000000","tz":120},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce4b90787f8caffca03a6782abcfc57a3596088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce4b90787f8caffca03a6782abcfc57a3596088"}]},"branch":"refs/heads/master"},"34e17577d49ef2b0e6f40269f8e4b0ce09fed4a0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-05-29 18:44:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/6"}}},"commit":{"parents":[{"commit":"d5310057e5fbad6e681f3fa645d47443029a3fbb","subject":"[base] Ensure full control over APT sources for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5310057e5fbad6e681f3fa645d47443029a3fbb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 18:43:39.000000000","tz":120},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34e17577d49ef2b0e6f40269f8e4b0ce09fed4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34e17577d49ef2b0e6f40269f8e4b0ce09fed4a0"}]},"branch":"refs/heads/master"},"bcf37e5016ecf1bc50dea111285cad887fd471b3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-05-30 16:07:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/950948/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/950948/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/950948/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/950948/7"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-30 16:07:48.000000000","tz":120},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcf37e5016ecf1bc50dea111285cad887fd471b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcf37e5016ecf1bc50dea111285cad887fd471b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
