)]}'
{"id":"openstack%2Fkolla~951090","triplet_id":"openstack%2Fkolla~master~I8e20c77a8f1538162fc1dac9dc92c837348a09f4","project":"openstack/kolla","branch":"master","topic":"2111828-lsscsi_in_glance_container","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-30 18:31:37.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-30 11:31:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-06-27 16:27:29.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-06-27 16:27:29.000000000","reason":"Change was submitted"},"37725":{"account":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"last_update":"2025-06-27 16:27:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e20c77a8f1538162fc1dac9dc92c837348a09f4","subject":"Fix glance-api: add missing lsscsi package","status":"MERGED","created":"2025-05-28 07:51:41.000000000","updated":"2025-06-27 16:28:27.000000000","submitted":"2025-06-27 16:27:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951090-2111828-lsscsi_in_glance_container","meta_rev_id":"caf90584674af371ac80580184c5a48da4cfada1","_number":951090,"virtual_id_number":951090,"owner":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 16:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-27 14:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-30 11:31:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2025-06-03 11:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-27 14:22:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-27 14:22:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-30 11:31:55.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-05-30 10:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 09:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-28 09:17:53.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2025-05-28 10:46:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-28 10:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 10:32:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-06-02 23:05:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-06-03 11:03:12.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2025-06-10 19:36:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-27 14:22:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1c60c2b109981a13465b7c73832e4cf8e09a941f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-28 07:51:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b51136062f05874904c81eb956bbf27fcc3ee9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 09:09:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f977870033d948839295ba0c5e22ef21\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eabfbf47f80544898754ec60160ffd22 : SUCCESS in 31m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f0813b36d92475aaf0774461ca081fb : SUCCESS in 1h 16m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff7b59265d094b22a9d928b2a840e167 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4559cecf47341a9b8c4ceea8ddf8b09 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5ff5c7c2ba04b7d807db10f72ecfef6 : SUCCESS in 6m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f03292f8c5304ceab494ccd672596a05 : SUCCESS in 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f6d2358c76c761be7792174f16b926aae3cc137","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-28 09:17:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01410776d0b31d3874b00ec29a0e5ec02bdcaaaa","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-28 09:17:57.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"26d7f372a8e8cd82f92e9d181aa8bc33ddbd9a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 10:48:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac360441e34d4b8cb54c07de37210eec\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/717b70a986804c0991f6dc1bc8060168 : SUCCESS in 15m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a4ae37f329854522b547cbe75708296b : SUCCESS in 21m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/03745ec3d02d4ba8912cccdc8f973e4b : SUCCESS in 59m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5db490d48e534497bd144dadbea6d7ff : FAILURE in 1h 39m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5a79b77bd44beaa2905f96ddfab169 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96f2142334c649df881fefb6f1f982ac : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/282d78875a5047a7a93ebc4a9b7ce321 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc07ab43952c4930ae892176aa7e5497 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aacda8205d254024a308441dfdbce85f : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae3ece38fe94bb9a15328ed7f7feeea : SUCCESS in 7m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b1d57ceb9c4f48049c32444562143c58 : SUCCESS in 2m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/78d5e72d8779432eb4f301af205d1e17 : SUCCESS in 23m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0c63ff385ba04a649bfe19b25db5b80c : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/32cdb262e09445b39981ffc4e66a71c3 : SUCCESS in 1h 08m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f68166c075547569c16647ed3785a1b : SUCCESS in 1h 32m 21s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/43890eb7b16a4f8c9df11cc245ffa0c3 : SUCCESS in 1h 58m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/92f07fefb0f442eea5427f238343fc13 : FAILURE in 2h 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a575eb2047ed4d287a617068680f6d74fd1e8365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-28 11:27:50.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a00a8662d96a7e0051fd01a511d47880a0c1da8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-28 11:33:43.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"63a046fd54d308822ded81982fa76a646b6bfbe8","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-28 12:16:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"500ba5f99450f6473d491d0682f3becb03de1c1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 12:49:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9e83390a412492086f350472d4afe95\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/335a92f3e48c48e497b433090dfab728 : SUCCESS in 35m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3eb65f9e500249f388b9bcb54fa1b281 : SUCCESS in 1h 04m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7664656a1e0a4ec8bd9426a491eb08ff : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/983058f4eeef423f9e66cc8707e9b02e : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35374406e8274293bb95fcc7b537a924 : SUCCESS in 9m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7ed5256788e147b29c6818489b25d87c : SUCCESS in 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43decc3fb9fa0621d5777e0dcd0674edb76814f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 14:11:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc509591451e4a4eb6b586c9dd5d4024\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d6c357e74fe54370a3ec3762962bda59 : SUCCESS in 17m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fd21210acf0440828ddb0fbaff24b626 : SUCCESS in 25m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/07215967dec54950a7c61a892143f3ab : SUCCESS in 1h 03m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1040093f953d4acf993e788436f573ba : FAILURE in 1h 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f6b4ff3d4c45c1be53b725afae1da7 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73d09944635f4330bf6071f3629ed590 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5afe71648e564d6e98cf2e5b9de0e760 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0a9f1fd8f2b4009a82dd83fb9b33bed : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cfbfe0afc704527a561bd93eab4ee90 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2650aaee59384c92a2bdabfe1250c3c0 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5de53fbd190046e6ba59bd99dd03585e : SUCCESS in 2m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ddd820e118604f10b436fbe5c603dc9d : SUCCESS in 2m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d3eac0d452a84099a9798a4a79eb28e4 : SUCCESS in 23m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f8ba309651c040e19d1900e25948afaa : SUCCESS in 26m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/10688f974193414f919a91c755805157 : SUCCESS in 1h 07m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0d36a15d5e554af996641689c95c5493 : SUCCESS in 1h 32m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1f611da2428940818039e11ff5a4fe5d : SUCCESS in 1h 57m 51s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3c3a485f28094ec898c9046d47738d1f : FAILURE in 2h 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d45e09c24ba7c7f7bc03ca7f20758e50d09e4c26","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 10:32:01.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6b40587ac9e864f09342fc26966817e37e067184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-30 10:51:06.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2566cd293fdf493e9f1685081ee243690e8fb4ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-30 10:56:44.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ddc9d53951447113ef2c63c57b7cb116dd102ce7","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-05-30 11:07:19.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"371f20113a84afa0caa4523f71540310d8770211","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-30 11:31:55.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3464e4818882bf01979419968e3e489b917c9c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:21:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cb2552886064cf2972de53c76ab57e6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9f0c61fbc1d04a8482a32e3eb7bde422 : SUCCESS in 17m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0e8e7b54bc374193aae5db197495efda : SUCCESS in 25m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3f6249bd6a20463389a75a358e8f80c4 : SUCCESS in 1h 05m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/502a5eb05ac0411eb6175998825b671a : FAILURE in 1h 41m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2d04ed12e14cceb050628145f0fbc9 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f75e37205fe4f5c8b955d10052dfe60 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b3810ddf28c401c8d9fd38cab7b1e31 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1a788df5b2d46c8a10188bde556d7cd : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/233c3f82227c44bb9fac0d2b0655c31c : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3bcb0ef4c14b06bd19bc1f5a7c65c8 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27731d345a2543bf919e42b20e0c6cf4 : SUCCESS in 3m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ef10fcb921e4e85a9472a04df0d961d : SUCCESS in 2m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2e276cf20c49479aa62680e2ec441542 : SUCCESS in 21m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/313a84f4d13c40f0b9437f75a596324f : SUCCESS in 34m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b0124ac8e20a4add89d6aa83345dcde6 : SUCCESS in 1h 09m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/171cc404608c47bd98418b9d06c4d32a : SUCCESS in 1h 33m 42s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2223e2884a974d19b67e211505015b8a : FAILURE in 1h 24m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/57eaea77f46c49a4985001105993c83a : FAILURE in 1h 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8e6fecc66f16bda405c64fffc5d86dc91c838a7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:32:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d419a491c804a8a8a76c8db05594a45\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/de03e7454c424568a23d672d9be25925 : SUCCESS in 47m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a27009fb67047e0bcd26493d1ed5020 : SUCCESS in 1h 10m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2359119562ab42afb2a14f5fd61c3af1 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/065741cb1b8b450898eb3640b4e8700a : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b43c08cdce064af4b22cd5753c25013e : SUCCESS in 6m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e798c37fe4874197b02841e044f2820f : SUCCESS in 37m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c472201dc8458f1d1c685db454a9a2080a1d0494","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 18:32:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78d089a00d9e364810ede52661162adbcbb51ebe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-02 23:05:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c3aeb40b54e89c02c9afdc9bd6753275cc594eb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 01:32:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/064ca18c532d43cbbc49a7d1f11cc519\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c388ac5a413b42e59e561269355369fa : SUCCESS in 33m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f6fb65478c584525b5c74e5c29f43ea3 : SUCCESS in 1h 48m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e22bf0b06a7542e9a82d9f2f4b7e1564 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c383420c6b84bb7845eb471f977dccc : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dbd6d74829a841cd938bd57663671985 : SUCCESS in 5m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/75a10cea4afa40c7ac1a00f172ff1312 : SUCCESS in 1h 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3d24c72ce0bc60dbfbc9edfe66cd4b5387ff8c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 01:53:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1e141d85664ca09d920daafbd0ad23\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b0bb58d6b2cf4ef3aae97ff7f23dd37f : SUCCESS in 19m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ef87ee3fbfc94a8ca2e6cda6cc797300 : SUCCESS in 30m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d98ed8e4dde1499aa1dd25b2b4aa95ab : SUCCESS in 1h 06m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f4070723fcf9461e806718d4e499dd94 : FAILURE in 1h 18m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae4f36bf2d142d5b69618ac743d9694 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7343648ea244a418f8cce57cdb7a819 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d73153ad86db44dc87ff4e42f2e03153 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8984bd2d794242a29b3e355ebb24b0b5 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c731a828c77d422b9670648fb1badd40 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc541d6247f1409e8fddfff9cc01f7a0 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c847d419dc5d4dcba7cb4ff1ac6abfae : SUCCESS in 4m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f8a09550b4724f0896af0174a3534d72 : SUCCESS in 2m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3b30ca6cbf824af2b4431127805c43fe : SUCCESS in 19m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8d4435efecf94bc283f52721a64cb549 : SUCCESS in 24m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b74fb320a31f42fe87bb45427f2bfaf9 : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a58a3c88a0084fcfa5bef4beb5dfdc1a : SUCCESS in 1h 20m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c2b24903dce445d798d19f294a32fb55 : SUCCESS in 2h 01m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0196871fdcd149e199da1372d3828530 : FAILURE in 2h 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed9a1ed09aeb738f2a58a5db746b902cd5d28948","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2025-06-03 11:03:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1af7233f24be2b184efda6824f1d192e7f747555","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-27 14:22:49.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"634557b243c497a41e5387061a6d2ecea8c7f3b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 14:23:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ace468f4e0cc7130fab170b61d18286d95b516b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:27:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a1bb51795ff42e98705bcabb398c91e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/56c17c7555fe4309a6d321d3f1aea5be : SUCCESS in 19m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/48d6d58b8c1244a28f1e0ae81a71e1f2 : SUCCESS in 14m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4e82166cd84f485d8aa23dc419d04324 : SUCCESS in 1h 01m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68647609a8ac439ab69178391c6e173d : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bb8d6e47067455bb5acf1d2bdcca66c : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9537b1bfbabb4b4eb889dbc055d7a9de : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86ded44ea354b42a79b3de070d98561 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c872faa77c5403c955e3e2b74b7e9f1 : SUCCESS in 3m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8512b4c3a9fd49dca68e97ff90f6deb9 : SUCCESS in 4m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/edbbfdc84dfc4e7c9c68366bae8ea31d : SUCCESS in 24m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f37cf8c7eab4169adb00f8cb88fec50 : SUCCESS in 16m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d5c7da7ab2f64588bc22384a65cbc44c : SUCCESS in 37m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/82430b8626b54237a2f38cc35ce5d319 : SUCCESS in 2h 04m 10s","accounts_in_message":[],"_revision_number":5},{"id":"5914140c1645075fb4f7499b07ad374692ec8c50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:27:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"caf90584674af371ac80580184c5a48da4cfada1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 16:28:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90083054e2d749c682f45ffb47d212a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018eae18cf74414e9f78d8e56c5929db : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c853396f484869add5ae9344694d2c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aa2e504a709bb186d271736b6042b68c42ad440d","revisions":{"340f5bb8ccff33c7b851432ad02256ea9d9c89fd":{"kind":"REWORK","_number":1,"created":"2025-05-28 07:51:41.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/90/951090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/951090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/951090/1"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"subject":"Fix glance-api: add missing lsscsi package","message":"Fix glance-api: add missing lsscsi package\n\nCloses-Bug: #2111828\nChange-Id: I8e20c77a8f1538162fc1dac9dc92c837348a09f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/340f5bb8ccff33c7b851432ad02256ea9d9c89fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/340f5bb8ccff33c7b851432ad02256ea9d9c89fd"}]},"branch":"refs/heads/master"},"20cc3ac1b33a51be7907216f643a3bfc3fc6f9be":{"kind":"REWORK","_number":2,"created":"2025-05-28 11:27:50.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/90/951090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/951090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/951090/2"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 11:27:13.000000000","tz":120},"subject":"Fix glance-api: add missing lsscsi package","message":"Fix glance-api: add missing lsscsi package\n\nCloses-Bug: #2111828\nChange-Id: I8e20c77a8f1538162fc1dac9dc92c837348a09f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20cc3ac1b33a51be7907216f643a3bfc3fc6f9be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20cc3ac1b33a51be7907216f643a3bfc3fc6f9be"}]},"branch":"refs/heads/master"},"13f15bcbc69504c445dcbb3e64381098c894541a":{"kind":"REWORK","_number":3,"created":"2025-05-28 11:33:43.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/90/951090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/951090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/951090/3"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 11:33:35.000000000","tz":120},"subject":"Fix glance-api: add missing lsscsi package","message":"Fix glance-api: add missing lsscsi package\n\nCloses-Bug: #2111828\nChange-Id: I8e20c77a8f1538162fc1dac9dc92c837348a09f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13f15bcbc69504c445dcbb3e64381098c894541a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13f15bcbc69504c445dcbb3e64381098c894541a"}]},"branch":"refs/heads/master"},"cc6208913ef2c6d40a68236b1a4f3e1b37168c46":{"kind":"REWORK","_number":4,"created":"2025-05-30 10:51:06.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/90/951090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/951090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/951090/4"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-30 10:50:39.000000000","tz":120},"subject":"Fix glance-api: add missing lsscsi package","message":"Fix glance-api: add missing lsscsi package\n\nCloses-Bug: #2111828\nChange-Id: I8e20c77a8f1538162fc1dac9dc92c837348a09f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc6208913ef2c6d40a68236b1a4f3e1b37168c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc6208913ef2c6d40a68236b1a4f3e1b37168c46"}]},"branch":"refs/heads/master"},"aa2e504a709bb186d271736b6042b68c42ad440d":{"kind":"REWORK","_number":5,"created":"2025-05-30 10:56:44.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/90/951090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/951090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/951090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/951090/5"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-30 10:56:31.000000000","tz":120},"subject":"Fix glance-api: add missing lsscsi package","message":"Fix glance-api: add missing lsscsi package\n\nCloses-Bug: #2111828\nChange-Id: I8e20c77a8f1538162fc1dac9dc92c837348a09f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa2e504a709bb186d271736b6042b68c42ad440d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa2e504a709bb186d271736b6042b68c42ad440d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
