)]}'
{"id":"openstack%2Fkolla~951329","triplet_id":"openstack%2Fkolla~master~I00bcb30662bb12ed8711b28a64ef1941dcce893a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-29 19:48:57.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-31 00:25:53.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-05-31 00:25:53.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-31 00:25:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00bcb30662bb12ed8711b28a64ef1941dcce893a","subject":"[base] Ensure full control over APT sources for Ubuntu","status":"MERGED","created":"2025-05-29 17:56:00.000000000","updated":"2025-05-31 00:26:41.000000000","submitted":"2025-05-31 00:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951329","meta_rev_id":"db5a3d8897069b0f21326e1f8ff5a87cca218bd5","_number":951329,"virtual_id_number":951329,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-31 00:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-30 20:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-30 20:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-05-30 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 18:44:51.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-29 18:44:51.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-29 18:44:51.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-29 19:04:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-29 19:53:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-29 22:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f5daabd304c887032d3eb75527d2550ebd8d78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 17:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"315219801298a03afddd56513245bbe9376df906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 18:39:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b993128add9f16d835b5ab685d54fd408a3ce3b1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 18:44:51.000000000","message":"Patch Set 2: Backport-Candidate+1 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"04b021f990e2b123dd2d31ebba752bddfd6b100b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-29 19:04:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"489d2a285abe8d179c577695261950deb1c7ea84","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-29 19:25:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a9434c2eef0bdb259b5f3cd0f7f7ea630c8b7d2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-29 19:48:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f93b5e70eb51aa30814c8c1c947909023de25d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 19:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a5e8fe7dad4415e9c69e5f3a6f28dd7\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9703400a25f548518af11b4bc47b248c : SUCCESS in 35m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/573bff1e62304dd8842ea2345daad7a8 : SUCCESS in 1h 12m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f11aec7cf0b346ccbf96e8dd0799e6e5 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6cb7ad1c0fcf4903915a817841963bd2 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c528314f3f6409087b1f0a1f88cb7d0 : SUCCESS in 6m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bf6d692e72684c8bb93c66b46b81c6c1 : SUCCESS in 1h 01m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d410535016e09fb714238ba9beee421da66ed9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 22:05:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0dc3404e4d24302bcc83e667deee125\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4847501a3fe64964bb1c47b16cf14d29 : SUCCESS in 18m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4e6b0f50b10e47b3b4346349d2c84b8d : SUCCESS in 28m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d9fa94669684bf0aed4253357ff774c : SUCCESS in 1h 01m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0f722b6357674ee4811c202a03641857 : FAILURE in 1h 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b430c63aa140ec9649a25f03809a81 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cba56f9f2a5427b9e90d168cb74d672 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fc09ff46bf74dc7985aac4b2fffc92f : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9bf4ab5bc3a40649db1296f5a95d551 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07d279f98ba74f5eb9f5fc2c1e7288a3 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ebf3a8f487d4b40b9cb325c78bbc08f : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bb66a19180547dc9d12ac2aad4421ec : SUCCESS in 3m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f91fd18703bc427984f0d4815a304553 : SUCCESS in 4m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/84f82d5d3c644168b43c3350ce02a4e2 : SUCCESS in 23m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/62a15f8470644d648342d1fd451789dd : SUCCESS in 28m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d75bbf295ce548deb4b71dc9723f128a : SUCCESS in 1h 02m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fad5c8f602d84faa9788cdb4fe7b7f26 : SUCCESS in 1h 29m 22s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6f693c29bef641b3accc03713235c1aa : SUCCESS in 1h 53m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/91e58602ce824de184daefcf1b415ae8 : FAILURE in 2h 48m 39s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ea2e9adb59204e388f4804eda486fb54 : SUCCESS in 1h 03m 53s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e728be74f5b64bf895660b21313c4f22 : TIMED_OUT in 2h 31m 25s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/db71ee4fdcb0419bbda987dd000e400c : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/27738fba9cf54412a8b54157989d5418 : FAILURE in 1h 09m 08s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f4293b4487204f87b86cf2cc3e26e8f0 : FAILURE in 1h 08m 16s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0dbd03c469be49ce83382da5a5d88b3a : SUCCESS in 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd4d5f18f9218dcb79a4af47445475e83a6fdf76","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 10:27:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f223ab72e62fd34e8d33603558ee3002036e53bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 12:38:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4fe8c53f4f4764bdab706e7492ca0d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c8b983aeba84925a7df98dc346391ea : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d402093e92c48848bb94dce9658174e : SUCCESS in 2h 08m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b139748579a6475cb2c5005e08ad3b48 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b6e88e198474d35b14e5fe1a03eddb9 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f70c3cf8fee41b48f1fd98d30b1cdff : SUCCESS in 9m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/df2f9ec3d2ef4071a97f591133cead5e : SUCCESS in 1h 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"710cb5f6d61f4da913d8b175bbfd2e2ab0e76a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:07:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6644b888bd94dc6b5febeb9c6594934\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/115bfc27adc64271857cfaf4c9579d65 : SUCCESS in 20m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/63be516939cb4187a04386dc6cc96e92 : SUCCESS in 23m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d275fd6248044d9be2fce0e1fdadf08 : SUCCESS in 59m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/594c3ea1acc54a4093097ffe7f54f32a : FAILURE in 1h 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634860c2052d4a9aba48e0274cd74776 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5df26554b5a7489ea7d977e71ab7b4fa : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5c592661df84a999c8dd7b673c438a4 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2242a0cc6f24509baa0f89c470895a5 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e95e750adb48b499d942598358d3c6 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55680eaadaf400e958b4c0c74da5423 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1069068df5d74161a081c19421569868 : SUCCESS in 3m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/aa29ebaa6a3d437cb20288501fec3041 : SUCCESS in 4m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c8485ec74f3143e8b1fbf4f1bbbf3d0d : SUCCESS in 21m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/737a36878bc94494b42f4f997fe17d43 : SUCCESS in 27m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9bef05ed8991432b9acf6597d5ca9b29 : SUCCESS in 1h 04m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/21e3c8f378fd4907910c135104ec3afb : SUCCESS in 1h 31m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/16644a70f1d746bd996f0d3dba28ca08 : SUCCESS in 1h 55m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3f23b5b3fb77417184d53ced6facfb1f : FAILURE in 2h 20m 44s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c2e46684220344e889e47bc53453a25a : SUCCESS in 1h 12m 48s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/497195e0e32e415481b4765fa4c25d1d : TIMED_OUT in 2h 31m 27s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/91bca6d93f5c4994ab3243aa4300d04e : SUCCESS in 1h 04m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/195d9825d7094716af36503ed2731369 : FAILURE in 52m 25s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e4b439b9d5f64561b0fc569f2e309323 : FAILURE in 1h 12m 15s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c3c492dcb5e943de8e5d211355910cad : SUCCESS in 28m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa1a227fca12e1039650992374b5804f5977078a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-30 14:39:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35869e04ab8ef3d5bf86cd625406200d3aa63566","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-30 14:39:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41e57bb716503405718a74c23e7f356539d37e5c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-30 16:06:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6d313f7825b6d7c497449f04ab5822c7c3b9d2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-30 16:07:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf4282f57ec4a86cafb8987b290ca6a29c7e837b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:08:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2a0b4743bbd46a68455c35270acd87c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0942d53a30ce498bb1ff4fbb64325ca9 : SUCCESS in 50m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c2c82e5ea9db48309dd437d7b0a5e2a2 : SUCCESS in 1h 57m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e8742c892664825a234622da8c29863 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae95510095c34195a6b569f602cef8d0 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e5ed86bc8544a3db9e53410691f6606 : SUCCESS in 6m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a8c289e665d94eb8a135f286a1967f0d : SUCCESS in 53m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36e9626a3f0bc93d10cd14fca33f38abd1347857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:20:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba3f3481b4d64ef889b41042f853041c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba984b224c2f4780b87265d167fd57b4 : SUCCESS in 21m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eb9ef3cd92284d19917166603dfcbb3c : SUCCESS in 28m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5b507737cb084f719f6c54aafe67f948 : SUCCESS in 1h 10m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f12fb23890da4c8885eaa732f7cb8f95 : FAILURE in 1h 53m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aebf4d26344e40b0b96825076f310efc : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84ee977cba4841e99c67379ca433c51d : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a1c7e274b564025a5c3a334773dde84 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2df2a31a183547cdaf7b38f6774c92f1 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e15ab65a351a4177965634bee40f07db : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51402c871a5e4788b36687d4857fdfaa : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/913303fc398f4f09928acb2acdbe8348 : SUCCESS in 2m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/804d0892a5f44f73b48a4fefe244ab54 : SUCCESS in 4m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/15653dee48f348dd8d99729f402ef648 : SUCCESS in 20m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6ab3ddc1a764b80854a9ec5937cfed6 : SUCCESS in 31m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/90b6371212bc495ca78710bad36942a1 : SUCCESS in 59m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f194033e5e1b42a8bb2e9b45e029d705 : SUCCESS in 1h 28m 15s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a58ff1f357104e35b5579af12a46789e : SUCCESS in 1h 47m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/293132d12b724d05b3a57c13792e1f73 : FAILURE in 1h 28m 50s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/43b94e2cad524f578e88b212deba6439 : SUCCESS in 1h 13m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/88cf813b7603453ab2cec237576db919 : SUCCESS in 25m 02s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5c25568940ad4a1689a27ee5914656d3 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/23805adbab674b2abed8c5741372894b : FAILURE in 59m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a1f56efbae9840d99af16d0e691ab7e7 : FAILURE in 1h 08m 29s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0e5c1d6fa04f48e18a1511700bc5d5b6 : SUCCESS in 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8e6ab879a868f7c531559a6a05403bf1f38722a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 20:30:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5f00f33054b482215ebfa7afe092c4d5884acbea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:31:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f636f2672f4895a51dc65ea43305f5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cc38444f4af047c69ba1b4101f98a818 : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36c5f634662845a7ade455b95a04d98c : SUCCESS in 1h 58m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc13a5a8c2954a9fa7e1fd0e93ce3d7f : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4c64ebb5771484cbd9f0d355e99436f : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50efb020a61e4713a60f96d47f514dd6 : SUCCESS in 8m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d9e9ed3ff2e54b03ba166058d330af28 : SUCCESS in 1h 09m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f7fe0f733827f609d56f59ac4f8e09c8f1d853e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:39:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13562772d1e9476ea20b33f3d0b4a563\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/039ee558ad314f2db79aae7378237a6a : SUCCESS in 20m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/56ce0216ac9e465fbcdb1f734d92b752 : SUCCESS in 24m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/32faf08b794d4b618bdfdf5dfcc09fab : SUCCESS in 1h 00m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0ec10dce522d4a99bf3ff5b1c0ec60b3 : FAILURE in 1h 33m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09cc9095e8ed4ccea759834bfc5290a2 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/510e28086c694579a8e4fb79d5a88673 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c60456e846040e09672a27f560128b9 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b101005fb4b4ca2be0e4098c2f06ce9 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ec7a2137724e79afd2b7864fe5ec60 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6dfeb330b9413783e406bc098052e4 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddf491f18b23427587eef1b71283099f : SUCCESS in 3m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d49f0062daee421e9a20702ff8805a63 : SUCCESS in 2m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5dadb68623074cf1bcc8a987fd035302 : SUCCESS in 20m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/066dcadf7caf4defae53610a637abb1b : SUCCESS in 29m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/adb43848b372429082ec61b5e0207124 : SUCCESS in 1h 09m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2d86be1cf9ec48858251491a013c8e75 : SUCCESS in 1h 47m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/45536d7235524f958e2a007506a687d5 : FAILURE in 1h 52m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/94648d90d05541ac91aae2f9af9e5ee1 : FAILURE in 1h 14m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f126d2944bd344d99ab0c58de62f06a0 : SUCCESS in 1h 10m 22s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/797c804110ff483bad185f6c9a94e303 : SUCCESS in 21m 29s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4332f0a81d5f47848eaf0d682aa8809b : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7933758e28d543c3a65a73f64a93f948 : FAILURE in 49m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e24ee5e70b2c4d7ea3e04efb38cb6045 : FAILURE in 1h 06m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4ef335023f284a34894d86b61628a925 : SUCCESS in 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d504b6c221e5e8d09c611e861632b019a57174d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:43:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9931712b3384db850fcb4db521f7ee37fa8185ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 00:25:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f99499f66fd4e5093397d9044fd1cb9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d8f57595fc1c43d6a1716d3f4e89f805 : SUCCESS in 14m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49678bfc29e642a7a2986cb94b8060e3 : SUCCESS in 21m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0d8a19b2873d4d3eb60bc6abd32b765d : SUCCESS in 1h 02m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53718e93dcda473b9b53a4c80f215a01 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66039b734b6d4bcf9bc85948b82b74cd : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ad6e90ccaef40fea292c6bfb95dda97 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/555a4757d37740c8a2f505a91b8b3333 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a460b21d2984d4fa481d15a7c30947f : SUCCESS in 2m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f5b8df4c578b4eaba9d80f06c5913a4d : SUCCESS in 2m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de92435a69d04d4c94af5ded5c3383c6 : SUCCESS in 21m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5dcfa954b19a4baab9a196587ef4b57d : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ca1a0f730aea45489b4bead3fd114524 : SUCCESS in 53m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/15e3567e44e144f9b6f4474ddc98dc14 : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e064bfa4d432bff57b11fd89ced86309f742a625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 00:25:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"db5a3d8897069b0f21326e1f8ff5a87cca218bd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 00:26:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbca68a9344467a86135b9b882536b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b0250798d64d2097abe1866656c529 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd32acb63bac439a99be4e07192fcbf5 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b5849adc45e3a39e0837dea3ee8e1fb3cd057dd8","revisions":{"0472457f93cf80993f9325af6c391154d957d10d":{"kind":"REWORK","_number":1,"created":"2025-05-29 17:56:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/29/951329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/951329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/951329/1"}}},"commit":{"parents":[{"commit":"4ce4b90787f8caffca03a6782abcfc57a3596088","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce4b90787f8caffca03a6782abcfc57a3596088"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 17:53:55.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 17:53:55.000000000","tz":120},"subject":"[CI] Use security from infra","message":"[CI] Use security from infra\n\nChange-Id: I00bcb30662bb12ed8711b28a64ef1941dcce893a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0472457f93cf80993f9325af6c391154d957d10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0472457f93cf80993f9325af6c391154d957d10d"}]},"branch":"refs/heads/master"},"d5310057e5fbad6e681f3fa645d47443029a3fbb":{"kind":"REWORK","_number":2,"created":"2025-05-29 18:39:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/29/951329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/951329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/951329/2"}}},"commit":{"parents":[{"commit":"32d2f1719a978c62a44ca998b8018692b55ac0d4","subject":"Merge \"Drop environments for nose\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d2f1719a978c62a44ca998b8018692b55ac0d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 17:53:55.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 18:39:21.000000000","tz":120},"subject":"[base] Ensure full control over APT sources for Ubuntu","message":"[base] Ensure full control over APT sources for Ubuntu\n\nUbuntu images were occasionally failing to build due to\nunavailability or slowness of the official Ubuntu security\nrepositories. Although Kolla uses `mirror://` URLs to\nleverage Ubuntu mirror infrastructure, the original\n`/etc/apt/sources.list.d/ubuntu.sources` file was still\npresent, causing the system to fallback to default sources.\n\nThis commit aligns Ubuntu behavior with Debian by\nexplicitly removing the distro `.sources` file and relying\nsolely on the sources configured by Kolla. This ensures\nconsistent and reproducible builds both in CI and\non local environments.\n\nCloses-Bug: #2112140\n\nChange-Id: I00bcb30662bb12ed8711b28a64ef1941dcce893a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5310057e5fbad6e681f3fa645d47443029a3fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5310057e5fbad6e681f3fa645d47443029a3fbb"}]},"branch":"refs/heads/master"},"b5849adc45e3a39e0837dea3ee8e1fb3cd057dd8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-30 16:07:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/29/951329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/951329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/951329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/951329/3"}}},"commit":{"parents":[{"commit":"bcf37e5016ecf1bc50dea111285cad887fd471b3","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcf37e5016ecf1bc50dea111285cad887fd471b3"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 17:53:55.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-30 16:07:48.000000000","tz":120},"subject":"[base] Ensure full control over APT sources for Ubuntu","message":"[base] Ensure full control over APT sources for Ubuntu\n\nUbuntu images were occasionally failing to build due to\nunavailability or slowness of the official Ubuntu security\nrepositories. Although Kolla uses `mirror://` URLs to\nleverage Ubuntu mirror infrastructure, the original\n`/etc/apt/sources.list.d/ubuntu.sources` file was still\npresent, causing the system to fallback to default sources.\n\nThis commit aligns Ubuntu behavior with Debian by\nexplicitly removing the distro `.sources` file and relying\nsolely on the sources configured by Kolla. This ensures\nconsistent and reproducible builds both in CI and\non local environments.\n\nCloses-Bug: #2112140\n\nChange-Id: I00bcb30662bb12ed8711b28a64ef1941dcce893a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5849adc45e3a39e0837dea3ee8e1fb3cd057dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5849adc45e3a39e0837dea3ee8e1fb3cd057dd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
