)]}'
{"id":"openstack%2Fkolla~951473","triplet_id":"openstack%2Fkolla~master~I4adb64095036e7e6b97789dfc064da405019b3ef","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I4adb64095036e7e6b97789dfc064da405019b3ef","subject":"[DNM] Add script for mysql drain","status":"NEW","created":"2025-06-01 14:46:03.000000000","updated":"2025-06-03 12:33:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"77dfadc19d94f2c5587c3deac6b431c85d77a354","_number":951473,"virtual_id_number":951473,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-01 16:42:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-06-01 16:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-01 16:42:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 12:33:38.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"CC"}],"messages":[{"id":"695c290039fb48c778e4c667ff8150f4304d26d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-01 14:46:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f36c6de4a369d006da2e637096f43764a0b18c2c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-01 14:49:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"034158cc278979a9df6944d5c824d6a60829ffb7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-01 14:51:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc577ebe6a1ebf8556641ecf9d2e4048d5e1d880","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-01 14:53:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8ae5c0f78662ff1eaf585ce80b0896303b6d078d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-01 16:28:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66423923410a47809af144bac6cefdb9\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/87a43070811743f58f947f4d2e147455 : SUCCESS in 28m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fec33b720dae4a89968184849d4e7b3f : SUCCESS in 1h 32m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7792e05196884967bcf2dc35e85f752b : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad2b20017f3047b3a23fb8c22dd43540 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e7e24479f3ac4365bccae9ab4dcd6df7 : SUCCESS in 5m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a1a4f15467474456b42734f718c4bf56 : SUCCESS in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54474421fb5fdedae4ac85b40f296b1f790d8ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-01 16:42:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf6a73650e6a41409ecd9ebccd75d6e1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cd03e253573a436f9e3e87f6a9275861 : SUCCESS in 18m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ee8b82e8abe849b5a582f4a0268907cf : SUCCESS in 29m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c56a69ef19674c57b95cfebecfedb424 : SUCCESS in 48m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a35b61463c2b42d1a62f15dbb1925892 : FAILURE in 1h 19m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876a9f8c110943f39dc5d457939e04da : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cda0f0b9c94487eb8506e7c3904e517 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63b33fae9fb44c239719a91f8df005f3 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94e7b3efef3a495c875cb986d30764ef : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1701e9bd7db845e7b27c9f4de4bee8d6 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92989934bf1043ef8b7d6dd9101ad668 : SUCCESS in 5m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ff43be9b80fc42aabfd6ede964a115ac : SUCCESS in 3m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0541960bcfb14c5e911a9c2fbac10307 : SUCCESS in 19m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ef94613206be4bc6a4a5cfa83ecea6e9 : SUCCESS in 24m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2c0484735f5f45d894096b619d4c41ab : SUCCESS in 1h 07m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2284a83ae3964bf6a4120d8ae9097bba : SUCCESS in 1h 41m 54s\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/6e46f25f8f68493bb66a221bd73b11cb : SUCCESS in 1h 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b5b17d95593649b44602a682a5de770136edc85e","revisions":{"f43f603ffc808a0863d49888d9db06e640ab1e74":{"kind":"REWORK","_number":1,"created":"2025-06-01 14:46:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/73/951473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/951473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/951473/1"}}},"commit":{"parents":[{"commit":"cea5658fdbba91ae387e8d5e9d19401014fcba56","subject":"Merge \"kolla-toolbox: Fix apt_preferences_rabbitmq.ubuntu link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea5658fdbba91ae387e8d5e9d19401014fcba56"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-01 14:45:32.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-01 14:45:32.000000000","tz":120},"subject":"[DNM] Add script for mysql drain","message":"[DNM] Add script for mysql drain\n\nChange-Id: I4adb64095036e7e6b97789dfc064da405019b3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f43f603ffc808a0863d49888d9db06e640ab1e74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f43f603ffc808a0863d49888d9db06e640ab1e74"}]},"branch":"refs/heads/master"},"979da9fc1a7018b4111fd33dfe4a26eac8663352":{"kind":"REWORK","_number":2,"created":"2025-06-01 14:51:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/73/951473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/951473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/951473/2"}}},"commit":{"parents":[{"commit":"cea5658fdbba91ae387e8d5e9d19401014fcba56","subject":"Merge \"kolla-toolbox: Fix apt_preferences_rabbitmq.ubuntu link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea5658fdbba91ae387e8d5e9d19401014fcba56"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-01 14:45:32.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-01 14:51:49.000000000","tz":120},"subject":"[DNM] Add script for mysql drain","message":"[DNM] Add script for mysql drain\n\nChange-Id: I4adb64095036e7e6b97789dfc064da405019b3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/979da9fc1a7018b4111fd33dfe4a26eac8663352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/979da9fc1a7018b4111fd33dfe4a26eac8663352"}]},"branch":"refs/heads/master"},"b5b17d95593649b44602a682a5de770136edc85e":{"kind":"REWORK","_number":3,"created":"2025-06-01 14:53:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/73/951473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/951473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/951473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/951473/3"}}},"commit":{"parents":[{"commit":"cea5658fdbba91ae387e8d5e9d19401014fcba56","subject":"Merge \"kolla-toolbox: Fix apt_preferences_rabbitmq.ubuntu link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea5658fdbba91ae387e8d5e9d19401014fcba56"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-01 14:45:32.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-01 14:53:35.000000000","tz":120},"subject":"[DNM] Add script for mysql drain","message":"[DNM] Add script for mysql drain\n\nChange-Id: I4adb64095036e7e6b97789dfc064da405019b3ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b17d95593649b44602a682a5de770136edc85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b17d95593649b44602a682a5de770136edc85e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
