)]}'
{"id":"openstack%2Fkolla~951500","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I562ce5e54752015863a20bf113bce74a69e02331","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2025-06-02 20:38:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I562ce5e54752015863a20bf113bce74a69e02331","subject":"[mariadb] Fix incremental backup using correct base dir","status":"MERGED","created":"2025-06-02 08:24:59.000000000","updated":"2025-06-02 20:39:45.000000000","submitted":"2025-06-02 20:38:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951500","cherry_pick_of_change":950948,"cherry_pick_of_patch_set":7,"meta_rev_id":"dadc118d8ff9a666c9b2c3046b68d75b92b26aad","_number":951500,"virtual_id_number":951500,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 20:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-02 18:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-06-02 10:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-02 18:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-02 18:46:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-02 08:24:59.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-06-02 10:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-02 10:18:43.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2025-06-02 10:46:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 18:46:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5b1a814874f514b807f41204909ba09856151fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-06-02 08:24:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c6de3df6843874d7b1365b40e31a04ad6772067a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 10:10:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f287551279204e10a120bff521b8df6b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dfe1bbeec55f4f3f9d3f2948c8cc8be6 : SUCCESS in 52m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2b925eec63b843a7be666b53d0c65671 : SUCCESS in 41m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90551c3ad5e34ef78416c9c333c9d46e : SUCCESS in 1h 43m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b40390577546472187cea8b7619d2f59 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e3ecae50bee4bfe8e2b03919e6f7745 : SUCCESS in 3m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7c926140969449c6b0fe92d7e3a19d63 : SUCCESS in 53m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fbbd38625e1a47c695724ae9d53b03dd : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ad89ea93faf6c963ec982d5ad6138d45c7f976c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-02 10:18:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f60c0ecd9a6b3046b1c13122deb82e4e578122b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 10:46:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed7e882092fb47cba91355d980747117\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e362cc5702864d83b8bb132b3e6fa357 : SUCCESS in 21m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6f627d32f8d24bd2a4046b805f5b4e99 : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cdf8f4d2d50042dd9164ffa5500676cd : SUCCESS in 20m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b54ba152be89474699041a883eae7311 : SUCCESS in 26m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/42dd7f354a084ff5976d8b7e95c08f98 : SUCCESS in 59m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0489b6373c77431bb855d827e1d4d56b : SUCCESS in 1h 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046cd1e41a1e4a018ada2c0d9c33b740 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72e2ee4655c74ee589f0bdde64f9e98f : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/023015a785f84286970943769a615837 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab44bcd9ace14cbb8a32c13207161751 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64511165fc3f41c58a2b63cc5ed41000 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36513e04d404d6694a5b22b6ad308c7 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c842097c0414b148a56d1c2bfde1cc8 : SUCCESS in 3m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e5fb12902f5844adb695953c211efff8 : SUCCESS in 2m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/406e07edde424ecc9c2326e0702f2ddd : SUCCESS in 20m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/32352b5ee8174580b6dcd4020e18e13a : SUCCESS in 20m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2c107b993d9143ceb9a3378190b7577c : SUCCESS in 56m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5363f73ba3a74375a2b211c1791e3314 : SUCCESS in 1h 19m 56s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/16985bc45db442709deb566a6f369c96 : FAILURE in 1h 38m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8dde9c7620c0404cbe571d4d6d08459f : SUCCESS in 2h 20m 23s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f47e5a3548a7449e8ac99584100e055e : SUCCESS in 20m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/284656aa30fb49829a4ca2ab05933e50 : SUCCESS in 18m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9ddfce8eeaaf4bccaa712f0bac9db3fc : SUCCESS in 23m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e77c23ee7c9840ffbac0477d4485f123 : SUCCESS in 1h 06m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2b8c604e3c8a43199404cbc0a46f1381 : SUCCESS in 1h 12m 47s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1c405716ab9d46848d14a7fc12709440 : SUCCESS in 1h 36m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/46ac158cf9614ccdabb08f08029a4f6a : SUCCESS in 2h 04m 31s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7b1a4d61d14643a99ab51ea5d8a30641 : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d283893d65861139a95d50101eddccac2d8d284d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-02 18:46:35.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c74a19020bb23e3e1a0325b733b62187ce4b8087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:46:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fb56a308697d9f114b6fa56c3047dc12008bc0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:38:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/715894ea4b58487588fa3f420b0a646c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a6d04ceb6efd47e6a1a8032eaa0feee0 : SUCCESS in 21m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b5f6fa2d393140e4b6b657839ed2aa02 : SUCCESS in 29m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d6bf838afca5478588e3d660c4e7ac89 : SUCCESS in 1h 03m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d89c6c79d5e44d3c815f53037d6d4c9c : SUCCESS in 1h 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c1568e01454928a66ed264a5ec0305 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef22c9ff7752468582f7ea2c9d1fe70d : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62313407a95244bfb05acd1d3417bb8d : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5b24f9fc7b4abc86db522b2a410f4f : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d392e68a972d420caac931ce5a7de952 : SUCCESS in 3m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c32e0e93d5a64bedab3744a68feee51e : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4fe10485cad94c72be5ade7d1f90f580 : SUCCESS in 18m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8baca381f0eb4cd3be7c42ffa0ab2720 : SUCCESS in 21m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cbc437db9bf842fdb13f883334db0040 : SUCCESS in 1h 05m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/843c213664114450a8ad0973fdaf3077 : SUCCESS in 1h 15m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/886ea95531b0428aa9eb6242c637e19b : SUCCESS in 17m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c18a13b3d0b04611bf6d8dcdbddc82d7 : SUCCESS in 30m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/390c184d031c48efb96a190e02c3152b : SUCCESS in 1h 07m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/963b6bcd74144ec99161b848a4cf77a0 : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"61bc0c065f3b1bcc2bb2973bbf8faa6ec9ab47ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:38:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dadc118d8ff9a666c9b2c3046b68d75b92b26aad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:39:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98929d79d104866990881c8ee5c500a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a77a67098940b8967ebcea2822f654 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a9c2c4d7e9d4a4892181977fe4a7295 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1aaf86fd575ace795edcf93a5abb4fa3c6cca921","revisions":{"1aaf86fd575ace795edcf93a5abb4fa3c6cca921":{"kind":"REWORK","_number":1,"created":"2025-06-02 08:24:59.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/00/951500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/951500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/951500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/951500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/951500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/951500/1"}}},"commit":{"parents":[{"commit":"895b6c4ce80714d925f40163a54d6eb556097625","subject":"Revert \"mariadb: pin to 10.11.11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/895b6c4ce80714d925f40163a54d6eb556097625"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-06-02 08:24:59.000000000","tz":0},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n(cherry picked from commit bcf37e5016ecf1bc50dea111285cad887fd471b3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aaf86fd575ace795edcf93a5abb4fa3c6cca921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aaf86fd575ace795edcf93a5abb4fa3c6cca921"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
