)]}'
{"id":"openstack%2Fkolla~951501","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I562ce5e54752015863a20bf113bce74a69e02331","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2025-06-02 20:22:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I562ce5e54752015863a20bf113bce74a69e02331","subject":"[mariadb] Fix incremental backup using correct base dir","status":"MERGED","created":"2025-06-02 08:25:37.000000000","updated":"2025-06-02 20:23:43.000000000","submitted":"2025-06-02 20:22:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951501","cherry_pick_of_change":950948,"cherry_pick_of_patch_set":7,"meta_rev_id":"a71dd28b8bcbc2ff264eed3116d239e78b03c41e","_number":951501,"virtual_id_number":951501,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 20:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-02 18:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-06-02 10:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-02 18:45:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-06-02 18:45:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-02 08:25:37.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-06-02 10:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-02 10:18:32.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2025-06-02 10:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 18:45:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1004e4408e38c1e7c32108ced50b57f38da595d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2025-06-02 08:25:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a67c054e59e636ccb21e6d46025c5a47f3244938","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 10:17:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb9854affbc4546b898716e65113e82\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b9d7022b3f8449b495ac57e9e3c5fd83 : SUCCESS in 56m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59de7bbcc5d04d5c9e9264904548e9a8 : SUCCESS in 40m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a3d915106b23476d80ba790e6fc4f65d : SUCCESS in 1h 50m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b7a632889bf4a738fd0bb0af22d3927 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dfca4d24414d4ac6b873f6125f8be220 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/accb50b90312452780834e46e101d293 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9a4ae44bb7054678bc7d9b2068694c16 : SUCCESS in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21ee7d4bdb97ec95ad5cc7904d95686856805ff7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-02 10:18:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4fdfa0bb0de3f49ffe8eb63b40bfc9a1aec50f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 10:52:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce77fda6939d4da7b86fd0eda2230a44\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c8c7547121504ac3a9a3b3d0c670f613 : SUCCESS in 25m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e6d8e6eb64cb475a887310cf1d21d4a8 : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/84471c1428c24494bbd50daf8abf957c : SUCCESS in 21m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d98e7d994ae04ea4a4784acc2f5ac79b : SUCCESS in 29m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/985be283dafe47c2b74d5e634831ccfe : SUCCESS in 1h 07m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1796a29d2b93413cadc3edc6b4ae9b86 : SUCCESS in 1h 33m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88aa2fa530dc4a549c59de866e917ab1 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55fd840877b443ff87a351701aa83ed2 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5229978a26fc4324aee6fe9c833f02a2 : SUCCESS in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/489a2910bec84dff99f8a46b3aec7bc9 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffd8958507a43869761fc683c112bb8 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d37bc1dd98544e309231047a13bb302b : SUCCESS in 3m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/43b264a17bfc4ed19a0973a0e4379d78 : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/994124d2ca5e40b681600a8f00e9296c : SUCCESS in 22m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f1d359fe5dcb42ce8ad022047e1fd26b : SUCCESS in 1h 37m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ae79de52fb3c492aad1d1e95b70985ee : SUCCESS in 1h 06m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/513c9c28da0b49d3a586e284332dd05a : SUCCESS in 1h 33m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1965ae14f0ae4fc6af1b940688698951 : FAILURE in 1h 50m 05s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9e463cb5663d4fe78b58fb28d9709856 : FAILURE in 1h 15m 13s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/634ae0a996f149b3b10f3bba5b2d8f77 : SUCCESS in 25m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e193a0caaa814bbda532ed0d0682eaa4 : SUCCESS in 18m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c7ca1e78b64e4c54aa1d0c59c055a8b9 : SUCCESS in 22m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0632df0866884fa9974de92d1d00d0ad : SUCCESS in 54m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/939d8f78a9ec4cecb32bd9df16ba6b49 : SUCCESS in 1h 35m 11s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f49868f9135a4143a3040168fb6109ac : SUCCESS in 1h 45m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a4bc4f42a93849f5b4e58b4c63b3ed58 : FAILURE in 2h 11m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8fc15e2c89474511af8b14713588e014 : SUCCESS in 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9e654854307801b4b11b9e484fa1009d352afc49","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-02 18:45:15.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0191f3a754bbe5b9228f7910fe47924ed67f9981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:45:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9869565745803306146712261c55299a19561fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:22:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318fa0444dd34f62bfdb980d2fa1ed43\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/23d13181e3b84ac19bbcec2fa1497b2c : SUCCESS in 22m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/830ee3843a3a449bbb8bef67c6dface5 : SUCCESS in 33m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4ffa9c99f90c4065b233ac2e69867921 : SUCCESS in 1h 04m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0be0f4020474b4ab5682ae31040354a : SUCCESS in 1h 33m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fba52f4696401d99f68206b446cbf7 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99bc16efcc3545878b2c3c74d958b549 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a666bd1034b54a828f162f36248640af : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef27d723e5a84e548ccd32f765491733 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd64b2df4344b1d91b4c7fa3672b6cc : SUCCESS in 4m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f15c5f14a64e45c391a9852255a2fbf6 : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/52d88c73966f4813b360cf8fa7ebbbed : SUCCESS in 21m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/452537ecc6e94f31a04ffb29b4c2c25f : SUCCESS in 1h 07m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ddf0c779be074f5f8c0b5fb7d593dd8c : SUCCESS in 1h 01m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4d6cb22c671e44229d0ef23d5156acf4 : SUCCESS in 1h 25m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c5ad7cf48852411fb5e16803e5a32367 : SUCCESS in 20m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/131e356459634cbea2e2872d1de53c44 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/70c18cf758204b449246b14eca2113d6 : SUCCESS in 1h 00m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/815c8b79149847c0a60580040682d2b2 : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7f32a76bb6f57566eeca63f780832d2066fb4ab9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:22:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a71dd28b8bcbc2ff264eed3116d239e78b03c41e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a17c0a19e8b4fc68e155b54181a1ddb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cdd27caa85147368088fa8ab9e83b2a : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e92bcc10c0af48199a89feffef20a2f1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","revisions":{"5f6f9f5df6e3f5d790787d29aea5ae820c759c12":{"kind":"REWORK","_number":1,"created":"2025-06-02 08:25:37.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/01/951501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/951501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/951501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/951501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/951501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/951501/1"}}},"commit":{"parents":[{"commit":"d9621f295ec8773f630edbdb9f442021a583f62c","subject":"[2025.1 only] Revert \"bifrost: bump Ansible version to 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9621f295ec8773f630edbdb9f442021a583f62c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 07:04:40.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2025-06-02 08:25:37.000000000","tz":0},"subject":"[mariadb] Fix incremental backup using correct base dir","message":"[mariadb] Fix incremental backup using correct base dir\n\nThe original script used --incremental-history-name and compressed\nfull backups as gzip streams. This failed because mariabackup expects\nan actual decompressed base directory for incremental backups.\n\nThis patch:\n- Decompresses the latest full backup into a temporary directory\n- Uses that as --incremental-basedir\n- Aligns the backup flow with official Mariabackup documentation:\n  https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/\n\nThis makes the incremental process reliable and fully supported.\n\nCloses-Bug: #2111620\n\nChange-Id: I562ce5e54752015863a20bf113bce74a69e02331\n(cherry picked from commit bcf37e5016ecf1bc50dea111285cad887fd471b3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
