)]}'
{"id":"openstack%2Fkolla~951846","triplet_id":"openstack%2Fkolla~master~I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d","project":"openstack/kolla","branch":"master","topic":"docker_login_before_pushing","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-17 18:01:00.000000000","reason":"\u003cGERRIT_ACCOUNT_2271\u003e replied on the change","reason_account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-04 11:10:51.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"37726":{"account":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"last_update":"2025-07-03 10:43:10.000000000","reason":"A robot voted negatively on a label"},"36702":{"account":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"last_update":"2025-06-05 11:53:10.000000000","reason":"\u003cGERRIT_ACCOUNT_37726\u003e replied on the change","reason_account":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"}}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-27 17:39:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2025-12-17 18:01:00.000000000","reason":"\u003cGERRIT_ACCOUNT_2271\u003e replied on the change","reason_account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}}},"hashtags":[],"change_id":"I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d","subject":"Add docker login before pushing images","status":"NEW","created":"2025-06-05 11:52:14.000000000","updated":"2025-12-17 18:01:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":15,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"0f42c7950d1749daa36895e233e9848728f7e4f9","_number":951846,"virtual_id_number":951846,"owner":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-04 12:00:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2025-12-17 13:52:11.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-05 11:53:10.000000000","updated_by":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"reviewer":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"state":"REVIEWER"},{"updated":"2025-06-05 12:59:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-05 13:27:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-06 09:43:34.000000000","updated_by":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-09 08:26:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-10-27 17:30:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-17 00:31:53.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"CC"}],"messages":[{"id":"3fd934bb3b032715acc1567807ceda43696dfcd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"date":"2025-06-05 11:52:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9797859476cf5478e14e2b27482c03433dea9877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:59:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/112c3656cfa74c5bba8e4594c2f82e4f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/40e45a856d094570b5ae1b6348809101 : SUCCESS in 29m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/35f8e27d32e94bb280bd83169c17bbb5 : SUCCESS in 1h 04m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10b63b36211242b099bf84c2d78b32ea : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86b02de3df104c7bbb8c187b124cd60d : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a8726ef5ecdc430f82c5f20328beb8e2 : SUCCESS in 8m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ff5d97dd9a684f48865e89b6673fe27c : SUCCESS in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbd1faa9fd7971ece357d5e76edb3442b3625fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 13:27:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730ae0e3c3904e8f9a74b66dfb88c10a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/52a23d7eb39f4655b66039171cf7a604 : SUCCESS in 12m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b7e253bdb2484958890d5da4f1af93bd : SUCCESS in 30m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/76de0dab81654a2dbebbd0040ac2474a : SUCCESS in 59m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/554bd3cf99f442199009d85410c06a6c : SUCCESS in 1h 25m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad1ec03b621494db6061ec9e31ad945 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f100eee351e4d7180aab1463939b367 : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f74973c271004415aa0e2f671e066603 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65346c2319c6440795336943081f1b46 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33c528d4ec3f4ee594fbfb5fa0f542e0 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399ab9f01a3948d3b1dd00216fd31bae : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7ed0f9c8c16465c9012694009614a77 : SUCCESS in 3m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4bb3aebd3716424e8d05df0813163ba4 : SUCCESS in 2m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0ec9feeb948641c2aa6afad762f8bed5 : SUCCESS in 23m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/60a6ee1d809344db922b2fb6c11ff402 : SUCCESS in 13m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/37f87404ae594182ab417bc9b2a76b3a : SUCCESS in 1h 04m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4c83c57c32134bd49e850da04443655b : SUCCESS in 1h 32m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3bfea2bee9db91d9e827bf66073d88e3e10cbbc9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 08:26:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd9f4652ff8a70e2e5dc8478ed9325dde7d8e6ae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-10 14:26:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"864e6de66f87c135bea72134d0af38db2fc64ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"date":"2025-07-03 08:53:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa060b0b798ad539676dce7f9651aeb54fb15eff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-03 09:12:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74194ffc3304a5c6163fba8f0fcf2c03e81a5c09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 10:28:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4024dc07ea24a27ae240208d665e501\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e0bb99eff284a3da19462689a3025e6 : FAILURE in 53m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/23170f552e2741c6ba50857559559f90 : SUCCESS in 1h 22m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ab88faa18e8432bb6bc09b39bdd8953 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/320c473b7ed845cd93d7400be9b7006f : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ef648cfa39d457aac70ec1fbd2067ce : SUCCESS in 9m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7af9f7c819df40b1853f05dd1cee98dd : FAILURE in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"39964e4d02973e92abb8563d579bf0a37ea31302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 10:43:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be1a2e8afc3f4d15b59e307636d51e24\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b4f99cf61bf24b9196320ea2d9bdc962 : FAILURE in 16m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/91ceedba1b174576af28cca751d00d21 : FAILURE in 28m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6885d7acc98f4d61ba30e933c9e8b427 : SUCCESS in 1h 04m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d12b4610b7f940d7af1400d2e33b7129 : SUCCESS in 1h 43m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fed3b7716c4638988ea21d4f6003e9 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3c096a8db6f4bdb9c628c8e508fdbc2 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f815542252cb4df9afc3b2e70929aba9 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed8b3aa740304fb5976518f19dd78c85 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/776a22078a094c0d80b99d240f3f67f3 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13111ea57e240b6a60f9fe1d6404382 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70c240216f2640e29b4471aef950aa30 : SUCCESS in 3m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/255609e5db28457fb06c36bcafa596e2 : SUCCESS in 2m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f852c944eb1420bb2a2de05b25961b6 : FAILURE in 21m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a4c8d2e326294e92b4ff3e0cabbf96a5 : FAILURE in 26m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ed57c4a22984f37821c2dbab906ddf4 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e4c26488dead4de79f2222a7f9f49616 : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d0c99aa66458d005f488d0c3ccfb9db7edce78f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"date":"2025-07-03 14:08:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c2c0c0fd431066ace0100d14acae0ba6f1a2994e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:13:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8c2b8a1630e441ab79656be56fb61a4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1896b49b61c42a28d5c5c21a59a7d9a : FAILURE in 12m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7ed2a456a46a4e14acd476c8f0847af1 : FAILURE in 18m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4eaac667d9fe4a9ca0b2a1fdc67cba53 : SUCCESS in 42m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d9c55593829a492abe64393fab3770e3 : SUCCESS in 1h 03m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9ab7795afa4a73800d219013acad9a : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6aa7f64812d041d4ae338534cd1274e3 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb4539c7404f4198bd5dbd9046ab63f7 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a93a74a29e4f4046abac9299c5e3c009 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3e9007faccb49d0bde5e488869d9256 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c9fbbf040d4d2f8a51e162ea183b02 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c0964f957534d6ebb67f2db4704dcdc : SUCCESS in 2m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b5044d59272b4f48a433f3a2bdf54d12 : SUCCESS in 1m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4e323225971a4ab791af8c1e5dac04e6 : FAILURE in 19m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3fd071ecffb145d9a4cc8240375cc8bb : FAILURE in 13m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c4b36cfc52024f4f9f1a47c3a5298a0c : SUCCESS in 42m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ffa0b82bc7e343e183fc515b55501d12 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":3},{"id":"467f69e23162adfabc2fc0131ae694e5b5b68119","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 18:19:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c988c36befb4186853667c7be567a63\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/899fa02c0994489fa6f5573e9e169aaa : FAILURE in 31m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dc6de5a7d2ab4ce1925a69e25b0bd3c6 : SUCCESS in 1h 54m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/746fe47aae424453a2e908cb0dde5297 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da32f7bf326649a495fd2e2ac8267bf7 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6246c7d16edb41348445930ad8a14c86 : SUCCESS in 8m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/af00739825de469ab2e1bf4d2be81485 : FAILURE in 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"360469e139452e6663bb14fd27791328d2f17bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"date":"2025-07-04 08:40:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b832b04ea2a4b11a6805b0ad78748946cb4fe383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:27:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48d98949aa4a4b6b8a0ee3699aa6fd44\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4cea8fe8b00d42099b8dbbbedc9d8d0f : FAILURE in 12m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/52c022f088364896b367e6df6edc8280 : FAILURE in 31m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/99cec2cf5793417f9529fe17fcdec0ba : SUCCESS in 35m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7ceea37f3a0f45d7a9da0198e500613c : SUCCESS in 1h 38m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c9007011cd4834a63b8862808febad : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/683ab2d882fb413981b238ecd7076837 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5c4ec80eb4d4305b17b8808ea39c26d : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b0ea0935b614126b36e8698a8888af5 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f98432d35a9b473ba48ce6e5b167e87c : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb7768ee44d422791f6dd005fa9f42d : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96820cdce94345868c34bca8f214a135 : SUCCESS in 3m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/73fd9c7cd91e4cbcb70a81038cba12d2 : SUCCESS in 4m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ce6ef89d018645e188fd8eafcaad8dc7 : FAILURE in 21m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e3afa4ae93494bb58741731709353664 : FAILURE in 24m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/847cf18898eb446a9595641f977131c3 : SUCCESS in 35m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c062dd6a8d8a4ec9a652c4daa284efe0 : SUCCESS in 1h 45m 14s","accounts_in_message":[],"_revision_number":4},{"id":"d2e33676ffe54a1c867f22aae12ba57d68e0bb7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 11:11:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d01aafe2b84a4dc69acb30c7e8311eb2\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8623589bfa6d4e14bc7857d774cb54cb : FAILURE in 1h 06m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/22f4c4a0e5b14278a867e88838a7d235 : SUCCESS in 1h 53m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e4d17c6606e430088731c62a37313dc : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8396dd613b7d479c8ae9b3c77bef2c81 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e7b591e5ef144ceb89459094ee345e89 : SUCCESS in 5m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c7b068c64077481a88c9406904464205 : FAILURE in 50m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9be35e00286696d11616b313d7b14bb26df6d230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"date":"2025-07-07 12:34:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5d456b92a582e0f31b50ced7a4a644fc1869bda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:13:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d7cfc5b47e942f2a080a5affca20863\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c1f53e57920940598dad28c275306538 : SUCCESS in 20m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84825a73d3d64149811c84ee739d7598 : SUCCESS in 23m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eb45110e027940ef9aa67d73f0185230 : SUCCESS in 34m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9ce670d0d834065aecc4038ed5fe800 : FAILURE in 1h 37m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a3e9ebe2b8484898638e4bbe02eaf1 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fc0b897336f4bb98b5160f19149345a : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ddbb40ae4804c3b9c86771042398d47 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3823d836ab3142c6913b8a09fc2507c8 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4633697515b04768afa87df7c8b64ce3 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8499f078a08433d9400fd1dcd93dc68 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87021d1673844d30bd970c53751ffc68 : SUCCESS in 3m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8517beaf94f749519f5866bcff21f2cf : SUCCESS in 2m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/506c592eb0dc4a8cb23e148899727880 : SUCCESS in 22m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e2c2e92535b0450798575a2337bdb65a : SUCCESS in 20m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3a7f9e3724a446d1966c336b9a95001d : SUCCESS in 1h 10m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ff6170aa780948b3aef9eb5b09d3840e : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":5},{"id":"5bf97975c4b254fc6612b8e57e75ad0871f4a05b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:58:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b220c060efc4c80b56650cb36c63eaa\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f277d25f703f44a297a0f85b1a6f4b89 : SUCCESS in 1h 12m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86015bc2ddbd4d74a292792104bff037 : SUCCESS in 2h 21m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8c294ac9684418c9da2975e1ab70264 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f77bb9ee4ec4e57b0a4f11800494441 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0968c10377f4cb688001ad92bb07aa5 : SUCCESS in 8m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4ff8a7f100aa4f82845396dd0c3c04cf : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e17e0958d49b577a9de1cb704ee197f9284d1bf8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:29:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d339f6d93b8411d6dbb030ab5d9f3c706b8c43b2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:30:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a75fd6f824602c3889bf10dd20fc2189b76620c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:30:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cd55cc51804d3292ed297ae7c7413c61eeebe0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:30:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_37726\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"}],"_revision_number":6},{"id":"44980403c54a8dccb6ee1ee2be70af3cd8e98f23","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:30:35.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"718d0b274984e01d74fa2f7673a6786f6f62d54a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 17:39:03.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b7ed0699727f7992582bbebafc292f41069d9337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 18:09:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a54b46a546140a6b9fb66b542b64418\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b7d558395c814f1f85c7370b982990b2 : SUCCESS in 25m 32s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/73747f69231c4648b9bd714b2f919870 : SUCCESS in 21m 55s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c88d1f2bb06f4fa682e0dff6a94ce1ed : SUCCESS in 29m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/871141f1af4247f28fb015850a2cd655 : SUCCESS in 18m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/06b3198fb7d842989f57f6c284660e89 : SUCCESS in 36m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97085656b7574b40b37307fe67e29ef5 : SUCCESS in 24m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8232903d39924927805275506bd32065 : NODE_FAILURE Node(set) request 411d5b771b6b414685532766426799ed failed in 0s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/49b96c49ee8349d5af0bffa51a4c48f3 : NODE_FAILURE Node(set) request 9b515aebb4fe43c2a4caeb5f56df81a5 failed in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb286d1ee5634dd1a99e86e372fa7ffb : NODE_FAILURE Node(set) request e0a8ab15b7fe43a8afdc40de87c241a0 failed in 0s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86795e67394c4e9cab7ddc9a26bdc5af : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bcfab9cfae240d08cfbf47e7c108777 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce5ecfc78b914d4ba9fc90edb26123d9 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd1f46441ab4e9e8bcb9eed346c73e6 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c0b7bd18af542b59ab1dd5f3af21cbb : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":6},{"id":"5432b410f6454b87ff79a0610fe7f8ef3caaa60d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 19:59:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025fb01df73d4f309ddc64a3a7a4a84a\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/4ab95addc1614b60928872c334d9ada8 : SUCCESS in 38m 36s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/3184e14edcfc43fa8e1359c1f6074ecc : SUCCESS in 43m 06s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/588e35e9b748437b9b15a5a545df81f9 : SUCCESS in 35m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/74542796cc3d4a40acae911bf64aeedd : SUCCESS in 47m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6272a834795449079991f4ceb81b4c3e : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5fe0e24e276f41b3b6a6a29f7ff44ea0 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b04a85926db94afbb1a7cfaa13c8e76a : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bcde02376744544e7eae91de8587e1f033ff38db","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-28 12:05:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f65b6c9a2cc84793dc8a4e5b13e8f0d010567baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:03:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea15b036abf24e8b9ba747ebc1efd9be\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4de70fe09f1b45bb8dde4d07029897b5 : SUCCESS in 22m 52s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/4b58e338d8d341c9996c9fc9f6585aba : SUCCESS in 24m 16s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/0fa4380213a6401fa883df2da796ca7e : SUCCESS in 26m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/79460f3f48e240c5b0bb150a31f515f1 : SUCCESS in 26m 42s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/60a35f3db2c04dcaab2a1edd56087313 : SUCCESS in 56m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e43879988ce840d5986402575c098573 : SUCCESS in 24m 31s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/02cd2681484b4d65b188d5c8d226f63f : SUCCESS in 15m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e37e5bcbefc44e85ad4fec656957e847 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70160fefc6e0416eb41afd82440332b8 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ccaaecc2b9640438bcd79990cf385dc : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6781ec5127534b59b55bbc1f68430d5b : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f468aa4bded04e6f99760209498ebe13 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befcc2ebf3e1413d847512d1185d807f : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fc39350331c4eb1b7796927e038760a : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":6},{"id":"7732024203cbba96babdcc4ad2d621b8e6008f4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:33:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9efaa130d29491b85de375612b8c279\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2c7d89971cc74e8f9abccdf8ed4de32b : SUCCESS in 32m 19s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/a2dd8a49e9e246459fb1196bcabe14ef : SUCCESS in 40m 22s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4cf5489ef1154a03b2ee9d1de3f4c87a : SUCCESS in 41m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5d83daeeed2a43029aac558fca461074 : SUCCESS in 34m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07e4e9d357dd4dfa95da94a85c6ad5bd : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db1f473cfcc949058ef17ff1aa7bf52d : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1726da50cc924a7fa0d1face22a20952 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"efa0c85d0480619cdce952c9a3202289eb79b551","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-28 14:36:05.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9ded9c31376da22660b98208a07ff9037ac5635e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-28 14:48:49.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"dae9fabd9f2c6acc6fc996c662adca822dab3f58","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-28 14:49:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"689d6a55ab77e037e9ff74c80893aacd7c6c97d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"date":"2025-11-04 11:10:51.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"82b70739ad1ca5862b36ee2c0021c76be0d74dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 12:00:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ae51fb649714efabc5da486edba6747\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c4976c518846465694c6d9c766509596 : SUCCESS in 14m 13s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c5e2698bcc94b7db2de8bf7836a6000 : SUCCESS in 12m 42s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/76bfb51847e04627ae9f950dba40bc8d : SUCCESS in 26m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/01fe33f69a134964a5b622b5cc266347 : SUCCESS in 13m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8cc3cbec1f13475eafa7ed0cd6b436fa : SUCCESS in 48m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b03626b05c68466b8e0f9244ef274ea6 : SUCCESS in 19m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1573d487f994abcb3625fc44c73a86d : SUCCESS in 15m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/383129e592d3477fb5d01285d5472b25 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc23ea4ca2e429ba893b641d9996fc3 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88fcb74de015466e97570f10ad53dde3 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7edebe85ff2c46cba98185141f52307d : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bc44df8cf0b46c7b3eb497cc2f27f7e : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0bede544a74571aad28d29e1dfe917 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4ae58b2f9d6491895262f4524aa1d64 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":7},{"id":"450fa22b572b43ba3349888ab51d365fbacd5370","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 13:22:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66edadad49e942aa880c2aa4dd3a6d33\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1b93ebd5c18444cbb89dd75d51b64e4a : SUCCESS in 27m 57s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/713e535de9d4496ebb0709d92c104ea6 : FAILURE in 38m 31s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/77f79b8596eb4a189f6f465458068eb2 : SUCCESS in 36m 48s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/405c58e65c9f4b3fae4745d8f2e0c375 : SUCCESS in 33m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0627270231442638ee289b7a5ef9f56 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c3e9df4b9991421181fa089ecd041020 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c32c2e3005784a9592f242a43bef9aa9 : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"09e379c21a6264722060403d5c98844e74e08558","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-05 09:30:23.000000000","message":"Patch Set 7: -Review-Priority","accounts_in_message":[],"_revision_number":7},{"id":"2dab8d91763e3fb06c2e0ccc1b29050677258d8f","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-12-17 00:31:53.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9bf60b5e0c089ae18d30d8735c4c1d7ddeaf4be0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-17 13:52:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0f42c7950d1749daa36895e233e9848728f7e4f9","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2025-12-17 18:01:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"48a4e77df240be651e8ba22800a17cc7d10191db","revisions":{"99403d1e93bf316912428a08b7f9af9f5722cceb":{"kind":"REWORK","_number":1,"created":"2025-06-05 11:52:14.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"ref":"refs/changes/46/951846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/1"}}},"commit":{"parents":[{"commit":"b643e8f9fd3e9d8e881e112e53183db743f1e5d3","subject":"Fix ProxySQL rule sanity check to support username-based routing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b643e8f9fd3e9d8e881e112e53183db743f1e5d3"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:52:11.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99403d1e93bf316912428a08b7f9af9f5722cceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99403d1e93bf316912428a08b7f9af9f5722cceb"}]},"branch":"refs/heads/master"},"51a8b8ae8c786850727fef44da8826a51ff586fc":{"kind":"REWORK","_number":2,"created":"2025-07-03 08:53:47.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"ref":"refs/changes/46/951846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/2"}}},"commit":{"parents":[{"commit":"b643e8f9fd3e9d8e881e112e53183db743f1e5d3","subject":"Fix ProxySQL rule sanity check to support username-based routing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b643e8f9fd3e9d8e881e112e53183db743f1e5d3"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-07-03 08:52:40.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51a8b8ae8c786850727fef44da8826a51ff586fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51a8b8ae8c786850727fef44da8826a51ff586fc"}]},"branch":"refs/heads/master"},"ee5a3bd4bb71ac4a5a492936a84629da362f9cdd":{"kind":"REWORK","_number":3,"created":"2025-07-03 14:08:54.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"ref":"refs/changes/46/951846/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/3"}}},"commit":{"parents":[{"commit":"b643e8f9fd3e9d8e881e112e53183db743f1e5d3","subject":"Fix ProxySQL rule sanity check to support username-based routing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b643e8f9fd3e9d8e881e112e53183db743f1e5d3"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-07-03 14:08:10.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee5a3bd4bb71ac4a5a492936a84629da362f9cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee5a3bd4bb71ac4a5a492936a84629da362f9cdd"}]},"branch":"refs/heads/master"},"f9a34b5a005912eefb3d80af4507f11d9d9d639e":{"kind":"REWORK","_number":4,"created":"2025-07-04 08:40:16.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"ref":"refs/changes/46/951846/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/4"}}},"commit":{"parents":[{"commit":"b643e8f9fd3e9d8e881e112e53183db743f1e5d3","subject":"Fix ProxySQL rule sanity check to support username-based routing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b643e8f9fd3e9d8e881e112e53183db743f1e5d3"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-07-04 08:39:47.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9a34b5a005912eefb3d80af4507f11d9d9d639e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9a34b5a005912eefb3d80af4507f11d9d9d639e"}]},"branch":"refs/heads/master"},"93203410ebc989210c9a020ed2b67bbf19e8d478":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-07 12:34:27.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"ref":"refs/changes/46/951846/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/5"}}},"commit":{"parents":[{"commit":"d2d9c9b3b62ef65776a9acc22e5a80eb3181e37c","subject":"Merge \"rabbitmq: fix Erlang pin in Debian/Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2d9c9b3b62ef65776a9acc22e5a80eb3181e37c"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-07-07 12:34:27.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93203410ebc989210c9a020ed2b67bbf19e8d478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93203410ebc989210c9a020ed2b67bbf19e8d478"}]},"branch":"refs/heads/master","description":"Rebase"},"2ad48ed769fa7e85b703bfa90d91e06d8ada926c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-27 17:30:29.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/951846/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/6"}}},"commit":{"parents":[{"commit":"9bc8d4d0816ac89614a703170601046e57f54c57","subject":"horizon: pin pip to \u003c25.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc8d4d0816ac89614a703170601046e57f54c57"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-10-27 17:30:29.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad48ed769fa7e85b703bfa90d91e06d8ada926c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad48ed769fa7e85b703bfa90d91e06d8ada926c"}]},"branch":"refs/heads/master","description":"Rebase"},"48a4e77df240be651e8ba22800a17cc7d10191db":{"kind":"REWORK","_number":7,"created":"2025-11-04 11:10:51.000000000","uploader":{"_account_id":37726,"name":"Marek Buch","display_name":"marekbuch","email":"marek.buch@tietoevry.com","username":"marekbuch"},"ref":"refs/changes/46/951846/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/951846/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/951846/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/951846/7"}}},"commit":{"parents":[{"commit":"b643e8f9fd3e9d8e881e112e53183db743f1e5d3","subject":"Fix ProxySQL rule sanity check to support username-based routing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b643e8f9fd3e9d8e881e112e53183db743f1e5d3"}]}],"author":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-06-05 11:49:41.000000000","tz":0},"committer":{"name":"Marek Buch","email":"marek.buch@tietoevry.com","date":"2025-11-04 11:10:24.000000000","tz":0},"subject":"Add docker login before pushing images","message":"Add docker login before pushing images\n\nThis change adds support for docker login before pushing images.\nIt ensures that authentication is handled automatically\nduring the image push process, improving usability and reducing\nmanual steps required for interacting with registries.\n\nCloses-bug: #1642856\nChange-Id: I85e4d48e93f2a18c097ae7fdf031d6c2f6e8a30d\nSigned-off-by: Marek Buch \u003cmarek.buch@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48a4e77df240be651e8ba22800a17cc7d10191db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48a4e77df240be651e8ba22800a17cc7d10191db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
