)]}'
{"id":"openstack%2Fkolla~951938","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-30 12:22:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-30 13:14:46.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-06-30 13:55:29.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-06-30 13:55:29.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-30 13:55:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9","subject":"Build RabbitMQ 4.1 in stable releases","status":"MERGED","created":"2025-06-06 09:25:08.000000000","updated":"2025-06-30 13:56:28.000000000","submitted":"2025-06-30 13:55:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951938","meta_rev_id":"95d9c0300f128818e2af3a1d701e2ed602dc1444","_number":951938,"virtual_id_number":951938,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 13:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-30 12:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-30 13:14:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-06-30 12:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-30 12:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-06 10:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-06 11:09:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-12 13:50:11.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-12 13:50:11.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-12 13:50:11.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2025-06-18 18:18:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"bb27c9019f6194676462c2d9a1ae5e56d4989092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 09:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67f37bbcf388bd51cbfa07b05465186914d49427","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 09:25:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d5bae98a3d783d3f4116d6f6438a180dbbf9655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 09:27:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f875a7641a0853effab138520e538719ec4aecf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 10:49:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607ababd76b74b17b8d1c5f4bde5184c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/472949831e9e4794a850f1871621b572 : FAILURE in 37m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b443b67713fd4df18bf519ebbd4504e0 : FAILURE in 37m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/945f98b2be184afda6751eaeaa118ed5 : FAILURE in 55m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ab7d0a8d21142deaf4a02558dc7a914 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ffffd6079214381abbbbf75faf5c10a : SUCCESS in 4m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fea5a2d283fe4e60b9704043d90aeefe : FAILURE in 1h 18m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3c523f21d7f64652ac51fe29a74201c7 : FAILURE in 1h 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"410d8b0d4ed04cc03765acee598982066d06c82c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 10:54:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"cb5c1c0eae7f71bbb71d188d128bc84ae096d3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 11:09:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb4d284af05a40789f45b19a94cfe2a3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e6c480e723174dc0955d2130eda591f0 : SUCCESS in 23m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/74c4492f92fe4221ba65d25026499184 : SUCCESS in 1h 12m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fedfd78cc0c04ac2a47f6bdf452438f8 : FAILURE in 22m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e061a44c3e124028b1ed184275b91eb5 : FAILURE in 29m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/658aa4b4b985443e94bcb45a5e908491 : FAILURE in 21m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/802bb743c4d4437098e95d375690c977 : FAILURE in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46e7ae455364a7db1fc8a9bcf0267ae : FAILURE in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9ea6c463a0d4489bd54a43dc2bffacf : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8492e45a99a14c8bba59332d622d78d9 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f5250c1f94443829b441037adb31430 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8242af2690424386bb99354aa17287f7 : SUCCESS in 4m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/82ae77aff6684b7f84d49550a05cf9da : SUCCESS in 1m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9a338895360c4ccda06fab781da2e69f : SUCCESS in 14m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/109da7c99b1d4888aba56badb45d5c60 : SUCCESS in 1h 22m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2c721d2081d0485fa14f660e73adc5c3 : SUCCESS in 1h 03m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/76adfee3d6ee4d309beed9822c31e9c1 : SUCCESS in 1h 26m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/efa0705f6c054707b9ce1537d53a8648 : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/040f862ea6a347d698d25feb2671ec77 : FAILURE in 1h 30m 04s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0a05eac7d46747c1b1bcea79e053b08c : FAILURE in 19m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aac9c316fa4b4e00bc768777d12299f7 : FAILURE in 26m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fa7232ecd05447a2b3af04308a8282b7 : FAILURE in 18m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0934d869dad64ff3b1739cd8c5107209 : FAILURE in 19m 54s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a17e14fa5d334f5bb27113c2709cc636 : FAILURE in 16m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c072386f3ab24976b1f377adb5dffb91 : FAILURE in 17m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f16c83a43f1e66982c573ee29fc353ba82976fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 12:46:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"14b14ba82300a52a638db60a844fda1f8566927b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 13:31:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1029b17da42b48f5ab5ecfd982b6711c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b3fdf5c7fc964ec094d78e2fd96c5197 : FAILURE in 35m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5afd80ab6b764808ae272d9abca46203 : FAILURE in 41m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1b63a1ad977644f5a7601c61f29c5048 : FAILURE in 23m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fd3221236b643319bd239173c787645 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/279ff1de20594acc9addab73374caa41 : SUCCESS in 4m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/59e2a3f5527041aea241614e07edf436 : FAILURE in 31m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f926f8d57f4642299ff2c09f4fa9db3e : FAILURE in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ebdcda797629afac0ce94e4bd284fac4f110f53c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 14:40:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66074b977a9845afafe810efc278cca8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6a0d74a9f9b441ffaddac38c099ef6b4 : SUCCESS in 23m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/468bb8a720c24e75847bce0b709b9075 : SUCCESS in 1h 09m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ea6fc4efa85544ea96f3b9f2fda4469b : FAILURE in 9m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/54d42c1f89cc4dd6bc9b958771b255c3 : FAILURE in 21m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/04f199f59dff414dadec17101b9a32d3 : FAILURE in 17m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6252c63f2dba4f8da3991b9ed127b169 : FAILURE in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee12f5432ef4a4993f991004f537978 : FAILURE in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62a827ce655e4c698d5624246c821987 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc9d5b4786f148529f960c4114def223 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b92b8251b7c34f7eb099dd7ca3278015 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770cd58b30474624b1b26589dd5dd6ca : SUCCESS in 7m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fb2565f04f6042bc91c7c9ac0ff466b4 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/050ee5e217eb4a81b17273f64f48e294 : SUCCESS in 18m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/661258ad91484d91809f5b9a03b706b7 : SUCCESS in 1h 16m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c8f710ae561c4e0a81f7f7883de3cff9 : SUCCESS in 1h 11m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b3834171d092425fbe99f78ba6735d21 : SUCCESS in 49m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/16e64794832242de84c82c3d342f3adf : FAILURE in 1h 36m 24s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ac66f373c1284b12a329d119330d614f : FAILURE in 1h 28m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/db302c3c5a31453e890247b3ccf13846 : FAILURE in 12m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c6c64ef79b514b5c8630e93e402cbc26 : FAILURE in 17m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/26e574868a2e4b699086c97a79729b99 : FAILURE in 20m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a6d34919eca344cba7116000c95e01f3 : FAILURE in 9m 45s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6df57cb203dc4cdebf1bd6038221cc9e : FAILURE in 20m 27s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1e86176e94c24cf499eb539c7a3d75d5 : FAILURE in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89988b31859c0c243b0ceb7defa9cc56563b262f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 14:46:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e135d18f92a7f020211c96183f0d172ef37be22f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 14:57:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ed4746dfffe51541c8def4ac414962c6bf0d09e2","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 15:07:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4acbeb17bb0db08e79daad6049daa20ded32cf95","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 15:58:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"31cf837af2627b3fcfde5b39b58f1feed7670651","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-06 15:59:20.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"92307850ac07538a07efeed43777805b43ebce46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 16:01:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/826d49e1462d4311a490cdbc1bbfae25\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a293bc8435ef483a8373c8121a034591 : FAILURE in 47m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a57e873e8624eeea1e89ea4f8992c6e : SUCCESS in 32m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/242746b9d44a47688e7df6bfb4fb4eab : SUCCESS in 1h 01m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/099fe72400d94ce9a25eb245d1c3c1b1 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81ed086cef704e68968d5d1f3c2d626a : SUCCESS in 3m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7b45548c3c3444f2bca418e50a66efa5 : FAILURE in 42m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2e77165032414748aaa0d73b55d84765 : SUCCESS in 38m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b81f20df11bba10e25af6039ed2167aa821cf905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 17:17:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661ec58463f647578293434a645227cf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/70c857ee6f3544f792a84c72ae9193b7 : SUCCESS in 23m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/09be19fa2b3b46f8bdbb6de01914b78b : SUCCESS in 58m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2d0ef3c9410542938c3a7a4e93feec06 : SUCCESS in 17m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a0c84053ec934860b4d5ae0e45aa9a33 : SUCCESS in 26m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/062cde1dc19a4f44b6de9d19c5a968d7 : SUCCESS in 55m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4316b2bdfb8f42b49d0d6c3f7164867f : SUCCESS in 1h 47m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4db99893f294498ca8ccc370837081e3 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9406b3d372b04c8e9c1f1e72a8b32997 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40fbc05297594a15ab582709cee81d4a : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f7f002f02345638b24f00a2b542a2e : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e2c9899fea4a7192e473a0de94902d : SUCCESS in 6m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f3c67e3da7d846f684a6881148f527ba : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/24c925c2152a43819f37cad74ac7c0ca : SUCCESS in 18m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c2781fb1aa514a85b00529f54d98b5cb : SUCCESS in 41m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4cad58a87da745eb907083d2bbeef811 : SUCCESS in 1h 33m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ccbc76a211b043c7816fcbe6b2afc500 : SUCCESS in 1h 43m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/938d2d24bf1a47e584e77798a9411f17 : SUCCESS in 1h 55m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9c73506b19e944f49d99b4befcb3a4e3 : FAILURE in 1h 27m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ced7fcf0063b4d5ba66ccdd2cbcbe94a : SUCCESS in 22m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/01caf8d9a549416095a64764f48a9a7c : SUCCESS in 26m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/28dc20edbd9347b19ef4bb611870e3d9 : SUCCESS in 59m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f58e603229c49f4be044692281cd978 : SUCCESS in 50m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/201670263cef4fd88ec517e7f13069ae : SUCCESS in 1h 50m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c8e6b17bdc0a432a9fd2c545fc0a7b49 : FAILURE in 1h 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"08641ddc49583ccd547a3550b687e10481243044","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-18 14:06:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aa84e0825727df9d03b067822d53a0e37c44cbfb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 07:43:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"973148c31bfe9db5f211d496bffaaa259980bfec","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-23 08:59:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8b50274c5e215269bd2b9455890735f920c3b83f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 10:00:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46fab611ca724987b70419aa2bcd9fc7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5cdb20a750ef421ea3371c74ecb9ccdb : FAILURE in 1h 03m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f40a25f1d21246ac8ab029c41ace4959 : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f2f4eae77fe4425c9072274831ab1598 : FAILURE in 2h 14m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e350648e7b224079b89eff796123f6d6 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4bf095d4d5a647499a618fea5aa7ac1a : SUCCESS in 4m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0e256ce2ae754e77a271dc72b9eac591 : FAILURE in 59m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2cf210e83bbb43e5923dec1ff0337f21 : SUCCESS in 1h 00m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b2c7d916688fb3ff466bb40f35e0f8e92356a1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 10:39:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c466cd06516e4318a85ddbef60f9ba84\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5bad8341d4554135aab66471f2b5c5dd : SUCCESS in 32m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7f9fc95b0a024907b2bc65ecc88b3cc2 : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77a587b46044487b86991352388f0d94 : SUCCESS in 14m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ff8f263d4f384806aa82b9aa7add4829 : SUCCESS in 28m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a32f54b7de14a72822d386032a023b3 : SUCCESS in 1h 06m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf21196893264f3aa557b4882916e584 : SUCCESS in 1h 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6ff8bcfa6a468ca5f611387c51628c : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/335bd60e7605455d91fe0e0e50cab036 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d4f303298bb4bddb4501a95acfd9ec6 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efbc9544962d43aaa55354287eaab7ef : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25affba0214641ff8eb223f020ededab : SUCCESS in 8m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1deba9c08487498098250d9b3935d1f4 : SUCCESS in 2m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8dd120e78ac4dc386e65d6d4f6def2c : SUCCESS in 25m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/06d36f6775834cbeb4cda6c297b8e26a : SUCCESS in 58m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5af58c8a9b38481e9fc3b5b8c08b2d16 : SUCCESS in 1h 05m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3626abc3770e4541b0edf611a78982e2 : SUCCESS in 1h 23m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/671feeafa1884f87b7516786ac6edfa9 : FAILURE in 1h 49m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6fc0e843fb3041d9be521c59bf980849 : FAILURE in 1h 09m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/da0809cd5bf34bf6a2823254a5219132 : SUCCESS in 22m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4e84e9c976a74752ab19dd4557ecb667 : SUCCESS in 28m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/056c7d06c64544519cf60d2b20e2e966 : SUCCESS in 1h 10m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cd4174535a9840838b96e794fb29d05c : SUCCESS in 1h 32m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/23b3dd2e8ceb4cf095faa20825ece885 : SUCCESS in 1h 53m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/99858b066b0b472fa6246815c940e5ac : FAILURE in 2h 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5edbe96c3df5add329aabb5d04a173a28bac9e39","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 10:59:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a8ee24d2ed4dfab3ebc9b9418f8504f3835dd13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 11:30:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"725edf1152e6c6652b2956b1ae31cb4a8c7df954","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 11:40:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d7927162bb26172b9d4c82d8ad9451c6b70f40c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 12:03:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7ec6ec0af994e10dd648659e5dae3ce67cec3a47","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 12:34:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e7ddf57959f39c246d9327857d31208c9ec27173","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 12:35:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ce1ffbc5dc32b71f0b81a56c32339a9beb92185e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 12:39:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d9408658aacc9644b1226c014ed6310c7663f3f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:35:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd7ea0ef2304e1c83a2aa5f36278019\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b6e46791c7db40949d7eb00b4635a32f : SUCCESS in 25m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a88ce22b78174d4495ba710a34235609 : SUCCESS in 1h 14m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/12e0f19669f24816999a59b7e978c713 : SUCCESS in 20m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49c947905cfc4782a26e270466ff78d4 : SUCCESS in 34m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f66cfe190cf145d5b7934198b8d94540 : SUCCESS in 1h 16m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2f237f3c8b27459db2d7cef3fbbd15a9 : SUCCESS in 1h 46m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb59a948c784b93ba2ba307a042d794 : SUCCESS in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f9d58033b746bf98b5bb821c21bd1f : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14be6af97a6045cc9a4db836f1e2b144 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c628aa442e14e2f9f4182685e4119c2 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fccd37378e104334b86dc5d8e2ed6012 : SUCCESS in 5m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5dfe6ec076f64d03a49722e368a558d4 : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d08e07aaa9964b8994914249c1da9271 : SUCCESS in 35m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/994cfc0d07be4adc8e5ea2e9c96be4d8 : SUCCESS in 58m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/224e405791b6436093ed816e25aa86b4 : SUCCESS in 1h 15m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/42b26e0d937d4f12ba80c03fd3acbf42 : SUCCESS in 1h 34m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0f5fb0f2ff9042bcad4a138827b3548e : FAILURE in 2h 00m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a198a2d1a16c401293d1135e5291bc61 : FAILURE in 1h 20m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/04c83ce614dc401a82055cbf307b3f9c : SUCCESS in 21m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/85587ef0329f41f1a1c67ec521d88e11 : SUCCESS in 28m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/02c52eae6cd345b198f5a898f8d1ef03 : SUCCESS in 1h 15m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4bc8cddc56674bc2a84d466f0d00c1a5 : SUCCESS in 1h 39m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8b6e627420c34876a1a8b0690aaea3e2 : SUCCESS in 2h 07m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/21895cb0b5b3440f893b004f74e61af5 : FAILURE in 2h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d6ca0435a1525ebf2fa11120f927a6af6be5f48a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:46:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/892e4c9cfd8644df8069b514b2259f4f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/041ec759ec654403a14a33ccd2ec2b08 : FAILURE in 56m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/baa6b33a02d84095ab880ae9eaaeb5a5 : SUCCESS in 56m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa615383de5443e5872f6abab8b4d53f : POST_FAILURE in 2h 09m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0df3bdbf07d348c5a3afe793a6b0dfff : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e614e99a8133436d8c04d71740c858de : SUCCESS in 4m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b7839455722441b496228031fa43d157 : FAILURE in 47m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/08c7fe5dda7c4815b8d236733c3bc051 : SUCCESS in 59m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"42c8ab08b14f2ee2e83720527e6c8b6ba7f24a3c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-25 09:57:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"94c57a45d62fcd5937564dcd98064866c76dbf7c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-30 12:22:24.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"cc66af99d7356ac20c384895106bb435b8d450d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-30 12:22:47.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f9f6ad32e62d8c3f4dc533c29894603b9042b944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 12:22:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"37318fc5a864902f72981eafa0a28843372a6b28","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-30 12:23:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"662fc931b55930ae89363a71c4145756fa9793b9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-30 13:14:46.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"afc6d15a5f0a47a0db5eadb9517102adc8df2f98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 13:55:28.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f630f2fa6464bbbb6e25d2ea2ec3f63\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5287d7248ff543f890201e19fb4f7b62 : SUCCESS in 19m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6384c35d502e4630b68f4777ed8ff79a : SUCCESS in 17m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ddc5f4f976634d20b7b52d8e64b2fef0 : SUCCESS in 1h 00m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d7d9ea7f613f4ea29ce4692f4e20c69e : SUCCESS in 1h 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc690e60bc146f98e64e59e2bc60ec9 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ced4c0baa0345fda698b980fe5f5e3d : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce61de17cac74087b15493157070689b : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995f634372a046e2b2e9f20a4f83e3fd : SUCCESS in 4m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b9108e54287541c29012563246ee63bb : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4079b47b734a437f9860589f1e48c15b : SUCCESS in 21m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d911005e09024418934b5edddb1c59df : SUCCESS in 1h 09m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ac54b0def3e64ae4a18b0f73ab860273 : SUCCESS in 1h 04m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6526c26a6a954234a76f597d1b737d42 : SUCCESS in 1h 30m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a98b1ebd6a0e428c97cd393e4d7001ac : SUCCESS in 22m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6db67bbd77d4b7ca7e3c2beef3e7569 : SUCCESS in 16m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/13168642cf1e409d81320dce2a23e89f : SUCCESS in 1h 11m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4dde464892ba4d8badfaa02fad6c2e62 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":7},{"id":"b80a1ac4dca97b182e304bfab4df739fcc43176b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 13:55:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"95d9c0300f128818e2af3a1d701e2ed602dc1444","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 13:56:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34137bd694f435fab6262777e351580\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f36065bd45d48c1a84de3b1acf7a104 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"af9f7cf13f22b1075c078e50cb40648771d20762","revisions":{"1abbd6ee96ed3e7b2473a28728cf3fc155e3ef54":{"kind":"REWORK","_number":1,"created":"2025-06-06 09:25:08.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/38/951938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/1"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1abbd6ee96ed3e7b2473a28728cf3fc155e3ef54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1abbd6ee96ed3e7b2473a28728cf3fc155e3ef54"}]},"branch":"refs/heads/stable/2025.1"},"6a7168f34591f6709cd65395bfae23ac40c615d3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-06 09:27:46.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/38/951938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/2"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:25:38.000000000","tz":0},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a7168f34591f6709cd65395bfae23ac40c615d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a7168f34591f6709cd65395bfae23ac40c615d3"}]},"branch":"refs/heads/stable/2025.1"},"72225b36238cdc78f0138a6dd4b4ef08179dcfb2":{"kind":"REWORK","_number":3,"created":"2025-06-06 12:46:35.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/38/951938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/3"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 12:46:24.000000000","tz":0},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72225b36238cdc78f0138a6dd4b4ef08179dcfb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72225b36238cdc78f0138a6dd4b4ef08179dcfb2"}]},"branch":"refs/heads/stable/2025.1"},"61adb21045b0bd4e2882dfa9a804fcb0affca1ad":{"kind":"REWORK","_number":4,"created":"2025-06-06 14:46:59.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/38/951938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/4"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 14:36:10.000000000","tz":0},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61adb21045b0bd4e2882dfa9a804fcb0affca1ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61adb21045b0bd4e2882dfa9a804fcb0affca1ad"}]},"branch":"refs/heads/stable/2025.1"},"3c717bb6935d80ca12044eef5996bdbf59b8fa36":{"kind":"REWORK","_number":5,"created":"2025-06-06 14:57:22.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/38/951938/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/5"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 14:56:39.000000000","tz":0},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c717bb6935d80ca12044eef5996bdbf59b8fa36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c717bb6935d80ca12044eef5996bdbf59b8fa36"}]},"branch":"refs/heads/stable/2025.1"},"b8dfe261121684261f1386c22f865ff4a412abe7":{"kind":"REWORK","_number":6,"created":"2025-06-23 11:30:51.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/951938/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/6"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-06-23 11:30:41.000000000","tz":120},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8dfe261121684261f1386c22f865ff4a412abe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8dfe261121684261f1386c22f865ff4a412abe7"}]},"branch":"refs/heads/stable/2025.1"},"af9f7cf13f22b1075c078e50cb40648771d20762":{"kind":"REWORK","_number":7,"created":"2025-06-23 12:03:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/951938/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/951938/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/951938/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/951938/7"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-06-23 12:03:40.000000000","tz":120},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af9f7cf13f22b1075c078e50cb40648771d20762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af9f7cf13f22b1075c078e50cb40648771d20762"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
