)]}'
{"id":"openstack%2Fkolla~951944","triplet_id":"openstack%2Fkolla~master~I4dd72b6a493a08a46dfbd6310122f53f803527be","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I4dd72b6a493a08a46dfbd6310122f53f803527be","subject":"WIP: Support per-image user definitions with UID/GID/home/shell overrides","status":"NEW","created":"2025-06-06 11:20:06.000000000","updated":"2025-06-06 12:01:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"68737d7090910b0376d437d401410fff5093c230","_number":951944,"virtual_id_number":951944,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-06 11:55:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-06-06 11:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d10d7c17fe2a590f6f879f381db492d25a071b95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-06-06 11:20:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9011c6fb63bdefc0b0f57126f209ceb432484757","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-06-06 11:20:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ed0327bd82d81d95346898d90743e889bab7710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-06-06 11:25:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f24ca2c7452909fb0076148368f751fa56dc4337","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-06-06 11:27:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ebb8e6565a280b329b5aacd8fae6228ddfd6556","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-06-06 11:33:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"3835a35274ddfb5e98638ad77b09ce1c56f7fec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 11:55:30.000000000","message":"Patch Set 2: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/015f6ef335d042a6a3d7bbf668f0972a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0c69a68e2ee04ab0a173a75ee5c8490f : FAILURE in 6m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ffcfa39ac23c46f3be339dc5d600b167 : FAILURE in 6m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/61841749b3144297a95d2a9c47839864 : FAILURE in 13m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2664ec4843b14da68b6011e3df75eb93 : FAILURE in 9m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13cc3df139d43a0837ec58567c7eccb : FAILURE in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8b20672f2f4411d848cdeaa1115219a : SUCCESS in 1m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ee916455e0340c0a2fa266cff1560fd : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b35291c0fc44951832c4b233dda5ff8 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a92bbcad037e49e4a694d7588dde0e93 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa035721f7841b88177d3c3e44ab1fc : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9ad7b17961d4d9c87a446d069f6ce85 : SUCCESS in 2m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a2afc0a8ca40474190100224a0ff77d9 : SUCCESS in 2m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f2886665713455382f863813fb49902 : FAILURE in 15m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fc6e2b7b52fc491b957af8988c48c0d0 : FAILURE in 8m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/12236f6d031d46c0b630ed7847a07a42 : FAILURE in 19m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c91d195292804028a926dcff531b1401 : FAILURE in 15m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/89de303ab4254d7f90a6cd59a1234aca : FAILURE in 17m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ec2db1a8f9ff45939fa0a6243ee0809a : FAILURE in 18m 16s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7e4a30f799d646a69166dae21844eeb7 : FAILURE in 17m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bc4d0f17f96943c1a202d8464ede1fd3 : FAILURE in 12m 38s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/efd0a57de8b743b0b7d056d1060939ed : FAILURE in 25m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2f934d61d74d4eccbc750785b0943e69 : FAILURE in 23m 00s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7f063acef3a3458dbbc36ee5d16f8744 : FAILURE in 22m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f2e966d01fdc43c085fac439b1c4662e : FAILURE in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68737d7090910b0376d437d401410fff5093c230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 12:01:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06bf29bbdb6847cdb79d6f3d4fd39a83\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f318d4fc4b754becbb9eb1a73485d012 : FAILURE in 31m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ad65514934740fb88e9fa24cd9a7dc9 : FAILURE in 33m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28a686a112fe4ac89bf4c784359c23fd : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f203d3e863ae49fe84de3b1eed8a7878 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c95793ff5bf4b6aa548ae9ad2da9641 : SUCCESS in 5m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/59c11e6b92bf4f4aa7d6f28519239c1b : FAILURE in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd35150bc14b589199b96c6d5fab1e7a8a92b21b","revisions":{"3344f99e263e9e6fa56748ca3d8d9beee7d4fccb":{"kind":"REWORK","_number":1,"created":"2025-06-06 11:20:06.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/951944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/951944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/951944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/951944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/951944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/951944/1"}}},"commit":{"parents":[{"commit":"627c451ebea6bfdb1c4f1008bb460bd9b64f9c6b","subject":"Merge \"Bump proxysql to 3.0.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/627c451ebea6bfdb1c4f1008bb460bd9b64f9c6b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-06-06 11:19:28.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-06-06 11:20:04.000000000","tz":120},"subject":"WIP: Support per-image user definitions with UID/GID/home/shell overrides","message":"WIP: Support per-image user definitions with UID/GID/home/shell overrides\n\nChange-Id: I4dd72b6a493a08a46dfbd6310122f53f803527be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3344f99e263e9e6fa56748ca3d8d9beee7d4fccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3344f99e263e9e6fa56748ca3d8d9beee7d4fccb"}]},"branch":"refs/heads/master"},"cd35150bc14b589199b96c6d5fab1e7a8a92b21b":{"kind":"REWORK","_number":2,"created":"2025-06-06 11:25:58.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/951944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/951944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/951944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/951944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/951944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/951944/2"}}},"commit":{"parents":[{"commit":"627c451ebea6bfdb1c4f1008bb460bd9b64f9c6b","subject":"Merge \"Bump proxysql to 3.0.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/627c451ebea6bfdb1c4f1008bb460bd9b64f9c6b"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-06-06 11:19:28.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-06-06 11:25:48.000000000","tz":120},"subject":"WIP: Support per-image user definitions with UID/GID/home/shell overrides","message":"WIP: Support per-image user definitions with UID/GID/home/shell overrides\n\nChange-Id: I4dd72b6a493a08a46dfbd6310122f53f803527be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd35150bc14b589199b96c6d5fab1e7a8a92b21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd35150bc14b589199b96c6d5fab1e7a8a92b21b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
