)]}'
{"id":"openstack%2Fkolla~952285","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I35b233a499c6bf1861b376fc441dc8081cfaf5d6","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-16 09:53:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35b233a499c6bf1861b376fc441dc8081cfaf5d6","subject":"Remove dibbler from neutron-l3-agent","status":"MERGED","created":"2025-06-10 16:41:59.000000000","updated":"2025-06-16 09:54:42.000000000","submitted":"2025-06-16 09:53:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952285","cherry_pick_of_change":951871,"cherry_pick_of_patch_set":3,"meta_rev_id":"3cdb422748b8aa14042abed465a2f1d86abe2983","_number":952285,"virtual_id_number":952285,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-16 09:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-11 04:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-10 18:30:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-06-16 08:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-06-16 08:09:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-10 16:41:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2025-06-10 18:30:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-10 19:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 08:09:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"4efb99b00d0a3e25a765d038444e20da21745e3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-10 16:41:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6404feb729379cc88e7e9826a3bc9d85ed9524bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-10 18:30:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8df4ea9a32010ab8fffbabd0f43c93a4e6dbddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 19:00:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d6196d4cf74ec193a8c0274b779eec\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/526c29dc855547cb870e483f50930ee8 : SUCCESS in 18m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/417f1758124f4bd5b33e416238f17472 : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bb3947d0b660478e9702e0b27465bd80 : SUCCESS in 19m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e8dbb2977ed64944b23944aeeff25218 : SUCCESS in 14m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/09d1120fe57f4482a26c6807baa08a5a : SUCCESS in 1h 01m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/05db69aa4aac4e07823e7af58a4656ef : SUCCESS in 1h 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4dfe6df71541c7af0f983a70ccfbbc : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08a86fa7165a4a4ea0d34e4c3bed5156 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a5069ee6b114b65b5e57c54310b1d30 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0316eae3ddc48f496729297647ff2f7 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be749cc7b384d86a665ae0fecd92868 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e36e38a144c04b8f93e768334a77cc50 : SUCCESS in 3m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4fb812fb4b2048c28423b2985e565f8d : SUCCESS in 2m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d7f5bb4655cf4583a4bd5d3e7b537eb4 : SUCCESS in 13m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bd58295e353a4a9a9ddb539c7af40f41 : SUCCESS in 1h 17m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/42905e723c9444d3802092a0d772c217 : SUCCESS in 1h 05m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1589a1de78974943810dc7a9da7ea4fa : SUCCESS in 1h 41m 10s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2364257765264ca6bcd0f3ee4c788490 : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4653a68cb2e04860be7fa7dfe483fa51 : SUCCESS in 1h 33m 15s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2d90e6ed67664d3b8be6e577f51bf965 : SUCCESS in 18m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb01b325bb784e7d8bfd22f63938fab9 : SUCCESS in 24m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4b447a96275742b496df613ff44b0d1f : SUCCESS in 1h 05m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9f6e609c47a94f5a8103ddb51ffb8756 : SUCCESS in 1h 38m 06s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d299df9e26584708ae68a38277a3afc7 : FAILURE in 1h 25m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4c95c040c710498a9cb992b550708a71 : SUCCESS in 1h 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22637bfd0131acd5464408435ae561e8890ea189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 23:43:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f416ec153c9c4b4294dd88d3bd418741\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fc16b6044c244ef386d5521144f34862 : FAILURE in 2h 22m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/20ae1aaacc56436f8487a9305a496bbe : SUCCESS in 2h 19m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f5d50fc5fea144779fa59b7cae423d72 : TIMED_OUT in 3h 02m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8b8d8a1e8e3434db5d0048e7dc2bcd5 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/459aa8c57e9c4c99a92bdfdf6f7de031 : SUCCESS in 3m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/58dfe275d0ae42ceb6bcc090158c4021 : FAILURE in 46m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6159207d00ce4c77acdf6dc389e9999b : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9981feb1f290afbfd791e27d5eaeb93eabd88992","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-11 04:42:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dcc646054f2d88f7de8062f2050115bef96d509","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-16 08:09:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae20d1d453c7bb1d50a39b29fa54f40ce846a513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 08:09:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32c60bee78efe8298e783dec67211f690227477e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 09:53:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dde762194cd04a1985eb8432f6ac034c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/851368b2548544fc9226f5c84b9d57e0 : SUCCESS in 17m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d1837e10203b4973ac1d936bb788c57c : SUCCESS in 24m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/545bd9756dbc43cda97143cfd076d558 : SUCCESS in 1h 13m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c5f0f8c54b3847ceb633dacc311a3d53 : SUCCESS in 1h 39m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576b8f8ac80d40efb992f0686387464a : SUCCESS in 12m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50a620bb10b14cc7888e0e12b0b579a9 : SUCCESS in 1m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18beffb258ec4fe38ae8214ef9d5ab2a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3d824e94aa4d16baf6432632e5e615 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdfcbb3a75d842119678ff432a07bde7 : SUCCESS in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/560aa7b6f65845b4aae59d16828c4c57 : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/069e8cab65aa401492b010a92d0613a1 : SUCCESS in 21m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5135ff36c03a4a778da9f4edd33d8c7c : SUCCESS in 1h 15m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9bda60813a744572bc813eb1af7cb3e3 : SUCCESS in 41m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1e003a8b14c045c2888b0128e678ef60 : SUCCESS in 1h 38m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e1cd02e407f4c11ae178e04774af87e : SUCCESS in 27m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0863e3aae9654af3aa92cf339510e861 : SUCCESS in 14m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5b60278b6e6c470bb0e48f2801668a28 : SUCCESS in 1h 10m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a26830699d354ade83f0c9239bd75bb6 : SUCCESS in 1h 39m 29s","accounts_in_message":[],"_revision_number":1},{"id":"15e7575b82cc8211c9825f27bb5070633e59c2b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 09:53:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3cdb422748b8aa14042abed465a2f1d86abe2983","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 09:54:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0410666b5e874a1396e8691669812474\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3834c3a981834ecaa373871497d372b1 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec07d502cf7341b8bc3240c709b35e80 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efc8a3cb1c9e485d7ab47013027b5de8af193e2c","revisions":{"efc8a3cb1c9e485d7ab47013027b5de8af193e2c":{"kind":"REWORK","_number":1,"created":"2025-06-10 16:41:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/952285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/952285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/952285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/952285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/952285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/952285/1"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-05 15:06:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-10 16:41:59.000000000","tz":0},"subject":"Remove dibbler from neutron-l3-agent","message":"Remove dibbler from neutron-l3-agent\n\nThe dibbler-client package has been removed from upstream Rocky repos,\nbreaking neutron-l3-agent container image builds on RL9. Since the\n2025.1 release, package is no longer used in Neutron has been removed\nbecause the project is stale and functionality was untested [1].\n\nThis change removes the package from the image build for all\ndistributions.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/934283\n\nCloses-Bug: #2112575\nChange-Id: I35b233a499c6bf1861b376fc441dc8081cfaf5d6\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit 6a1d383c48eda9fd9c1c55474d29bdba0f06a52f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc8a3cb1c9e485d7ab47013027b5de8af193e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc8a3cb1c9e485d7ab47013027b5de8af193e2c"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
