)]}'
{"id":"openstack%2Fkolla~952331","triplet_id":"openstack%2Fkolla~master~I71003e0d454deaf75f8e2ad2080077a1118af12c","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"38005":{"account":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"last_update":"2026-04-03 07:57:30.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-11 17:55:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I71003e0d454deaf75f8e2ad2080077a1118af12c","subject":"add venv path check when set dev mode","status":"MERGED","created":"2025-06-11 07:54:51.000000000","updated":"2026-04-03 07:58:21.000000000","submitted":"2026-04-03 07:57:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952331","meta_rev_id":"54dc25b92872e1d7df69c318004009234be37d17","_number":952331,"virtual_id_number":952331,"owner":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-03 07:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-01-13 17:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-12 11:49:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-04-03 05:56:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-04-03 05:56:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 09:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-11 09:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-11 10:19:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-06-11 10:19:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-13 17:44:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"197d69cd4b15eaa5947a9e39588b9265167ef6ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"date":"2025-06-11 07:54:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4dbfafd3ccc889b199d3794dc2d21016f5d3933","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:32:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9d6d8d2b39f4d6dab6431fd40eb2a2d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e64bce5535ec46c0ac346d3c84e64a09 : SUCCESS in 1h 33m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dcba926ec3d14dab8928f08e0e707015 : SUCCESS in 1h 13m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cedc57c551d14a1eb32f20b5ea2bba02 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e3e01e5bbb649898cb42aa638bc744d : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/650cf43a58c548bda53c711bc04eeba0 : SUCCESS in 13m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/78da4b9650cd44bab7e280f2dbff22db : SUCCESS in 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9571c8311ff72f92fde422f5e27b41734bfbd73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 09:32:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e194ea5b614e5189bd8a75b4180471\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f998c0ed8fcf4e23b589128fe93d1991 : SUCCESS in 18m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ebc66ed48f174ca6b17ca467e8b9b349 : SUCCESS in 17m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a6f5828cc5a14b03aacaa1614556b1be : SUCCESS in 1h 24m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d4f6fb281d9245478f5b0bf026eaab92 : SUCCESS in 1h 31m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4cb31c04262476581794e80b88270a6 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c012b946311a4d3f9b8583850bfcb824 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df631b68a1ca4e0e98135bc16ae4c2db : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5134d04bbe2d4dc0937ab8ce29cae53b : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b4ac91550b54e99b9876b303d674708 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d56a2c7c52448caa83d5959867127e : SUCCESS in 5m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b6d81f50c1414ccfa9531068f88519ae : SUCCESS in 2m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/27ee8540aa1e4401bfa29dc763146dd5 : SUCCESS in 17m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/868b999c26f4402e89a569e5711b9131 : SUCCESS in 14m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7833cb2b214f457ea75a3abe7dca4e90 : SUCCESS in 1h 05m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a85805484a7648afada14015089d08e7 : SUCCESS in 1h 21m 32s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/ac475ec46e0040d0a35802b23e6829cd : SUCCESS in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d91719eb3359a56f1c16267aef4253d137db1bbe","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 10:19:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0684247f88bec1b439080660840b000c5b5ae452","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 10:19:13.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"26164c2eb9485fc9253a1bc9505ac57ffdf1e2e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"date":"2025-06-11 11:04:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1183b2bbadf6217a82a4ba59fc1c1797edc60b35","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 11:23:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0a350402cd03c9fbecd34dbc972556e92d40fe6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 11:25:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b2d9b5e32652ef4c6e0f9ea27964e7847885533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"date":"2025-06-11 12:27:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"816e97b9573d1f3ae2b2b059cdc12d3e88e6a70b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 15:59:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeffa467a0a44e2c8d383c7af94843d8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6b4ed0ffffbd474d81bc3ec38179b06c : SUCCESS in 21m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/29867141d7be4446aef9051ccd5215de : SUCCESS in 29m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1ded48bb1cbb47539a71aa409cb1d9e5 : SUCCESS in 59m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b4653a1b60a24129ae023a358f07745d : FAILURE in 1h 47m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c340225c3d9044e482dcbfef820ff555 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7d991c0f19c4690a1ae6ad47a629b93 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd1d406c421a48949787c1b5305eb9d2 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a489afc700e4f24a9d58d130857c8d5 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5e32fac55844e6493f6667bde1c203f : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9651d267784461a8199bfc9180ece0 : SUCCESS in 5m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/88d24c83263641bb9dbf7c2aa31f425b : SUCCESS in 2m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/46885887749b4a04846dacb9dec18e0d : SUCCESS in 11m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36ac1460589a4fa28ea892b84c7b30e1 : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d221a5ca70114f4e87972e1b1699d776 : SUCCESS in 1h 01m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c974c60c04ca46c49fb01be391cf56a4 : SUCCESS in 1h 35m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8ba8d170942f4ecb84abe35958cabc6f : SUCCESS in 1h 54m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8ff81574336c45a8af2291391f991580 : FAILURE in 2h 53m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0953b25c8086494f8c16241285a7d940 : SUCCESS in 33m 51s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/697cf14d10134ba6ba172d4eb9820714 : SUCCESS in 24m 20s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9128db68efcf472db6e24d9c6f0a65c5 : SUCCESS in 55m 31s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/eb48f9ca86444eaf920a46fe6bff4270 : FAILURE in 58m 19s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0bd5d87269c8401682c3e71c822cc943 : FAILURE in 1h 09m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9c354544511d4117bf0c7f9a63e121e2 : SUCCESS in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"989e0355e860c46c4eb9d098341d47b96a75b4a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 17:37:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f34c9d0d04401d9eee52c3eaaea761\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e1f4d2fe3847498c891dc74cd35f2da4 : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5dbb1350ecfb478f8f2d4eb4a4c8cfe8 : FAILURE in 2h 02m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10ff70a9517b4b4f8e40899299975ac9 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4d54061eae34aa090174a5404b71f82 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4bf87843b32f41dbb8bd340e6a743b05 : SUCCESS in 9m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/348318b9f3c14c9b9a5b3fe59441a904 : SUCCESS in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a96bb3d8f9049937b4b33983fac05f9a7b0bbb83","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 17:55:04.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"de6cc67ca15458f461381b12e641a4a6908e4f6c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 17:55:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"78ae5c3f60523c20e29b48dabce25a9c4b90a4c1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 17:57:30.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b48211af68627a32e95458d417510f3b2bcc588f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"date":"2025-06-12 01:58:08.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"11767bee80f384dffe077e84da9c97826ed9c1da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 03:48:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3677ff239a94082a504bc8136a45e85\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8203109a65e24fb0b981f9e647c3c437 : SUCCESS in 40m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1aecb91baccd4a5b826fe5a6d6b75182 : SUCCESS in 1h 48m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/561ff3400f524bf6b193ee881182f982 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e21ed7f233654d5681b011dd48b134ae : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f7b37ace8074c588d65733a7474e67f : SUCCESS in 9m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c70f2c89bf8c438cbec2f7ecffbe945c : SUCCESS in 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9e37bee095615503964370ec0d16a07bcea1e845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 04:48:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6598bab8c615454fb262d2b39bf2fe28\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f3ca8b4ea8c840d3a9d26784bbb57240 : SUCCESS in 15m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/68f48786ffb948a5880e8d89efd5cec1 : SUCCESS in 27m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3a1074bf3fe240a89ed05559a1ad5216 : SUCCESS in 59m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f1f1eef63fce4e968dea81e02429ed4b : FAILURE in 1h 38m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f392cc37174af5bf6bb6182e15728d : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/755fbc58aaff404a9789e58c363ef498 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/598544bbb77648379d229cf010814eab : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7633b308f9e74c1eb02743ad68a224f7 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0a23c32f0294449a1705b169f590a91 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574412e03ebc46e09d81f26786fe7e72 : SUCCESS in 5m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7330106b21ac438eace4a23c02d21307 : SUCCESS in 2m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/710a3da156e94054ac11baeca0c49c84 : SUCCESS in 22m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/eb9825c631f440ab84acb8717b3c7a01 : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/29cc7cd7a8ae4228bcde2877c61463b6 : SUCCESS in 1h 10m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/033bdc4d971b42aeadec493f9c8b60b9 : SUCCESS in 1h 26m 33s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d9212d0925b64fd0a3c0e930dfe0a28f : SUCCESS in 2h 05m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/229375e838b74545ae4a6cee9796c4e4 : FAILURE in 2h 44m 49s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c2c777ad264c43529c640d721fcdd20b : SUCCESS in 43m 19s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a5ef16cbd1a34a03a5656c58ee4831ac : SUCCESS in 24m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a5369c7c17c94e418659e7a67a5587b8 : SUCCESS in 53m 00s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0317d927eb5d4b00803dac6024a162cf : FAILURE in 1h 10m 57s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1f6b910d9be3413993c576dcccd35c13 : FAILURE in 1h 11m 14s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/21572e7073ee4af78177d01d925a0a2a : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"715c0fe8ccc3872cbd9dc050549e488cc045df4b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-12 11:49:13.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"437160aaddb3465acadb5919037f00d2a3414093","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-13 17:44:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"57cdac7c96d9a7f6dba7df43fc1dfa33207c2a0e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-03 05:56:29.000000000","message":"Patch Set 4: -Workflow Backport-Candidate-1","accounts_in_message":[],"_revision_number":4},{"id":"09884e35b6279cde361c43f43e1b327ed1f3367e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-03 05:56:32.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7255b525a89614f279039a5da8b6f97a8e34383c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 05:56:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"93da4deb5323e7a30d19ffdcea6c087e487538a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:57:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1859637e582f459eb1839b4397ec80b5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7d5d4e37914d4a9288442900d52da05f : SUCCESS in 1h 19m 21s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/dee51b2826474ccd9c7b75a4ab7255ba : SUCCESS in 19m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/62cdb94b671e49c89485b2a15008dea9 : SUCCESS in 1h 04m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/85076e8da64844b09aa72c6b8b3305e6 : SUCCESS in 1h 22m 27s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/924bb07fe27d4538a4093a43d5634b9f : SUCCESS in 56m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62711934662041aa987e430f1538b145 : SUCCESS in 29m 37s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/dd7076e3e26e432ca465f779d4e2978b : SUCCESS in 1h 24m 13s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/f677967a205d4b1ebb9d43d7f9a39577 : SUCCESS in 1h 41m 07s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3cd142ee087f4dbda54eb0dddfce7f60 : SUCCESS in 1h 30m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/61c49ca1222d47aaa7ff8a5f5f3d74ca : SUCCESS in 1h 59m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/721e4eccaabc4f738f1b7b9072b637a2 : SUCCESS in 1h 25m 46s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8d8dd3ffd0cc4146a785f841c9f8bceb : SUCCESS in 1h 56m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2f7a281def794b8ba0b03aef2c11e28c : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc92095cd61413b913786de17ba553d : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/503cb4cdd0554fc6b64f9c15ad39e3ae : SUCCESS in 1m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1ab3b4df73047e99dca75cdf9ea2ab3 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217b08e1ff0640abb12e99f74c1babce : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":4},{"id":"76f46251f4e471014a818376e4a72ea94eea3e12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:57:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"54dc25b92872e1d7df69c318004009234be37d17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:58:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6595a248792e462c98634291f7a25b30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be621c7f2908410b9855ea9ec5c3e33b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cf40e03d9e2bf40e1e55d3f0c42dd5f0c56c49ac","revisions":{"0f14d541cff38481e0397ba2e331eca725ecb9a4":{"kind":"REWORK","_number":1,"created":"2025-06-11 07:54:51.000000000","uploader":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"ref":"refs/changes/31/952331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/952331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/952331/1"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"committer":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"subject":"fix nova-libvirt container exit when kolla-ansible set dev mode","message":"fix nova-libvirt container exit when kolla-ansible set dev mode\n\nCloses-Bug: #2113874\nChange-Id: I71003e0d454deaf75f8e2ad2080077a1118af12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f14d541cff38481e0397ba2e331eca725ecb9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f14d541cff38481e0397ba2e331eca725ecb9a4"}]},"branch":"refs/heads/master"},"791f035dee6589261bfe2d00a7cbf982c0f9a1b4":{"kind":"REWORK","_number":2,"created":"2025-06-11 11:04:49.000000000","uploader":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"ref":"refs/changes/31/952331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/952331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/952331/2"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"committer":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 11:03:23.000000000","tz":480},"subject":"fix nova-libvirt container exit when kolla-ansible set dev mode","message":"fix nova-libvirt container exit when kolla-ansible set dev mode\n\nCloses-Bug: #2113874\nChange-Id: I71003e0d454deaf75f8e2ad2080077a1118af12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/791f035dee6589261bfe2d00a7cbf982c0f9a1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/791f035dee6589261bfe2d00a7cbf982c0f9a1b4"}]},"branch":"refs/heads/master"},"dbf9eeb281e33f9dc7aed290ffb356bf016f5e1d":{"kind":"REWORK","_number":3,"created":"2025-06-11 12:27:21.000000000","uploader":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"ref":"refs/changes/31/952331/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/952331/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/952331/3"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"committer":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 12:24:36.000000000","tz":480},"subject":"fix nova-libvirt container exit when kolla-ansible set dev mode","message":"fix nova-libvirt container exit when kolla-ansible set dev mode\n\nCloses-Bug: #2113874\nChange-Id: I71003e0d454deaf75f8e2ad2080077a1118af12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbf9eeb281e33f9dc7aed290ffb356bf016f5e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbf9eeb281e33f9dc7aed290ffb356bf016f5e1d"}]},"branch":"refs/heads/master"},"cf40e03d9e2bf40e1e55d3f0c42dd5f0c56c49ac":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-06-12 01:58:08.000000000","uploader":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"ref":"refs/changes/31/952331/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/952331/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/952331/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/952331/4"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"committer":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-12 01:57:30.000000000","tz":480},"subject":"add venv path check when set dev mode","message":"add venv path check when set dev mode\n\nCloses-Bug: #2113874\nChange-Id: I71003e0d454deaf75f8e2ad2080077a1118af12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf40e03d9e2bf40e1e55d3f0c42dd5f0c56c49ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf40e03d9e2bf40e1e55d3f0c42dd5f0c56c49ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
