)]}'
{"id":"openstack%2Fkolla~952340","triplet_id":"openstack%2Fkolla~master~I4b1b54983144f5514a6703c255d2cb4fa052a0fb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-23 12:08:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-11 17:48:21.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-06-23 13:55:19.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-06-23 08:43:14.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I4b1b54983144f5514a6703c255d2cb4fa052a0fb","subject":"rabbitmq: Improve healthcheck script output","status":"MERGED","created":"2025-06-11 09:00:52.000000000","updated":"2025-06-23 13:56:16.000000000","submitted":"2025-06-23 13:55:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952340","meta_rev_id":"c2e57b0f4f29458e4ed0a6b10507c5cf86aa4906","_number":952340,"virtual_id_number":952340,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-23 13:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-23 12:08:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-11 17:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-06-23 08:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-11 16:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-23 12:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-06-23 12:08:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-23 12:08:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 09:07:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-11 10:22:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-11 10:59:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-11 11:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-11 12:49:25.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-06-11 16:00:32.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"038c75ec7851271ef3109646105dfd95a8cdbaef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-11 09:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"806f0204fb8a0678eee347bef6f145fd518c79be","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-11 09:07:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ccfb5fb80d797f23bbc670b36f3fc9670bb4410c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-11 09:51:04.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"bf8881d506e388898953e7664d4731c0faaf9b9e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 10:22:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4f74e6b47cf2431b3bff789e8421d666de59e1da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 10:59:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc956c1343447aaa6e8e8dcf38d1e91\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/116375953ee14cdaba88508a97cc9b79 : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f308fba3f254e28b2e2aff0d7e18972 : SUCCESS in 1h 11m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fda4dedb765c410a8f60729b16e3d07e : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11c5d8d0a2624bcfa7edd7851c103be0 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a29f1dbe4fc940fc93884100084f211b : SUCCESS in 12m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/19188ea39fb74a16b51788f87e856b06 : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd4f9386a793d2d63785aa330a5fb2a664d2ad9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 11:50:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf1470f271ad41698d505c6bca1183a5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a631d8ff50424f31acd2bdcb23c9c662 : SUCCESS in 15m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2dbb86b96c9c4b88953d864fbef1afdc : SUCCESS in 18m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d8fffe40ae7f46c1b60001ec428c5d8e : SUCCESS in 57m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6af794bde934eb1bae8931dea15ac5e : SUCCESS in 1h 24m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e38d373cf9497fa2f93b4e6cb0ef21 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a973cbcd96bb4a058b162283a2194e06 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/101fb59869724305a29642095a80228c : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bed87eebc1784095a0b5ee71c661b9d0 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e13389de3a0d4205892e2b5bde73011e : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df09c145264c417287b151f5c8e2b09f : SUCCESS in 6m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5923a2aadfea4c309289535ab6a33fca : SUCCESS in 3m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7eb6da0a34c74bcf99375398a5f31e36 : SUCCESS in 19m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e5c807d6b44347daab61069ec387d8ed : SUCCESS in 14m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/76391a90e4c24497bfea13d652b0c81f : SUCCESS in 1h 09m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3cb6f5deb9754055bb26d0788722eeb2 : SUCCESS in 1h 22m 49s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2dba1d5a0fe940038ed1e8b835e8feb2 : SUCCESS in 2h 01m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ada0f1f2c1864f52ae95290447719587 : FAILURE in 2h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87615e32252d707e83b2a6371b41bc657d43476f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-11 12:49:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"497157b93827800d00301d0062f02ad1452b1d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-11 12:54:24.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f17b71ad186a4caf50f22001262cef0b1e2a27e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-11 12:56:22.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4cd4902ab4e5eaca37acf68d5561a767e4bed75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-11 13:25:43.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fcc0c27285b648e8eb837ad0426e1ffe01c4adf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 15:55:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc47c8f72a14bd3b22bac45520520fa\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4e1b5a8e46ee40d7879a301a465c01cf : SUCCESS in 18m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ae927aa2c93444c9bb23fe4a44a76daf : SUCCESS in 20m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/67c1c5f3c63041cfb86fdc196ddf780a : SUCCESS in 39m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/52abfb3ba92a498da27a7128cf099e4f : FAILURE in 1h 48m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985590c0361f43a4b0b5da2de6cf3377 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2c591cc152b4976b00416a47ddb73ce : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe1ca3ebd7154e7db14723dae4fa4e15 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db2e02fa837a482f884cbba5ebf1ecee : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8226c2b0fc71474fb77ccd5abf346227 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c7115a156047069623b4af95eade54 : SUCCESS in 8m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/30eb403d9a0c40328397247b0e6bf2db : SUCCESS in 2m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ee11edc67aa4452a18bfc5a281b78f6 : SUCCESS in 11m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5780d6742ff4498b8e2f300cab685ce7 : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d5e34769357b498aaf855ddbe5da1a67 : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f4d5846cb1f94c529bd9744e06bd17d1 : SUCCESS in 1h 44m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/931ba8d66edc45ba9b05a33c8f9c97f2 : SUCCESS in 1h 04m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/55ba7778afc343fc9da85c9f9f6dd1db : FAILURE in 2h 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"03efe80c611a30576b28f1ed02bbce6b961118fd","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-11 16:00:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"33ec18631d681a3691354fdc6f0d7344d44c1ab7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 17:48:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a04d94ae711839e90d01dde7096537978cbcd790","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 18:55:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8e88cc530604b22a24204ebb55e4b7f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a668ae7397d74f318dba8345b3382dcb : SUCCESS in 38m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/55d38551f92c4c4084fc740da2511f1b : SUCCESS in 1h 12m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12fc61c44daa40a1b89a5b7c4116cc52 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a247476b11594f3f9f826d9b65292578 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9746b21f2d674b15aeaab95b31a31371 : SUCCESS in 8m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/81a0c19c9b9e49f4b141697888cbdc03 : SUCCESS in 1h 09m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54db88b10a8062be039c93a13971f9d94b6cec36","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-23 08:43:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"94db7ce3630d61c388fc0230d8eecc4bf7757c40","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 12:08:53.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95a4aafacffca74b460195d88f2da52768e06d09","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 12:08:59.000000000","message":"Patch Set 3: -Workflow Backport-Candidate-1","accounts_in_message":[],"_revision_number":3},{"id":"178eab480bb684e4379861e485ddf079390321d0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 12:09:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"50f1d6c41bd37292e10daad86453491e308fcffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:09:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ce73edd5182637f2e999293a6625fc711413736","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 13:55:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e88437be9ca431a991915a4205d866c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8472e3807062499d934fd5b0cba660dc : SUCCESS in 18m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d6994cd6d8f44129858dcb69e3c2f0fc : SUCCESS in 30m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/43f1fa34aa4a4768bd61d851feb92c8d : SUCCESS in 1h 07m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a718436cc17a446080c4aee753d90b7b : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/215c4753be144e16b6bd83cf2af7011a : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/907de76640144a99b812dc5d99eafa08 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582aebaa8b58495191ed8ddfa1da5c78 : SUCCESS in 6m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/50ca05ea9920456494796c4b53786da3 : SUCCESS in 3m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/508eb98a96d04f8d9cae1263889e6a74 : SUCCESS in 17m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a5e7fe7a65bc43f8ade7db745e04a980 : SUCCESS in 21m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b8825813c47e42f7bcf131d002d032f1 : SUCCESS in 1h 24m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7c0871ae12a84dad90e16dd633f6f88d : SUCCESS in 1h 38m 47s","accounts_in_message":[],"_revision_number":3},{"id":"1235098d2e2fe9ea062b92fc98ddec1e2ef691f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 13:55:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c2e57b0f4f29458e4ed0a6b10507c5cf86aa4906","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 13:56:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98e3e3fe1d3343079e527857352b05f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ef95b5dc9349c4b14bd362d27649df : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbe23bd107c16d4990b26e9aca5eb888f98cc781","revisions":{"2bcfb1dab7776f33f8b98ca4b6ebf2ce32fff700":{"kind":"REWORK","_number":1,"created":"2025-06-11 09:00:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/952340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/952340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/952340/1"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-11 08:50:28.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-11 09:00:49.000000000","tz":120},"subject":"rabbitmq: Make healthcheck only do ping","message":"rabbitmq: Make healthcheck only do ping\n\nWe\u0027ve seen multiple occassions of all these healthcheck commands\nnot being able to be executed in 30s after the switch to RMQ 4.1.\n\nAnd actually healthchecks should be minimal tests, not something\nthat needs to run for nearly 30 seconds or more.\n\nChange-Id: I4b1b54983144f5514a6703c255d2cb4fa052a0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bcfb1dab7776f33f8b98ca4b6ebf2ce32fff700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bcfb1dab7776f33f8b98ca4b6ebf2ce32fff700"}]},"branch":"refs/heads/master"},"35e4fb95874afe72c4bdc653a143e4126a302699":{"kind":"REWORK","_number":2,"created":"2025-06-11 12:54:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/952340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/952340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/952340/2"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-11 08:50:28.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-11 12:54:17.000000000","tz":120},"subject":"rabbitmq: Make healthcheck only do ping","message":"rabbitmq: Make healthcheck only do ping\n\nWe\u0027ve seen multiple occassions of all these healthcheck commands\nnot being able to be executed in 30s after the switch to RMQ 4.1.\n\nAnd actually healthchecks should be minimal tests, not something\nthat needs to run for nearly 30 seconds or more.\n\nChange-Id: I4b1b54983144f5514a6703c255d2cb4fa052a0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35e4fb95874afe72c4bdc653a143e4126a302699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35e4fb95874afe72c4bdc653a143e4126a302699"}]},"branch":"refs/heads/master"},"bbe23bd107c16d4990b26e9aca5eb888f98cc781":{"kind":"REWORK","_number":3,"created":"2025-06-11 13:25:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/952340/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/952340/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/952340/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/952340/3"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-11 08:50:28.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-11 13:25:39.000000000","tz":0},"subject":"rabbitmq: Improve healthcheck script output","message":"rabbitmq: Improve healthcheck script output\n\nWe\u0027ve seen multiple occassions of all these healthcheck commands\nnot being able to be executed in 30s after the switch to RMQ 4.1.\nLet\u0027s put more information in the Docker/Podman healthcheck logs.\n\nChange-Id: I4b1b54983144f5514a6703c255d2cb4fa052a0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbe23bd107c16d4990b26e9aca5eb888f98cc781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbe23bd107c16d4990b26e9aca5eb888f98cc781"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
