)]}'
{"id":"openstack%2Fkolla~952416","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I71003e0d454deaf75f8e2ad2080077a1118af12c","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"38005":{"account":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"last_update":"2026-05-15 15:16:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I71003e0d454deaf75f8e2ad2080077a1118af12c","subject":"add venv path check when set dev mode","status":"ABANDONED","created":"2025-06-12 02:06:27.000000000","updated":"2026-05-15 15:16:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bcb606bc6e9c021fde310e36edc92d376f1cbe1b","_number":952416,"virtual_id_number":952416,"owner":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-12 04:57:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2025-06-12 11:47:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 04:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-12 11:47:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-15 15:16:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"a399a7031ac8f8b6d57120c8acc1bbfa864a7b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"date":"2025-06-12 02:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"240772c21e1fd1529680e3394a11bc03d0ed5c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 04:57:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea148a6e35c34afe8c7528e4fffce4de\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fe7b1f250c8f4484830065887b60bd5b : SUCCESS in 23m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/91ab4c165b4848289092b2ac3d68a379 : SUCCESS in 1h 11m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a72a811bb4864536a9b0e4022abb334d : SUCCESS in 19m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49cdf51675034be892f1ef904ed71a2a : SUCCESS in 23m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7945eaf3f9fd4635937389c1670cdba7 : SUCCESS in 58m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cb9cabc4a0154f50bc293c45eaeaf630 : SUCCESS in 1h 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc16bc1901ae43ab81a96070cefd8fcd : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dcfe53aa4dd43ebafc51cebcd3ee436 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6434d3442a1a43a285c0ec7ca977f0d6 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c980481c2a1b4bcab04e4a24ac744b56 : SUCCESS in 2m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa44d7832a70453f89ac6607ef4c1832 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160bc73790354a3da05745b1218605a5 : SUCCESS in 4m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a672387e3dc7465a970dc5adba1ec423 : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e5f911e0bfcc4fb585d7ef960df5050f : SUCCESS in 17m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/81b10337c64448649f7a78d1a7777a48 : SUCCESS in 24m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/28b505fc425745b5ab8b9bea92779c0f : SUCCESS in 56m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/77c16021492346c6b66e66f9d41ae5ee : SUCCESS in 1h 17m 30s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a021dfb18479457bb4dbe87e99794161 : FAILURE in 1h 08m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b862e3efdf954c07b05b7c662dbf0927 : FAILURE in 1h 49m 19s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/bd153ae40cc34e8fabe8eb4ef5d7a3ab : SUCCESS in 1h 00m 05s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/20fbd3a0cab54a1a9773f6a249e45894 : SUCCESS in 1h 12m 15s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7f7a3b2cf2e948aca01237dcbb171a0e : SUCCESS in 26m 14s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/60db6ff63caf406cb046d5983b5ca0f1 : SUCCESS in 1h 02m 53s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/04a0593166f848f5bc3445d1857a578f : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7a8aaa586e5744cd9f9a471a14ae608c : FAILURE in 54m 25s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/47105628f54d4815bb893a2e2d0f297d : SUCCESS in 31m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1eca33d031cb42c7884d706ef43bff71 : SUCCESS in 21m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/64f92f35fb4a4bd99f58da50756b1cd0 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/28deec7383484018b6dd3b723fef478f : SUCCESS in 1h 08m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3d47ebea9a6242da9b20a4a936074ce1 : SUCCESS in 1h 27m 52s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/010dc05d4d624597b39ff7127fa49498 : SUCCESS in 1h 53m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/26e6993a6a0c4edf9a170f30c90fab6d : SUCCESS in 2h 06m 23s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/14e93fb6891445dab5ed6dabec4c7be9 : SUCCESS in 1h 13m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f95bf1b38b5a43b985b326d4e24e0fee : SUCCESS in 1h 24m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3d0fb301429243a5b8818e60089bf045 : FAILURE in 24m 26s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4d867d9f858d48fea07b883eceb404f5 : SUCCESS in 1h 00m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/012b02ffa99f48128411cb90e6cd9cb1 : SUCCESS in 59m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6553b8d51e0a41cba4bfae3ebbc0d03f : SUCCESS in 1h 19m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/94718efa9f1f4125a0db455143a16683 : SUCCESS in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef36f41000fbe4b5cb115fcfe94a8747926c6fbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 05:04:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daced79748524f809f14ff183e0a8321\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f8d02a99d866454d8a01a7c29135e3ef : FAILURE in 40m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/254c754442744f6ba9dbb5df4e304042 : SUCCESS in 50m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e19a81c7a27d49038a39bb5b9973ab48 : SUCCESS in 2h 11m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3398fbf3c0048cbbad1b16b744c0d86 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c3d3b0d7df843e4b803e46da7187fa5 : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c60a9d2893f04b71abae13d951166c93 : FAILURE in 39m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5aa6b56cba5842ed869d93bd0376f2c4 : SUCCESS in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba8a684816d5b50e82563d5bd2daf2446676eece","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-12 11:47:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcb606bc6e9c021fde310e36edc92d376f1cbe1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-15 15:16:16.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/kolla transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5131c9653c70f454f524d58a78ab825d464e0992","revisions":{"5131c9653c70f454f524d58a78ab825d464e0992":{"kind":"REWORK","_number":1,"created":"2025-06-12 02:06:27.000000000","uploader":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"ref":"refs/changes/16/952416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/952416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/952416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/952416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/952416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/952416/1"}}},"commit":{"parents":[{"commit":"cb204aa173b364d3d5021afaa7995a215645791b","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb204aa173b364d3d5021afaa7995a215645791b"}]}],"author":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"committer":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-12 02:05:33.000000000","tz":480},"subject":"add venv path check when set dev mode","message":"add venv path check when set dev mode\n\nCloses-Bug: #2113874\nChange-Id: I71003e0d454deaf75f8e2ad2080077a1118af12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5131c9653c70f454f524d58a78ab825d464e0992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5131c9653c70f454f524d58a78ab825d464e0992"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
