)]}'
{"id":"openstack%2Fkolla~952417","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I71003e0d454deaf75f8e2ad2080077a1118af12c","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{"38005":{"account":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"last_update":"2025-06-12 11:47:41.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I71003e0d454deaf75f8e2ad2080077a1118af12c","subject":"add venv path check when set dev mode","status":"NEW","created":"2025-06-12 02:08:04.000000000","updated":"2025-06-12 11:47:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9f38a12da9ced16f7ae33499b67f7974b91b368d","_number":952417,"virtual_id_number":952417,"owner":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-12 05:03:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2025-06-12 11:47:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 05:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-12 11:47:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"3d4ed510feefb5f4d0b72aa64966d6c57736bb2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"date":"2025-06-12 02:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9af41f3ca715d7630ab001a7ff0260a30c6865c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 05:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f87f8c8de4a4be7a977098dd29712b0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a986ba78ae2a4bc4943cba9de83cf1d3 : SUCCESS in 43m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3ca45e792c67467cbfa367e2f59d15bd : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8423183322641c6a52948bf2e3ad40b : SUCCESS in 18m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eb5c887255764990ab42d97850ab913c : SUCCESS in 28m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d58a4570a3b040b9b128391314f5a249 : SUCCESS in 1h 09m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5460f36ca226403b8f7ac9aced594f26 : SUCCESS in 1h 47m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a9eef1dd994486a5531309478fdd2c : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/064eeeddbf7a41aba32a5232c384c840 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca7e9b2ec1c2464d9a62c19efc6e75f1 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/809d3ffc2f964805956e6f5656e60532 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98bbd34269e46448b53914844481cb4 : SUCCESS in 5m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9c0735f30f484ed0a6f22006660f1244 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7fd047688dbd4f569049247f65329b3f : SUCCESS in 24m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/45cecf3a96ab4acc9f767486e27c585f : SUCCESS in 1h 03m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c690f23ec2b24204a04b4dd6c11b559e : SUCCESS in 1h 18m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5fb16134e64d4a26b57b15138e7784a3 : SUCCESS in 1h 26m 48s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/fb6313ac8b674a4f9fc9c064fb779a30 : SUCCESS in 1h 56m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7441eba4427940e48c43b5c245a5d521 : FAILURE in 1h 14m 22s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8001d5d47d0348a8a4130d2692f8f970 : SUCCESS in 1h 10m 47s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/31149c6904394752972d73c7b3bd81bc : SUCCESS in 26m 24s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7e64f2b9ca4344b099c251b0cf0b69ce : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ef6d7aef998e4c1a813977861b42846f : SUCCESS in 1h 23m 29s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/034ff8eda06948cd9e5d7257dc86486e : SUCCESS in 1h 26m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ce2779af165a496a8b2393acbdc72889 : SUCCESS in 31m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9dfa408b0065483f80edc53d0a892f7f : SUCCESS in 19m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ab6635481c094e9d84f03a896355c5bb : SUCCESS in 34m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/17a2b843987f4b16b0a0420dcdad6257 : SUCCESS in 1h 04m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4a07675fbf754a99ad00b1df44966f5a : SUCCESS in 1h 32m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a80a38ae71a34a06a39aadd02668a758 : SUCCESS in 2h 01m 03s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a2120b1b9101436eb350fcd1ba8a2457 : FAILURE in 2h 30m 06s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7f29defe758e40f4bbabc88e34c52b98 : SUCCESS in 1h 20m 21s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/302b8391db8b438d85b73913e7796af5 : SUCCESS in 26m 11s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/53b0c22c300042b8ab930a1bb3bac2bc : SUCCESS in 52m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9bf90bc48dfa4cd0af843a7151349c65 : SUCCESS in 41m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/47c107e18f1d45f386e084d09615c17a : FAILURE in 1h 26m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/656a8252f0ff4b7a86946dcd2325b187 : SUCCESS in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82412dda131fa51d88b4f99b4166d949c416ff28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 05:20:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a91cfddfe28473c856e6118b16d60b4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0d00fa986a484ee6a3188f28e35857aa : FAILURE in 42m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a52f2ef166264a28b895870a3e411ac7 : SUCCESS in 47m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/343c8d7bfb684d02acb9b8412ddd4cbd : SUCCESS in 1h 55m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/892b69271cee474c90a0c9510a27e447 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82457921fcaf472c8e526ab232a615cc : SUCCESS in 5m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3b2ed419666f4007809d03fdf5e40392 : FAILURE in 1h 03m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f00121229d3d495895f1dcdb08170008 : SUCCESS in 1h 07m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f38a12da9ced16f7ae33499b67f7974b91b368d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-12 11:47:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe16ab16de99e2b6cbb7ab0536f60bb9de9ac1b9","revisions":{"fe16ab16de99e2b6cbb7ab0536f60bb9de9ac1b9":{"kind":"REWORK","_number":1,"created":"2025-06-12 02:08:04.000000000","uploader":{"_account_id":38005,"name":"zhengzongwei","display_name":"zhengzongwei","email":"zhengzongwei@foxmail.com","username":"zhengzongwei"},"ref":"refs/changes/17/952417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/952417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/952417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/952417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/952417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/952417/1"}}},"commit":{"parents":[{"commit":"5f6f9f5df6e3f5d790787d29aea5ae820c759c12","subject":"[mariadb] Fix incremental backup using correct base dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6f9f5df6e3f5d790787d29aea5ae820c759c12"}]}],"author":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-11 07:53:23.000000000","tz":480},"committer":{"name":"zhengzongwei","email":"zhengzongwei@foxmail.com","date":"2025-06-12 02:07:26.000000000","tz":480},"subject":"add venv path check when set dev mode","message":"add venv path check when set dev mode\n\nCloses-Bug: #2113874\nChange-Id: I71003e0d454deaf75f8e2ad2080077a1118af12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe16ab16de99e2b6cbb7ab0536f60bb9de9ac1b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe16ab16de99e2b6cbb7ab0536f60bb9de9ac1b9"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
