)]}'
{"id":"openstack%2Fkolla~952566","triplet_id":"openstack%2Fkolla~master~Icb0e47cf95a67cb4585c71f863e542bc59ff5009","project":"openstack/kolla","branch":"master","topic":"kolla-bug-2115244","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-10 12:11:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-10 14:31:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb0e47cf95a67cb4585c71f863e542bc59ff5009","subject":"ovn: Fix log permissions","status":"MERGED","created":"2025-06-13 10:12:09.000000000","updated":"2025-07-10 14:32:35.000000000","submitted":"2025-07-10 14:31:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952566-kolla-bug-2115244","meta_rev_id":"711637ca0d986f28732c177bfd8284690f879291","_number":952566,"virtual_id_number":952566,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 14:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-10 12:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-10 11:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-07-10 12:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-07-10 12:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-13 11:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 18:23:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-06-24 08:43:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-10 11:51:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"1cc3b2d64e1781cfb6dfcc20d0013ed52877d6be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-13 10:12:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35fc621307a8bbf74f49b2788f5d090f3cec6c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 11:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/603657a37e1f4645b8cd09ff340b9967\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8a74296f036d4c2baf217aa1615c6f27 : SUCCESS in 20m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/854139498e30441ca674a3dcd784c941 : SUCCESS in 29m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22098e1970b44102bc33727fcc92bbe0 : SUCCESS in 1h 11m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4981288f60644ca1b9e2c5947c83015d : SUCCESS in 1h 24m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5354786eefb426b8958a045419df744 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66dfa68d2cbb4b3c80b37fc0fcc25694 : SUCCESS in 2m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2f7fc1ab274444d89c757c741d21f44 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/718fa4aa2ce14c8ba3e7d9cbf60edb6c : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36601112e6384435b4382d1c61568cd5 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1bb6e15055541c081b920ef67f87459 : SUCCESS in 8m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c59da6ad067144a0a75f2aa2d134b64f : SUCCESS in 2m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/36feedd90b824602be8ff37f3aa0803d : SUCCESS in 10m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aff485e865454bf8acc96ffaf25cb2bd : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4f0a37f77c49471292390e8b2bf91154 : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8cdf034d415d4e23b60dfed4985827ca : SUCCESS in 1h 29m 44s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/512d7a27e6f24f9c803ef2c4cc5c5b08 : FAILURE in 55m 25s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/966893a6559d4481a0b005389c63ef2f : FAILURE in 1h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d14399994ac5bccad7dbbf1f0c668b59f2857e8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 12:22:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c91cda4b2cb45ceac138fdfcc75e374\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8197d8c497a941879d05e64906c85705 : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/812bdb98a28d429f9d7c1418b673c871 : FAILURE in 2h 08m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db60da19f68f42dba6ddb72dccc8717b : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f9a3ba2254b84066835d3b3d811a2d60 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f1aa56d57bc94f0298932b4a8edaacd1 : SUCCESS in 5m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cff697cc2559423baee5667914516fb5 : SUCCESS in 1h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"202d70f930b8c4ad18ac37772c8242f17ba9306d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-16 13:42:15.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4f9c14ea13e0b4639aad5664fca8a0ed8cfa6444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 14:55:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2673bc1dcb2a41b6b44299f2e959a586\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dc7b6f74e6a242a68dfb731b4f30fd6e : SUCCESS in 35m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/15dfb4f724534d1b8d9878554272f150 : SUCCESS in 31m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d1d4378c8f34b3ebef19e14213f56e0 : FAILURE in 9m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/71f7857fed0149fe8875dfc0cfb1c5e3 : FAILURE in 9m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d74ad9db344e55899ea10090ed7251 : FAILURE in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97c71a5899dd4ecb9f50f0033219defb : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3fac9dfce1c4211a525b83a443cb737 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efe5dd485248425ab08341de85d9adf6 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3034ea55e3a84d49af433bc25b7ad17f : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0126d288b7468c94220336f1ceddf5 : SUCCESS in 6m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/98d3f998e790407082e548f062242327 : SUCCESS in 2m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f2d8103f23754bbf9a0dbda4ef04ab9e : SUCCESS in 17m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f31aa7108b88442fa3fc4e3d086620e6 : SUCCESS in 33m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/505a8ada43db47ab958ba22a473f9168 : SUCCESS in 1h 11m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/45051285a355411e8552d76142449e9e : FAILURE in 10m 30s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b63d5912701d461792b811d6806ea705 : FAILURE in 11m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2823d10b49ac449c950772a08c528a8f : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d389219e3f8526d17d31ada416bfb8c194aa76e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 15:50:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef558e09428474aa08ea101ed06299c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a7a3296f0cc4c08940a31249c777b24 : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51372afc6e0845f797394a0d65e382f6 : FAILURE in 2h 04m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35cbfda69cc14fdd9d69eabaac029fb9 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9d18ab2de904ef984e1dfd5b8412a66 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ddb547e929764e189d3e15aea7699f87 : SUCCESS in 6m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/df9f301946c94fd6b86f6ecfc412c0f7 : SUCCESS in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c58ddafd21e86549892f4b1ac43cfe26c10501d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-17 12:37:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c7383c30e5b9b76d3ba239f9bf00b8f36f04242","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 13:51:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab6180b73154a76a9bf24e473694a10\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f97cb885da89461daa5fac7f4bdb46b5 : SUCCESS in 40m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02923df8c025420f9c6a986759e9cc47 : SUCCESS in 1h 12m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07c3bba6fc684c139c7372f9b82f24ea : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d2e9b22e1054d6c9b97bb68be0e386d : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a415f53b1adc4a0f8d1acf40f509c179 : SUCCESS in 8m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d267e4a988fe4aa8919a0224e21f8b82 : SUCCESS in 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c07b972ceed6a2af49552353a34f0cd2dbc3d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:10:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d3c37024b7c4e4dac025b97999d07e9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/409c4d4dcfe94b9d87878a05cf9ae7d2 : SUCCESS in 20m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4da8cdbc8f2f423d980e854b3c7f650a : SUCCESS in 28m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64b140c907044b0eab2f878824fd56d8 : FAILURE in 7m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b83a934f245e4ddba44e9cea08b7b05e : SUCCESS in 1h 24m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f220cca71441db90d77f57ff89fb97 : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4a50274149d4c3d9864f719ff44cb21 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac496dbbf71e44d3bd4f431ecef82bab : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cc01910e5164b729d15b0c576e8edab : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f12854fb2a5465bb818e1548ad9e60a : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e2bdbc6d1f4a34a2e20d527c729b6b : SUCCESS in 6m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d20cad3ad2d1472fb4c407352828f426 : SUCCESS in 2m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/24ed6c4f976f42bd87464856fcbbacc2 : SUCCESS in 17m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e0f409dbc4334f38a39dc96fdf35814e : SUCCESS in 28m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6fc21685d6b442c99cee24d37d59d307 : SUCCESS in 1h 09m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/205f45b161e84ecd956c7e7184294032 : SUCCESS in 1h 30m 20s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ae6e0be044d64570975295eb46ad9320 : TIMED_OUT in 2h 31m 48s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/dd33e14ba31c4c4d86508a2c629d5e43 : FAILURE in 54m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"604d3ac3a219f32ff50ab9b87f92b77b02a58d62","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-18 06:42:03.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"80b508741383c777befc439acce251a96d9c1fff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 08:02:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b3898d605424c68baf3641b1a96da58\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/389484fdb14e4403825b54ed07ebd8e9 : SUCCESS in 35m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/85e150ccfc324ff0a0260b281e37454b : SUCCESS in 1h 18m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b870bc5284544f3b9ca1dc5e533a105 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a094a05217a0463887bf249da1d82a3e : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/451f0e4367814efcb117bb50697c16c8 : SUCCESS in 12m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d5f5fbd824a41f6837c40a14b7adfb6 : SUCCESS in 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b19aa01150516fd58525b45016b7fdf7679e2a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 08:11:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/658554c5e19b46c6afb678f5205c6252\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed515a307e8d448f9ffdab446f4f7933 : SUCCESS in 17m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/81d604f8459d4e13b827ccf01bfcd4c2 : SUCCESS in 29m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7628c1c2c11a45db903987d8e9fc43df : SUCCESS in 1h 12m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/736deb1cb0bc47b699b43b2486924959 : SUCCESS in 1h 25m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8eefb213d704558959bb063c349f9c3 : FAILURE in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e350c641dae744b4a29ee4c35a513237 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0b87dbb001e4789803e65b73af6854b : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3734549313b4715b4d069605de42a2e : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632e9c3b4b4d4e7692b5f21ba9207f31 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78a1ce5bec842c1ab89524fae5cebb8 : SUCCESS in 6m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a988a4168a1f4b08a12593fe4024ccc8 : SUCCESS in 3m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fd35fb89a30946d49cce9730c0beb5d3 : SUCCESS in 26m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36111dad4e0f4710bcd86be2a553bdd5 : SUCCESS in 29m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d054b5699dc34593a2bea252278fe6b7 : SUCCESS in 1h 04m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f88c02c98d17437fb0e5c63f1325f5c1 : SUCCESS in 1h 27m 21s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a91b5bfafbb54488b6aba8207cba9961 : SUCCESS in 1h 20m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/93b06c3bcefe41ce968363a80adb7550 : FAILURE in 1h 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5ed906fe3dc1272b0432f100a2eb4875b620f74c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-18 15:57:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c80a6407d794814b6460373c1aa004ec2ee7e594","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-18 15:58:23.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e8367de8338b08ca52e751883ec4134e9588b46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 17:43:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d09c7453c42e42418b2ff9fe4c9ba8c9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ab0a86a5aedf41749cc6fe3b92e441b6 : SUCCESS in 17m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b4b34d7d3cba4be6bc5c51672dc93eec : SUCCESS in 27m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a00aac609d214fcd9534c16568bb1469 : SUCCESS in 58m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8b58152bfe97427b8472153c6b1dfc31 : FAILURE in 1h 26m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbeef958b024a83a99f4d454c6afbf6 : FAILURE in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b4ba7d10b5f4de5b4df575a8fc1a7b7 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/351d3616bd114b03923e1146afc5f6ce : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03a7d48f13f8491e895d0ce6b62181e1 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f028975402f4ef28ac5b7dbe0b2f36f : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3aabd32bfc463993a4a39cdbd99c1d : SUCCESS in 10m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b5d3671398344bbcaca7da93b38adba5 : SUCCESS in 2m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a46cc607b89a42f19e90409f03d91663 : SUCCESS in 24m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/62bba6b4f40f48f485e66cdb0412c509 : SUCCESS in 22m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d6e205009ca844998396414e557069e1 : SUCCESS in 1h 10m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1d10cc9a52b24e7d9b7b0fd7c5ebe89d : SUCCESS in 1h 28m 24s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/12b74034bf72479891c0833d64a05ca8 : FAILURE in 1h 00m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f32de5cfd8ec4c408118fe6ae6b0ee35 : SUCCESS in 1h 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"47654a4bf44d871977899491c85df8e1a059b2b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 18:09:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67f44774e00b4f7ca47fcc549bfa35bc\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d2b59ffce1844017a740d42e76aaee9b : SUCCESS in 1h 03m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/87e2bd7602114069b5b85620155535b4 : SUCCESS in 2h 08m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d913da67691b4686b769a30efecf3d7f : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d02afcf39a064275ad2831b62a9c8ea7 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/07134bd0686a43938e914d5dd43400c1 : SUCCESS in 11m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/71926ebae1fb4541b860839404ed5642 : SUCCESS in 1h 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fcab27e623319986a593df042e76f6e283c2a880","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-18 20:04:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c5cd66dd6efde238524a412939e3911a00687b79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 21:13:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970393ddc6cd4725aaa2441fba4d0cce\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1c93274d3f540cca2ed8f5cbb406575 : SUCCESS in 30m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f50f6891736e4bc683b15183c3d94161 : SUCCESS in 1h 06m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d15b99dff93d4d6d9faaad33563459ba : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/caff77e4729044d780f95fa24d843378 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9ff53222df844cabd5f1a108e495256 : SUCCESS in 6m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b4716b1a7154210a05732436ae6e46e : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a4a7811da9b90e7b44b3e6eb39332dfe96785755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 21:52:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e318c202dc14a67b27588a652e60214\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bb1cafd9c37e4734981e0aaa7338d310 : SUCCESS in 17m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5c71b139831f4c17bbdccdb8b81c6ed3 : SUCCESS in 19m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d4e98e7f438d427a8b6b867451fdbfd0 : SUCCESS in 1h 10m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9046347b85c3497482b1d24a62c8bd10 : SUCCESS in 1h 18m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3dd6792ea7f4cac8b0f0952b185fbb2 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8fcb1d6f978457aa4fb0d05dabaee12 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cee038d8cd2f40178be206f0d76267a7 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63527d2db6de4aad90d6292d28a7062d : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c2a85eb65614f058aff0f0afb57c920 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da45a50bb084fe0933f739554f3c2a6 : SUCCESS in 5m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ff7c1fb71be14c23a09646104924d50c : SUCCESS in 2m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f0fb632c10b247e4b2fc3256702caa7d : SUCCESS in 21m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6ce11d0a5f874bf6a5f4f0b19af74e49 : SUCCESS in 24m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fb34d4f7b68a4362aba06b307aa1e5a7 : SUCCESS in 54m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a238c158788c46768f1d241473b96ca7 : SUCCESS in 1h 44m 57s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5ae051ce743a4f328d28554d72f023ae : FAILURE in 52m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3849716090434532b46759d6a4820b55 : SUCCESS in 1h 42m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1b2905510852b96cc2b297369ef34544a65916a6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 16:33:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"c1200b676fd1227b54561cfc455f6d2454f72f7c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-23 18:23:04.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"59f9056aac74663097878cdc28e14452c9bd8a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-24 07:38:18.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"519abdceb994a550abd18d150b113b09bf5a84fb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-24 07:39:47.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"89cf1d794e2e13d9358c739aa9d22dfeaf2e7278","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-24 07:57:20.000000000","message":"Topic set to kolla-bug-2115244","accounts_in_message":[],"_revision_number":7},{"id":"587e23a2b61cd4904a45533ab6e5c8079c43a279","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-24 08:43:15.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"168ca6808fc9ada3c93f743d09ccb57b453ac575","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 08:57:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2af047b38746878adcb22fdea76690\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/33faeb1885aa49d79c15f7ad14f1c7ba : SUCCESS in 52m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a24bff4d63348c59aa18aa79b074282 : SUCCESS in 1h 17m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/556634470d6e482d9e55be38bc7cfa1e : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7ced0454dd34585b83f31868fc6a3e9 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e42038576dce455ca8655d99d812343e : SUCCESS in 8m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6e406fe867fc4c089235a1b32d3a919b : SUCCESS in 59m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bacfe8af60a5955f03b4b5ffeb62cf1c6c2b4ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 09:23:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a246614fb5649f4b81a458356d67f0d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/85f80dd857e84a5b8196373afd50ce5b : SUCCESS in 19m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1970e71bce404a5ca67350fa51f1a168 : SUCCESS in 26m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/302772cf457e400ebf69f5405a35ea3f : SUCCESS in 57m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d5067dac659a4d9db67e9436f09e33e4 : FAILURE in 1h 43m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa6725cda9842f58f88550a61f8a28f : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a89ab634df214e97854ce9ee4f3fd081 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/250d67cf54f04fec951e540169ffb87b : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90a93581df064cdaba28b190c76e035c : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68a6817a5e1f4a63858ca84c5c754a58 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c4b58f93774d44be6555255f6120b3 : SUCCESS in 6m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c5e48747f9674bdc94aa9cac8752a80e : SUCCESS in 1m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c008d83fab1d4822b2339574e901faff : SUCCESS in 18m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a098b1194264b879d5cb0a6955613e8 : SUCCESS in 21m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/800ad6d163b74199b30cb5ba7b537ccf : SUCCESS in 39m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aa3846503c9344b391b4ffbdd6b88688 : SUCCESS in 55m 48s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e667076aa34a48bfb1f49449f9dd4209 : SUCCESS in 1h 12m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b26ccec035ea4cbf915e4efc38c7443d : SUCCESS in 1h 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"39270c3a81b46380e45c5d6c3e1fe799f672590e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-27 05:48:23.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"12d7e3afd28f36b2448aa98514f80f5b5c491d53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 07:37:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea20e44f56834223910a86d884ed0227\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3cd8b69edc76474e9a98330ae0cd0efb : SUCCESS in 1h 26m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e79bf68986154a9b9456b7f857d9521b : SUCCESS in 1h 49m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5400e11a0b65464e83ee23a41259301b : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d55935830f6f4da3b058d32d7993e4b6 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6daad74678d84ea9a556ec7bce09ba2a : SUCCESS in 9m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/82dd0758b1194206b98db5540b644a7c : SUCCESS in 1h 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"756cb5c65246f65a8786d44d627405fe29c3f57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:21:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121bd1af4e6444d1b9140c12f26e38a4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d037c599a57548709b73513b047cb28c : SUCCESS in 12m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7810bde49264472a903ce7dc5c9d8612 : SUCCESS in 22m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a159ec61aeb943debf8a4bac30ae002b : SUCCESS in 51m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9ca4dfa0b3de4bc3b2f5b5b9efcb9155 : FAILURE in 1h 21m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf718899c79436aa6aa842078030547 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eec6b7cf2e7f4b489f85b6ac51d5847a : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a4933cb6d0f4b15b8db6d5f7e41e6b1 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cf8904e03a2496fbc41f3d766e6047a : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffce05c4f4974771b96005b3bd5d19c1 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170faf5f80ef47118f68ad0607cc2eae : SUCCESS in 7m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c4330856794c471a938c719bf19aa04c : SUCCESS in 2m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8ad4f81296304137becf2845525c9ee4 : SUCCESS in 12m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a7bceb37f4644dd3af7cb4fe4a0681d3 : SUCCESS in 21m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6f13ab3c97a34e79bb444a6f9d57308d : SUCCESS in 1h 10m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/abcc2c08e1e1458d9f4263ca09900882 : SUCCESS in 1h 27m 47s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fbc9afaf3d5e47bbaa9e703d28f5bd94 : TIMED_OUT in 2h 31m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/29d45d79ece148d68977f6dcbcdceb92 : FAILURE in 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8e12eff5cf2f0457b60e781c7863cc33f1839ad3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-10 11:51:57.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"cd3e605b4e10998b61b13f5f9b77408c83228104","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-10 12:11:10.000000000","message":"Patch Set 8: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6f786574fc7b849ca2f4883e5d4c51891a12ee69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 12:11:35.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f0b9478fc83ca283823eaa7d50e1f12301deb1c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:31:08.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/168d4f0cd5ef4ef58fd13635c3379a7c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/eb94a7fc7db24409bc07ecd5130b1bc7 : SUCCESS in 19m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2a43ad565e784bacbb8bbd1512daf53a : SUCCESS in 30m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/60c17e7f024945dd95e45f5f0718c395 : SUCCESS in 1h 04m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7213473c594cd38c1837df10e71f8a : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bdba20e1f254aaa9cc3cf2fd4e60b4c : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73c33aa3e5c448c68219a40b26627ea4 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a9016d9a4e49f590c888404d432a53 : SUCCESS in 6m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d17ea9ca56504d1dad15af0fca4b7e9f : SUCCESS in 3m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dce415b548514b76866a830b66236d72 : SUCCESS in 23m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aae0da8437f94593b283a622b68af3e7 : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/435aa1b1beee44029fd99d2098d96973 : SUCCESS in 1h 04m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f67b3262abba4260a54370cb21013705 : SUCCESS in 1h 39m 37s","accounts_in_message":[],"_revision_number":8},{"id":"46ac6c0fe84c6fcb167c19e615dba4cc2a95b1b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:31:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"711637ca0d986f28732c177bfd8284690f879291","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:32:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f684004ba1dd42099346335e97dc162f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a004f6ed7fe4ff4be87a72f6d051bdb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5121ff7679f416844ef74774c825b6be1382c74b","revisions":{"9ccec94a1d914d123f67e2fd26aa222ab2dac9cd":{"kind":"REWORK","_number":1,"created":"2025-06-13 10:12:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/1"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:12:06.000000000","tz":120},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ccec94a1d914d123f67e2fd26aa222ab2dac9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ccec94a1d914d123f67e2fd26aa222ab2dac9cd"}]},"branch":"refs/heads/master"},"9088297cb49398a12ec4e6e2772c906b708c0ede":{"kind":"REWORK","_number":2,"created":"2025-06-16 13:42:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/2"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-16 13:42:15.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9088297cb49398a12ec4e6e2772c906b708c0ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9088297cb49398a12ec4e6e2772c906b708c0ede"}]},"branch":"refs/heads/master"},"d912299fd1443796a9c429f1b984932e0159925d":{"kind":"REWORK","_number":3,"created":"2025-06-18 06:42:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/3"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-18 06:41:58.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d912299fd1443796a9c429f1b984932e0159925d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d912299fd1443796a9c429f1b984932e0159925d"}]},"branch":"refs/heads/master"},"43e66b65a6264247f78a2eac87ab11ebdeff7be7":{"kind":"REWORK","_number":4,"created":"2025-06-18 15:57:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/4"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-18 15:57:30.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43e66b65a6264247f78a2eac87ab11ebdeff7be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43e66b65a6264247f78a2eac87ab11ebdeff7be7"}]},"branch":"refs/heads/master"},"1e09213b547b4218daeca5316120206342f0f3e3":{"kind":"REWORK","_number":5,"created":"2025-06-18 15:58:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/5"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-18 15:58:23.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e09213b547b4218daeca5316120206342f0f3e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e09213b547b4218daeca5316120206342f0f3e3"}]},"branch":"refs/heads/master"},"939ace78317850fc8c98bcf5645b8b12ce1e7762":{"kind":"REWORK","_number":6,"created":"2025-06-18 20:04:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/6"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-18 20:04:59.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/939ace78317850fc8c98bcf5645b8b12ce1e7762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/939ace78317850fc8c98bcf5645b8b12ce1e7762"}]},"branch":"refs/heads/master"},"23712a82a194e9ff8506c7f0e458f44319158676":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-06-24 07:38:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/7"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-24 07:38:18.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\novn is running as root, and fluentd can\u0027t read it\u0027s log files,\nso we need to precreate these.\n\nCloses-Bug: #2115244\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23712a82a194e9ff8506c7f0e458f44319158676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23712a82a194e9ff8506c7f0e458f44319158676"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5121ff7679f416844ef74774c825b6be1382c74b":{"kind":"REWORK","_number":8,"created":"2025-06-27 05:48:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/952566/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/952566/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/952566/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/952566/8"}}},"commit":{"parents":[{"commit":"2319d121451167c2519938c234de2c65e7d3a797","subject":"Merge \"Remove dibbler from neutron-l3-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2319d121451167c2519938c234de2c65e7d3a797"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-27 05:48:23.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\novn is running as root, and fluentd can\u0027t read it\u0027s log files,\nso we need to precreate these.\n\nCloses-Bug: #2115244\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5121ff7679f416844ef74774c825b6be1382c74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5121ff7679f416844ef74774c825b6be1382c74b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
