)]}'
{"id":"openstack%2Fkolla~952694","triplet_id":"openstack%2Fkolla~stable%2F2024.2~Ie6719928e4ec2901d661f44c9f259add35afa269","project":"openstack/kolla","branch":"stable/2024.2","topic":"add-rsync-nova-ssh-stable/2024.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-06-23 06:06:21.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-22 10:48:29.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-22 10:48:29.000000000","reason":"Change was submitted"},"30523":{"account":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"last_update":"2025-07-22 10:48:29.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-07-22 09:11:35.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ie6719928e4ec2901d661f44c9f259add35afa269","subject":"Add rsync package to the nova-ssh container","status":"MERGED","created":"2025-06-16 16:21:13.000000000","updated":"2025-07-22 10:49:29.000000000","submitted":"2025-07-22 10:48:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952694-add-rsync-nova-ssh-stable/2024.2","cherry_pick_of_change":952463,"cherry_pick_of_patch_set":3,"meta_rev_id":"ab71862f2ffaaff961e163d992040fd1db49a0f4","_number":952694,"virtual_id_number":952694,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 10:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-21 09:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-22 09:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-22 09:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"CC":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-16 16:21:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-16 16:21:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-06-16 18:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 08:27:30.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"CC"},{"updated":"2025-07-16 21:15:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"23d41f7808198537430c7301d9087ec3bedf5be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-16 16:21:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1022721afe45328f3f7da9c76d36a59ac2504067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 18:09:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52668975f75c49078f0ca245cbd34f63\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e1d91cc80e24094b993369249e2ab97 : FAILURE in 20m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1c46625931ed488186045bf6519b95e6 : FAILURE in 25m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4d187b37c7b445e68daa238846da5412 : SUCCESS in 9m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/822b519bc93048a39ddf41b4f0e78197 : SUCCESS in 27m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47625f26fbc444a590fa9e40f18c6e63 : SUCCESS in 1h 12m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5e30b7235d914e06a0a5555684cc96e6 : SUCCESS in 1h 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e0c5b8fd97045f2a09e363a860be431 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d6d65821c0a4baca5642f8128bfb537 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b2c7240e44f4c709de46aedfdaa2b1b : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/377f134e70b949bfa82f5e89ffd9389e : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/286c05fc72734654a2c028c86a91ac04 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2d7617f9bb412189aeb4d6532638d1 : SUCCESS in 4m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c2f33fb291724949a74db3a8c0620a72 : SUCCESS in 2m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ed0870727b0b4a208fd17cf59cd9eb6e : FAILURE in 16m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a73485f25aa04c4ca6d4a16158cbeddc : FAILURE in 28m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d4afffa22da43ebba70184778da97af : FAILURE in 24m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e80f79ac690745539dacbe31f4e0131d : FAILURE in 22m 06s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/894a292104844d6c9100464a3417d983 : FAILURE in 20m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/98ddf9d9f893451bbb1f3d4a25e018bd : SUCCESS in 18m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ebb7073b71494fdf964ccc2bb305a871 : SUCCESS in 24m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/28b8c5cc15214a2c8f87e0d0d2fde6f0 : SUCCESS in 1h 00m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b477787578a547e9a8405c990b3933a8 : SUCCESS in 1h 22m 42s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/1df646f5a22343839fa11c2f78ebde96 : SUCCESS in 50m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"882eb07778363493aa0338ec8158582fad9e5e3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 18:20:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc1f9e7dd4f412d907935c366115c6d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/67dc024f31ec46008f1171583635c79d : FAILURE in 56m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2b946091e134029a2bd6b0114347049 : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2c35dc693da45e8a9e1e383c84d3103 : SUCCESS in 1h 55m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b69d9de88414430aac216ff48850be6 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b821d99dc5684b2b973546b8593b2278 : SUCCESS in 3m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/49fba3a515594f01b99de0519b677487 : FAILURE in 48m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2e5d9d10ce92436e8186d1346a481659 : SUCCESS in 56m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69cbf7588145da1d307ac97a23a70d277ab44fe6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-16 18:21:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c32236e9a999dea20994c6b971d26dc5a29b4a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 19:57:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17f37797f0e047f1bbb2bb810602afd5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/347d4bf576b34bf1abc8f7f73e594eca : FAILURE in 17m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c46c75efbf85479e9563d139f87a6260 : FAILURE in 21m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/880a0a1e63004df9baca4d80c5d19b21 : SUCCESS in 21m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3ae7b5cd948746bd9b40d3d962ebc7f5 : SUCCESS in 23m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3896edb2449a4696973ae8b650f65bed : SUCCESS in 56m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c060972c54d94e92b6edf3bbf1bed1da : SUCCESS in 49m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a20dbb487d049a8abdded208b93f2a6 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e96aa5b6ccd34fc295b09a9af28a961a : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12c2a474d1ae4c0dbd9b1295803eb75a : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9304329839634abfa3c6815f9f208e7e : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04b55ebcbe1d4008bc699d9a9b51d858 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bf919cd93d401b94b92a3f30b496d0 : SUCCESS in 4m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/76052b8d2a80476a917a5108203d4b1c : SUCCESS in 1m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8632b3b2dc5348389da68036e2ca3dd7 : FAILURE in 17m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/daef3ebef3214096988857b82fbff24a : FAILURE in 29m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e19dc390f8c54ba19f0162cd84ee54c7 : FAILURE in 20m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5b772ee9b16149fc8e7f7a4515ba4074 : FAILURE in 20m 34s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/b27a564d85794d3c8ba93b40f16396de : FAILURE in 19m 48s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ac47035d5cf41ec8cc7011c7bcf913d : SUCCESS in 21m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/20e5ead751594ee9820c34f1b5fd4806 : SUCCESS in 21m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/19d14685a4654873bebf89f78042dd20 : SUCCESS in 57m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00994bd8663b42fd9af13bb06bf65114 : SUCCESS in 1h 30m 40s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/8a7b0650e64744eabc7e66a69cec7978 : SUCCESS in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5bdcac57dd4f3697a8e6422a8d1959b58ad7450","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 20:30:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8445b4646884edc9eb2d28f4cc8f303\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/51f54cd4e0774c83a9b72fe0ee8e174e : FAILURE in 1h 09m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e177a4e3f5ce41478d2d26e3e3afc852 : SUCCESS in 1h 21m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d37c48f22b13498f9b3b0f358031a04e : SUCCESS in 2h 06m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6be0659e279b4dc78fae7253d0fa93bc : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c85a256a9de24b548f0bfb18c7b0066c : SUCCESS in 3m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7636395395b04b0f8272f16b51b9d574 : FAILURE in 1h 13m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/55f549d67cf848a38f4498b936473655 : SUCCESS in 1h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32d948205eac8f390ac6490a5ec876931e1c7c33","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-17 05:58:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecbcff3059401aad11331ffe4ed458d2aa47c14a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 07:41:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf735e1bd9ef4adf8e870f6dca0bf048\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e943c9081bf2493fb0900405bcd8b5a0 : FAILURE in 15m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/53395059b8374280b3917aac7efad0da : FAILURE in 23m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e8269205ec164adc88d201fc3e708081 : SUCCESS in 20m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3e6e7b0806fe47a2bd9d34a8bc5289c0 : SUCCESS in 26m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4e9d80b0a2ed43bb84bce8098ee1a7cc : SUCCESS in 1h 01m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/465cbcc4860c48c180be36a951d6e467 : SUCCESS in 1h 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46212bd5d6d4757983de62fd4702445 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aadd1b4766e4830b0c66d23a7e4cab3 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5616cdd7f5b41b4a492f7c54b73aaa5 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb40ff9dadfe4ac798e7b4e81802184f : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dce8b44fdf444c46a1b179f9fff1d0b6 : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0ddcb0ce104ca6a1266d115a8c3c30 : SUCCESS in 5m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf9b303e760149e18ede848e1918537a : SUCCESS in 2m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b5f525efeb4e4c298c617409c66513a9 : FAILURE in 20m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/be1a444ae45e4a94be22968f00e089af : FAILURE in 22m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1d252e6c36e544a19f0322bc7bfecbd6 : FAILURE in 16m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/24d87c1521c84e44918364fb8d280f00 : FAILURE in 8m 54s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/f1326903c38a4946bb22a9e4b64e69b4 : FAILURE in 16m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec3d7bf6850e413e93b483ea74d37155 : SUCCESS in 21m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5075875259f745b3bafacf2237437a3e : SUCCESS in 27m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1388ea7bb1fe472bbf7ee9b72456f8c3 : SUCCESS in 1h 05m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3318065f2f9c429eba2f4f449104a951 : SUCCESS in 1h 06m 38s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/6185ed0f2dd44db99e8b2926d3a5e501 : SUCCESS in 57m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51b939faffd06a4bf4bda0c3e421fe45cc183b22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 07:51:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b60811bbb87b4dba89f266ec61866bd5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a033df14365949d38b10c6350a92922e : FAILURE in 1h 00m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14d4f604c1eb42dd8ed0de5e47dd02eb : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d6e5944354634607b191ce134a90af2b : SUCCESS in 1h 50m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63732ea0dcef4c39805d3d933334daaa : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d87e91b710764e46bcb4b8bb97120354 : SUCCESS in 3m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/679e4540375f4c3fa92bc2bb6f8a1bb5 : FAILURE in 42m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/09618022f3674610abc4cbfc7a41619e : SUCCESS in 54m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7504afe4599642cf9d9c4013552532c29077f259","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-18 12:39:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"106ad8f41ec1ad74f32a1bd24beed81b279f6e46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 13:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08e4185cbc1f47709408de88cea79b30\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a17b766739b34102bf612963bb0581e8 : FAILURE in 1h 10m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25d44a311ae0403983fa0b484265ec95 : SUCCESS in 1h 13m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02ce4957f5374d47a7f5d710eb853d58 : SUCCESS in 1h 10m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca40662728bc4011bef724744d5b5c96 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/abee6fc6b69a49919f005973adccfc08 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1b156b9444c54dc4b7cd64cff7541734 : FAILURE in 1h 06m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aac045c4b5f945d984cb441b883e3af5 : SUCCESS in 1h 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebe3d346eaeca98b698e8f3915f2f985a5df7b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 14:59:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07113956672545e9b8586137f499b1b2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/76c5d55e0db04838bf8ad2fd97a9e2b4 : FAILURE in 18m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/823791a6848146ef856be9f5c52a5e34 : FAILURE in 19m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2afd803a025d4f1d8af0839eef78c556 : SUCCESS in 18m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/02eba9226c30488fb84f69792263d791 : SUCCESS in 27m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c2f0533dc0d244e3ac1f3f16ab106f20 : SUCCESS in 1h 01m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/127802079f6b40b1a420b0ea33855cec : SUCCESS in 1h 45m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e078e05595334f6b83acdfd910c1e6fd : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe8c3fa04ac24599a046a7398bb64a82 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0433b4a591c84c2ba5bf56ceb8f158bb : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4250ed9b80c44405ba40f9e0460472d3 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cffeabb2206c4d5f89dadce85b5dad51 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0d93f54a234460a3721499c720021c : SUCCESS in 4m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/477bd98893704a85acc18b6847b5c947 : SUCCESS in 2m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f6c02c85fdd84371b80e8a35c28f1529 : FAILURE in 20m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/de1915f736f443f89fb3119d444caf7e : FAILURE in 28m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/08849e7bd23f49b3b89abf6efb47f0d4 : FAILURE in 19m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20cdf3f1301b424fa32248899b2f4bb6 : FAILURE in 17m 08s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/ddaefc4690a64b4a8f18d04d1180ce90 : FAILURE in 21m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3f241fd237c344159856caf4a55d347d : SUCCESS in 23m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9625b7027d8b4ddea3417818f4fc919b : SUCCESS in 35m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/00784385cb1b4ba99f7f2aa81aa91919 : SUCCESS in 1h 08m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f23507508f5147d99c9e11fc27884cf6 : SUCCESS in 1h 15m 15s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/1fa833d21ce54e84b878cd43a53ee4aa : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95c9e153f32e025a5fac5838b8926a241392abda","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-18 18:19:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46133bc60f8e0fa5a8c8fb9f62e4223d4e5e89f6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-20 08:26:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6570694fffe36fb4e81a5bb4c764a7941d36994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 10:14:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa42c8a16edf4cd58b45b8d0adc7d632\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8a3daff669a847759c91dcf1dbb86a7f : FAILURE in 20m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ab0e02e382e74e96b75ab07bb2d8178e : FAILURE in 16m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/09a4ee2710264a018084ec1a97eb6362 : SUCCESS in 15m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bd4692796dbd488bb6ca91564da25679 : SUCCESS in 22m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8b0471b639f4a0684ee27726ba0227f : SUCCESS in 58m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/62aedc8c1cb44fcfbf742fa0fda7618f : SUCCESS in 1h 40m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cdae485ff5412093b5e5ec9e204bdd : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cc33d1b80eb46969733e0d15b068ca1 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2719ff4d2c848b7adeba495ae6eb03f : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd60b58c0f08417381459eeade4a57bf : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93031aaf5a6b4743902bf1a2cec9f579 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaaff86004a54eb1b71875b619f0dd52 : SUCCESS in 4m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c64f8b83d9c24963bd9ef74d0deded27 : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a912e9125e7416ba7e3db4d82da9d1b : FAILURE in 16m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9424e0f194634f458498f8532cd2d53f : FAILURE in 20m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4ce9e4eb89514a82bc0b9465d920b7db : FAILURE in 20m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c9e47c05110445768677671b4a4111c9 : FAILURE in 20m 49s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/6e22ad3a24c84488adf761bdbe7c7265 : FAILURE in 22m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/458e3472e65e4861a9390c1b8144cee6 : SUCCESS in 19m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a72037e3d51048f789540ff4faf3bb74 : SUCCESS in 26m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/90a2a22b7a374d6ba02cbc392b7396ad : SUCCESS in 1h 06m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/30eaa1ebf5154f229ceab8f35c3ec266 : SUCCESS in 1h 27m 49s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/661adb84766546709e5e85733195439f : SUCCESS in 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b4366cca66ae4ed2d43afda0e7f8489c4fdd44c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 10:15:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26a9c0ee84ca4e6d8b30f8d22684b972\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/65af1ba31dc14692b1a72214b0437e35 : FAILURE in 39m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aa5837fa45e74d8fb1e529f867cce1b3 : SUCCESS in 41m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0206a22636244ed69c97b1111fbd36cc : FAILURE in 1h 46m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb60740bcdec474c8ad6119912a61bf5 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ed6e84c80f6044a59deb9820b4e4ca08 : SUCCESS in 3m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8db4a70083e447f193177bec57d2e595 : FAILURE in 47m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b3be129357964e1baa62828bea3a2b51 : SUCCESS in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49d728c79af7f752148eee89239dceb7d968ebdd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-23 08:03:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab9c24cf3dadbbf902bbabdb031974aa9a45e7a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 09:45:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea8b3bade614973b814b2aa87654f80\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a3b7c4ef31dd4833af5843a36ac2111b : FAILURE in 39m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/18f8b30c115d42968320eac0d2de6abf : SUCCESS in 1h 07m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d2f1b729df14742ab481c8154349671 : SUCCESS in 1h 37m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/694331e30b934a849feefc42adfaab29 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22ad04897b824da29feb42326854d5fc : SUCCESS in 4m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/35a120ae99864d3f9f08ce6f058fd678 : FAILURE in 40m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/58f830574cb348f7a1a26bedc182c2e3 : SUCCESS in 1h 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74d71e3a26e1acefa56a162fcd8a3ac29183ce4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 09:45:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a285219420554d169b101e32b5e16c32\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c9c15692975b43b295df755b98f0279c : FAILURE in 17m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5df0240644dc42dd90c83146799dba8f : FAILURE in 21m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5de71c8693924490a5613f47cad29413 : SUCCESS in 18m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6f89e9a7efce488799bbd198c8c11ca5 : SUCCESS in 20m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/54b32b4e2c8944e7bb8ddec338d681de : SUCCESS in 58m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a8bc3af1ad2a40dba8eb6dfd1c3a2fc1 : SUCCESS in 1h 38m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758f845ee2db46bfb97e6b8488723dfa : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f315057a47084624bf465ae5bf2e9843 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/790597f9de134865b7291d670d670a45 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc69dc63fc174c80a50ef4a7352753c9 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f88eca985199434ba34c48d0c52ca9a9 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a7beb3a73f41cb826c4018ab66b391 : SUCCESS in 4m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c206a41c30f3430db0ccb1bf3ad33c19 : SUCCESS in 2m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2a607e99fb0947d6ab187f2544f17c73 : FAILURE in 17m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ced70c1238e4b549bdbe2394954b5c1 : FAILURE in 20m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/62e3802e289640a9acf16f7e0bb55b1b : FAILURE in 19m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4c9135ce0ba349559950c3278d446583 : FAILURE in 17m 32s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/a28002eeae864cd79721e1468ce62658 : FAILURE in 15m 54s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f4e86bdc639349ebb287b46d229a3902 : SUCCESS in 22m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3cb7039163f24edc9587f6d187fcae2c : SUCCESS in 21m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/099749fe411a48aeb2bb709f904d7b01 : SUCCESS in 56m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73abc6fd5fd743fabd3c79471f57a53b : SUCCESS in 1h 09m 05s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/aed1ad2e778b4eec8e772f80579b3efa : SUCCESS in 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acc9f2d76281970da71773adbb9eec443cd64eed","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-25 12:49:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e93100dede1d5779fc296ceb467fb9f00f89fabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:22:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7550729d175437e9d27e34becca2642\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b0716d460d354abf9b90a414cf2bbe24 : FAILURE in 23m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c517d0be068a41098c5c0c3265e0fa91 : FAILURE in 19m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9880a2095ac542188a2f8db1acddd7ae : SUCCESS in 19m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5c7b02f293124ec29f8b06551d8f45cb : SUCCESS in 27m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/317ece86db1143e5b4d78cff2b54dd6e : SUCCESS in 1h 04m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/29467b6425074a959b81f244692cb77c : SUCCESS in 1h 28m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd92821425b4e46a0adc801f9c3f437 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2338c380a09d452e9d32e779dabd8740 : SUCCESS in 2m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/482784ae76594145bf7092d0f3d091c8 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4538d2a40119472f82d771028fc47b5b : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0872eecaf3643048c1095fc64f7db1e : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28771d18ddb4fe1ac5d97e1f75b07e5 : SUCCESS in 5m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfcc3a4fbe384f7caea08d7e681ffedd : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3c2bcd0ad6624e56b803a05d2a1aa8ed : FAILURE in 20m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0bcde3650aed4d9abb13853701933dbb : FAILURE in 15m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fe57763abe914ff5bbd0d336da09eb34 : FAILURE in 19m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a34abab0f3c648eb9ecf46501d939c12 : FAILURE in 19m 27s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/b20fa43ad2b6402f9edfd2aa7ed1d300 : FAILURE in 20m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/34ebe16a79054d9da3a23fafdb950274 : SUCCESS in 21m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/58436ddb734a4b57a8bc537eb54e2cab : SUCCESS in 24m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1d4b67c97d874e5291093cf35ed783a5 : SUCCESS in 1h 09m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8e2f397310cf467199be56b21a6391b7 : SUCCESS in 1h 22m 17s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/9b1ef80e5b134aa98d033c8abff9ccbc : SUCCESS in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"836ead615a0b776b7da8a3d52c7f4524fcc41019","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 15:37:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d87f20d0dd4c5d8fee826455caf25b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f3956795110e43b7a48262e9bdc1af0c : FAILURE in 1h 35m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ad0b4915cac24ee2b13e33d4361eb19d : SUCCESS in 1h 33m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8a3e5cc1ed443a9b179f7555fba413d : SUCCESS in 2h 33m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d9d6ee195b440f2b6c478a14debea03 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c3a57bb7bfa47a9851236fafb95420b : SUCCESS in 4m 08s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bc08dba4e6ea474e80fd050d78463916 : FAILURE in 1h 17m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d97e4ef4443240509820e129835b8bd8 : SUCCESS in 1h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b61cd5125847d20c5651192c47e61939a4153164","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2025-07-09 08:27:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daa3067480f2c63ecc2a86ce25fa149b1fe37807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 10:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841245ac669d4e1ca8e184a19600331f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/84a5728768684b6fa1c170fc179e77e0 : SUCCESS in 22m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a0ae3101cadb49a7bc64ebd9ae6e13f8 : SUCCESS in 48m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8164d1151ea04432975db1c94ab0ca07 : SUCCESS in 18m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b3e75c9d16904b2984e28695542f2cb5 : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/81b4dcdf5cae4a1680dc5241c0d097fc : SUCCESS in 49m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/27012f90e4a84135a70f2e5646acd916 : SUCCESS in 1h 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9225283a25ff4fad92e99dc202deea08 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9de5c6cd0dc047d3813557104e3bcbda : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2d27e5bf5af4b81845b40f3c964a6bf : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4de11df778304f4fb4f44b9a90526d16 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0f6d265d674492c8ea05359d06e9dd2 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f31c0ae4034e7882daa2befcebd2f1 : SUCCESS in 4m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/245ba4e649e341668c0c764ca73aac88 : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2d4513e4c4524ecb916cfce6fcda71e6 : SUCCESS in 19m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/40a34a15d8274374b24624b868f87fbb : SUCCESS in 25m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fab94ff7576947adb542c6a701d76465 : SUCCESS in 1h 06m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b7de383ec28843a886431317f5e03294 : SUCCESS in 1h 19m 32s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/90c65989fc5346d88c157a8fea9f5843 : SUCCESS in 42m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6be9a28d7e084fcf87d68069d654d270 : SUCCESS in 16m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2c87109020e74252a0ba2fe2cbbd52d1 : SUCCESS in 21m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/91bdd2e5713746b7b7024497a922155c : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/63592dde923f402d921569ba9e1f273b : SUCCESS in 1h 31m 51s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/c42119919e41437c87304867259edb1f : SUCCESS in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"996ad41c23ef355942e787d1e03e11f35715c12a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 10:37:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4883ea262f8c41edbb09c0fd6c7b1e17\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f6d44d3049194074bdcf2425573cbf0c : FAILURE in 1h 20m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d00afcbe60724705a4c30adac9151399 : SUCCESS in 1h 27m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec9c3b7118474b4a9138697a2939bbf1 : SUCCESS in 2h 07m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9602ac05095a447ab91d75fbc014a140 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/82fa07cea63b48ac90d3e24edbcdcb4f : SUCCESS in 5m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c919d2da2eec45fda12c2caa18c78e16 : FAILURE in 1h 20m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d7ad96a284d2491992da6a42549215a7 : SUCCESS in 1h 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fddcf9b08c9ea8944148e75c23453a0ba473d18","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-16 21:15:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4053214f4ab1b3ecf2d34a8ea7872059a22068b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-18 08:50:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c6ee2298cb53c26dfac2d2183f32e6eae1dbc2c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-18 09:00:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"445f59b7159eb4bf5e42a5b745d071a39ec99ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 09:01:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8cf41eeb6238576943b933e5c2878c2191e0bb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 10:49:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41a59e89c12a44a78bc193a5b13ce1a0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/96196c7d8bd44294b72fe7e8671b53e8 : SUCCESS in 13m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d5297b3ec8104314af4a490f6e7bf9ff : SUCCESS in 29m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/565de0f1cc3e4dc4ba64858029d0021a : SUCCESS in 33m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7ccbbae8a7fd408893e3846885931b59 : FAILURE in 1h 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cac5d010e5e4744a90d437b3baa2362 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15190fb0a24b4bb7a33bb367c1f0fe0c : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2150c3badeca44eca1d147c116f436e7 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121ba235e5344c8d9aba7f5ab449dd18 : SUCCESS in 6m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/74ec19a0fe3f4c7ab4a39cc02d389000 : SUCCESS in 3m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7619993ccd1e4d63b1c6ef5340ca121b : SUCCESS in 18m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9f801ce7c44349c7a5da56a76a6510e9 : SUCCESS in 29m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e4dfb9d6b8054693b759b3e70fcbd1ac : SUCCESS in 35m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b52111980d2c4549b85838d86c3ea156 : SUCCESS in 48m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4ece9264a930446197e26ac10dad1dea : FAILURE in 15m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ceac69d430eb4798bd2dedb8aadf4757 : FAILURE in 17m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/367bf641a02b40cd998127c6aa0636b8 : FAILURE in 11m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5325fb5349c343249d719618437a6a4a : FAILURE in 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f0088cc71e525ffc1c70cc9febd4db965e65f45b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-18 10:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94c46d5a94708d4872cd8000c769ebf33905ec8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 12:30:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c525f97d3a4147c2aac6862b77720327\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5c0bd1c38d65496091dcd09023290e5b : SUCCESS in 22m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/301ee8b75c4644ba9f018f5a3185b07a : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7aa89644083a4401ad4d4626d1a96e54 : SUCCESS in 18m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8b860126a8c34b3b8aa02f6226d2605d : SUCCESS in 25m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e4e1d337ab994570811f65fe3b172db4 : SUCCESS in 1h 08m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bd42c9ffa127472ba3e75523694d43fb : SUCCESS in 1h 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a1553521a0455d974c7f16cfe9a116 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69b628364e3d41d1901710df0b7bd98a : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e7bb88789724f2ebd163dfe10909b30 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7acc4d386e9c4cdd9e98df83d8c8157b : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8edf67cfda540fcb3d6bccaddec77fc : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fe692239b348598eab13d1d2c46ad4 : SUCCESS in 4m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa7e775b556e4bd5aac809ae52f90de1 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ebd72c24a544276a2f24772e23a8428 : SUCCESS in 21m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ec3390dba9c64a3ca210d7a27501e5bd : SUCCESS in 18m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/16e4cf9559ed48abb8276220b7888022 : SUCCESS in 1h 10m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e27a0a6781f140fc9414c4d3630429f3 : SUCCESS in 1h 33m 47s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/3f6b810a117746ce809404aa3a40577e : SUCCESS in 47m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4ddc06b6f33243fab4391a7a88368935 : FAILURE in 22m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/590636d65b444e08910a9be92b4879af : FAILURE in 26m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae1dbf1488d146b6ae23424e4379562c : FAILURE in 18m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/050f5dde7b7f4d2b80953528282c933b : FAILURE in 17m 32s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/d3e191d74d934dcf99e452af6cf8e25a : FAILURE in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bd0e8cde9c589e5dd88e2f03ea5d868bece3223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 13:21:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f402782e4e445019dcf88e7ad7d0096\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f27f8b51f2e486bb2e60865fdff7bd3 : FAILURE in 49m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/95130ed23845481885354351e59f4c39 : SUCCESS in 48m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/291f218b531b47a896f49cde7943ad35 : SUCCESS in 1h 56m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0021e41d868a41e380a91d5dd58f9b6b : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0240fdc28d3548f389b05b1afac27fe0 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6d0ad504cf1548ed9dca63b558114edc : FAILURE in 51m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ddf0c072d8242efaf824c7c59f2f84b : FAILURE in 1h 06m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5d0d943232e76135bbfcd05820e7b482959945f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-18 15:39:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a095fe3c6526c7dc153b3ab29ac35827e001a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 17:00:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2effc0a670c0436bb791277a3ee6bbda\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/55c210b9f51b41a4987675c51d2e73ae : SUCCESS in 24m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fb62c29807be45f3ab6170ce47168f9d : SUCCESS in 44m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2efaa5c99cc249e08f492e717d02e7ba : SUCCESS in 24m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dabb827f52a94117b6d2802bc4b7f56f : SUCCESS in 27m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef1d36a7e09f4ea9a3d00339f3fd77fe : SUCCESS in 44m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/051cb592a3084cf6b8c751c90743665f : SUCCESS in 59m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8992ee25fa47d3b67b07254c79f4bb : SUCCESS in 1m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94befac9f26343e98588edcb5f59dbb8 : SUCCESS in 1m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ba5a3d0ed724390bd76a9a2ed94097a : SUCCESS in 1m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4727a0317a8421480774ba6a0fde693 : SUCCESS in 2m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e23ae11010e4d088b05ecac4b52313e : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7334c491a1f64627a462b262789810d1 : SUCCESS in 3m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a0294bf4d4b8438e918c04d9fb817e21 : SUCCESS in 1m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/449738aab36d42a08a04b6fc048762c6 : SUCCESS in 26m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/033b179609f943e49c05e2f7f1ab65b7 : SUCCESS in 24m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/deb38787ed004514bfe27fcebeb27252 : SUCCESS in 43m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/26a870cbc1d04f459b3314bc279a1c7d : SUCCESS in 56m 55s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/82a33e7ed913445fb3e799896e5ff808 : SUCCESS in 44m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b59e7c3ba8c54fbfbafa809c66b87e0e : SUCCESS in 25m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2b6aedbe171d4c748a315ad0cb7ed69d : SUCCESS in 26m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/73ddcb556b1f43f2972c050848be00f3 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6e26c2f33fdc4f46a52e0c7e7c7b43fc : SUCCESS in 1h 16m 48s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/5390d812e3a7421ea10a1710aa8e105f : SUCCESS in 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"406af1339052577c0da25419d9f7deb7d7e747ca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-18 17:02:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9a05f0c2c403b2b77f6f93d67a3a7bfc01f4b7ae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-18 17:14:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c10b223e3d37e195343656df723fab19e3a5c47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 19:04:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b965e641514ae0b81db7ed91aa8543\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a144d79a380947648576bcee2e14645a : FAILURE in 1h 03m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce6471430a804e67a5d29e8823839270 : SUCCESS in 1h 19m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02a35807639e4670b6367b4b89533585 : SUCCESS in 2h 08m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04f12ee8a37c4c1482959abdc6588cca : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b56b64914280442baafae71000bed955 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1c9cfaca40ec4818bc25477afbb61622 : FAILURE in 1h 12m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/198753e6c29e4c879f3ffc7550d65fd7 : SUCCESS in 1h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"234f88ac833558f99e7a8f7ceb8480091d3aa427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-21 09:22:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_14200\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"_revision_number":3},{"id":"6326a5b3345a7b82e065a8ffebaf318c076f36fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 10:51:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a56b636cdc804e15b3974a050f48e3e2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/06e5d42e24234afd8ade36164a527554 : SUCCESS in 24m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7f151548a2eb4420bc2822ffda50982b : SUCCESS in 31m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a07cb013e35445b693e4ab34f4b25ab6 : SUCCESS in 15m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8f909995e32e4af9bb639a42a79e1a76 : SUCCESS in 27m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8f28faec4b154298b8bda8543305b708 : SUCCESS in 1h 00m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/15a949dcd5d342b6bf7e361ad632ff59 : SUCCESS in 1h 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6687c3c2e4674830b58190095ecd73ba : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbc62a550fb241f79dfc41dc236bf434 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/149961d309344033a12292ed28596a4b : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adbe5ee677d34b29b991eebf6ac6306c : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f8757de27a94121841154e20accc028 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4c049e50ec4a41a62443313ff5de0d : SUCCESS in 5m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/251b641446bf498994b2a230ebe8f01e : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8cd7681472c04a9ebddfe101ae4883ce : SUCCESS in 19m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b3328b087fc4d1eb224064fb1cfee33 : SUCCESS in 29m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/efc313c5ec2940f2887a52556148de45 : SUCCESS in 1h 03m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2abb61d402ce4d61a5ba5e559bc9e3f8 : SUCCESS in 1h 23m 52s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/e56766cf01b7427eb804ea6b6651d586 : SUCCESS in 42m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5d4c83551a9740e0aa048b9107420f2a : SUCCESS in 17m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1363befcb8964d3892e049ead1eab549 : SUCCESS in 23m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d3b546efb3d41a69b19eb836edd6985 : SUCCESS in 1h 02m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/543f555be38d44ea983753f782252047 : SUCCESS in 1h 07m 58s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/5d72c807eac34e46b3cfc4097dde1419 : SUCCESS in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"956236c972e02b6c44351c1b2ade06bf9d7803e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 11:28:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/651ac91be25345b69ee9e166cc66bb4d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1e8b63e58dbf4ea089a863bb078e8c55 : FAILURE in 58m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d7851faaf1543528ee2f75945a4127f : SUCCESS in 1h 00m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02bb0a7dd4fa496c9188054138800342 : SUCCESS in 2h 01m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ed312c63ee8494bbbc7b21b4e33afd2 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/137e42ecd066479883a0992c67b69d39 : SUCCESS in 3m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8ef5fee0e54b40fdb5564334fef796df : FAILURE in 48m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a68c98e374bb4221b9ff567aa0c4f920 : SUCCESS in 1h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cf8c1f9ef5529dc83a06f271e6d1a1c738597e5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-22 09:11:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c8084945948b3ae7d5d0626448fda4c3735add8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 09:11:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"439dbf6be6b520db0d5e2f50ab0cb3e2ab22d572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 10:48:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451c892ae44545ff85a32a95f112519a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e779fe2b272e4d8e817d297831b2a717 : SUCCESS in 19m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/66191838d1f944389823bde6072af399 : SUCCESS in 27m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9a29d8fa7c8e4e93bfefc51226ce30f6 : SUCCESS in 56m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0ebf20bf0b0e4bdea42d16c6907d319b : SUCCESS in 1h 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7862fa1c4a98451e949195fb1257a47d : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5d3de00f613456fa36b1ea6fcee3032 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f260635ecc54f5ab5edfffcce7af607 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dc59371f0a648e4b8a091d963d0f9da : SUCCESS in 4m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0250f056b63d46ffa66453397a03513a : SUCCESS in 2m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c58170da7af24c8e98a8bb0da4198331 : SUCCESS in 22m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2cd30d5f23ec4e5686a0c32fb8f383b9 : SUCCESS in 29m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b4eeb52a6436480abcd5416cada2b971 : SUCCESS in 58m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/955fda43904945aba08967db27c5273c : SUCCESS in 1h 30m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/37db6374a9ef4507a1e42224ad55ffa4 : SUCCESS in 24m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b959c949cf7e4954a3a4b29ce41bb794 : SUCCESS in 21m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5a882053eebb4f1b8499387b5b06a2aa : SUCCESS in 58m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3a63adb485a549978b6becf85af7a508 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ba801f9821a561f4dd5597addd38a8f28900f586","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 10:48:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ab71862f2ffaaff961e163d992040fd1db49a0f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 10:49:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2f8ff7eacc4599a1767a1cc247432b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a365d4ea74461f89f40615abd97274 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"371aed3dfc2600a9afd0ce9b2ef45705cfdb254a","revisions":{"7f4c49f3224cf856302568b42081160f69b7a8ed":{"kind":"REWORK","_number":1,"created":"2025-06-16 16:21:13.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/94/952694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/952694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/952694/1"}}},"commit":{"parents":[{"commit":"cb204aa173b364d3d5021afaa7995a215645791b","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb204aa173b364d3d5021afaa7995a215645791b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-12 14:05:37.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-16 16:21:13.000000000","tz":0},"subject":"Add rsync package to the nova-ssh container","message":"Add rsync package to the nova-ssh container\n\nThis change fixes the issue with absent rsync package when\n\u0027remote_filesystem_transport \u003d rsync\u0027 is enabled in nova_compute.\n\nCloses-Bug: #2113926\nChange-Id: Ie6719928e4ec2901d661f44c9f259add35afa269\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit bc018d398b5eb1fffe1f4e14baa508703df3a1a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f4c49f3224cf856302568b42081160f69b7a8ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f4c49f3224cf856302568b42081160f69b7a8ed"}]},"branch":"refs/heads/stable/2024.2"},"4d5b8d5eae87b7b7085204a006b6c89f61328bae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-18 15:39:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/94/952694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/952694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/952694/2"}}},"commit":{"parents":[{"commit":"ce411f7e09a34664cf8c9c90db7bd5225da4377e","subject":"Build RabbitMQ 4.1 in stable releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce411f7e09a34664cf8c9c90db7bd5225da4377e"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-12 14:05:37.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-07-18 15:39:08.000000000","tz":0},"subject":"Add rsync package to the nova-ssh container","message":"Add rsync package to the nova-ssh container\n\nThis change fixes the issue with absent rsync package when\n\u0027remote_filesystem_transport \u003d rsync\u0027 is enabled in nova_compute.\n\nCloses-Bug: #2113926\nChange-Id: Ie6719928e4ec2901d661f44c9f259add35afa269\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit bc018d398b5eb1fffe1f4e14baa508703df3a1a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d5b8d5eae87b7b7085204a006b6c89f61328bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d5b8d5eae87b7b7085204a006b6c89f61328bae"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"},"371aed3dfc2600a9afd0ce9b2ef45705cfdb254a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-07-21 09:22:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/94/952694/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/952694/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/952694/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/952694/3"}}},"commit":{"parents":[{"commit":"68214645d02bb4401e59c7d776070eb441372bb6","subject":"Build RabbitMQ 4.1 in stable releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68214645d02bb4401e59c7d776070eb441372bb6"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-12 14:05:37.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-07-21 09:22:25.000000000","tz":0},"subject":"Add rsync package to the nova-ssh container","message":"Add rsync package to the nova-ssh container\n\nThis change fixes the issue with absent rsync package when\n\u0027remote_filesystem_transport \u003d rsync\u0027 is enabled in nova_compute.\n\nCloses-Bug: #2113926\nChange-Id: Ie6719928e4ec2901d661f44c9f259add35afa269\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit bc018d398b5eb1fffe1f4e14baa508703df3a1a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/371aed3dfc2600a9afd0ce9b2ef45705cfdb254a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/371aed3dfc2600a9afd0ce9b2ef45705cfdb254a"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
