)]}'
{"id":"openstack%2Fkolla~953358","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ifd664de82a81830eeda8ccc019b22ee93768635e","project":"openstack/kolla","branch":"stable/2024.1","topic":"mariadb-fix-stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-09 19:55:00.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-26 07:40:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-09-09 19:55:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd664de82a81830eeda8ccc019b22ee93768635e","subject":"debian: Use bookworkm suite for rabbitmq APT repo","status":"MERGED","created":"2025-06-25 23:14:18.000000000","updated":"2025-09-09 20:25:37.000000000","submitted":"2025-09-09 19:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953358-mariadb-fix-stable/2024.1","cherry_pick_of_change":950531,"cherry_pick_of_patch_set":10,"meta_rev_id":"42c3159efb8bfb496d3d70ffd1229887b3c62e86","_number":953358,"virtual_id_number":953358,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-09 19:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-09-09 18:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-09 19:11:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-09-09 18:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 00:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 07:39:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-18 08:52:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-09-09 18:46:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"8c4a960388fa3bcb85272bc243c0343d5bd51687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-25 23:14:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c20d305aa119c60a014a5718d341c65f12608874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 00:26:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db4cca175fe248bfa6238117e811e545\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b423a8fce5ec4a608baa3ec83ced586a : FAILURE in 22m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/15571ad5d5644a0e8735ce82258dc271 : FAILURE in 18m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c8bf0556765949e68c95386d34a23b29 : SUCCESS in 18m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/180ac4d5cd884733a0e1ec2ac1a8f82f : SUCCESS in 30m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e9b279f9c24b4cfba16a2324a2895e0a : SUCCESS in 1h 06m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4b2eacc6824e95aebc1db20dade3cd : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21a9630d9dd47b997f7c7164306f3c3 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a72c59e13fa48b48beb0376a8718232 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c7586992f6a464e9adadb21c157064d : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adb7ce67248b402dac51034a1867fcec : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80ed03a791d46b0a1371d41eab470b5 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea60e54b70e4474c9d84633b13f22b02 : SUCCESS in 2m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/30e551302f5c44b99e35319641b252d6 : SUCCESS in 2m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c0ea0a8589d34bb2b5d05d03de930ed7 : FAILURE in 16m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6c03d656fdb8403abfed9396a17d6625 : FAILURE in 14m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c7a1e503156044729696a0ce4973022f : FAILURE in 18m 08s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cb6f924f37845009cb5f41416ee0626 : SUCCESS in 8m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5762216b9dd546cb84692709e34692b8 : SUCCESS in 10m 37s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/cb25d7d692364561bf920b755e5d2786 : SUCCESS in 34m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e7eca7d148484a2f95d86faf3463b55e : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a001f5515f7a62f8128687a394b0c0dfc6a4942f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 01:23:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1deaaf971d2e4574b4f9820990d42ca4\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bacb0a014d9b4dc68051eb9307d89bf3 : SUCCESS in 44m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/656ad5701fec4298a4c5fa976c1db89a : SUCCESS in 1h 50m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe5e4630ab2b428494b84c3640e3b406 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fbefd5e4292642e0abf06b23bea01de4 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/879bf534256842dab56d3381c0128d3f : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"538c24d7aacc3dd8a415aa77b3130879764d4ebe","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-26 07:40:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41e4107346bb972022dd70b124a7ceaad7caddd4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-18 08:52:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"924c1850d5dd9768d07fdf7960eba498d6e9e76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 10:07:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9cbce064b5f4ab3903b088e9f882802\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1529cf8191fc40b7a128067d2d477b50 : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c62d29dd261840abbce9dc146bba8466 : SUCCESS in 1h 09m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2f5784c5409947d89013614281ec3ced : SUCCESS in 15m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f37bd5f57bfb42959b0afd5433329d2e : SUCCESS in 27m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/726aceb83b294c82ba3005e8534edfdd : SUCCESS in 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a780c6a80cb34fd0ace93ccb7c93389f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96a5d4dd18f74b8f837387159c761f39 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec703d9305d742e586f8a9b0c28d6b6a : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd21b7c2d87745a4a2e0487da26ad5c1 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe027f0bdd349e3872b1a5ccf7a4589 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7063b6870244e6a410a85b47d04eaf : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/384e5fc341db43849e149b98a64dfbdc : SUCCESS in 2m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ba022ff2e5e0449da35b1a261196d8db : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/685b6801069a44728b76ecb01bab9434 : SUCCESS in 13m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5a5fccacc0894221be0161580cd16e4b : SUCCESS in 29m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4b611327bb014a199ebbcd840c1feb2b : SUCCESS in 37m 33s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a0d56424ce4777abb40e069d5d104d : SUCCESS in 16m 11s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09e3fbf5c0e248c2b759e0b281afc66f : FAILURE in 24m 28s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/fa957d79f9e74da09ab375c744008ec8 : SUCCESS in 38m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c4d937748154948bad2493414b52e94 : FAILURE in 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5f8a41ff1a92b423f62378fd97ca7c2830d5e1da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 11:31:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ede998efe6d49d6a96549417d50dd75\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69862518f0ed40a1b4ebe7c9d54dd4fb : SUCCESS in 44m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5764aa19c9194a8daedf4afd2b1ab4f9 : SUCCESS in 1h 47m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cd13f15964749308d4f569170c3f46f : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6184a6b14c1e48d7bf6bb9ba4386b36f : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fb969a76d054ef3a788b95dc68f4323 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3f98ea4f379cb05c4aa230fe44998a966a938fb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 12:59:00.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c42903654587c9939a55b3c325047477aa6b3a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 12:59:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7c9813765636bc534c417be846f2f2af8782219e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 17:00:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75f949c82ba043ccb522cb88fa83eccf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8ac6f14cee554c8087bd0f491a891898 : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b305485cfe0b47d2867c319f37abc95e : SUCCESS in 1h 08m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/560c09408f8e4a93bdc6eb079bb95b6b : SUCCESS in 19m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d1acdb96ad704bb8b2f20bbbd3d8587f : SUCCESS in 29m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3da646a592d44a1a90403150b2001cd2 : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79166a2cc36b4139826df5f899c75e71 : SUCCESS in 1m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/855c3f3112c84df3815b8b731fcc832d : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9dab411314d8459d8b375e8d71cd6f0f : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ae225042a354e61a1675ff94d967eb7 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/602149d0394d4eba98b265a702ce7165 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937e9e166134468ab1903f2a6c6a0633 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7e501b32a9a44f7bfbbcbd1ddb49a85 : SUCCESS in 3m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/40d3de305cd3490881ba41b325e39601 : SUCCESS in 2m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e59e17e80fcb4987a7772fd39da3686e : SUCCESS in 19m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a7c6beee25641a4acc0a138beca1752 : SUCCESS in 25m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5b411d40b614f8886ab52cc0724588a : FAILURE in 5m 39s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebe6b69f9d1f421faa673c9623261721 : SUCCESS in 15m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4f1d2ba58b046c893b66413a23e4a49 : SUCCESS in 10m 58s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/4b8870f72a6147adb01570476ba1e4c4 : SUCCESS in 56m 30s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb445fb04edc4bc0820bf5a423e66f17 : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":3},{"id":"966cbb5a51727f6d56d3e941fbc36bc8a78b2939","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:09:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451dc6b1f8a54e0a8a46de6eb7d22549\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/63f2674007164f7896ba81d798b9ae90 : SUCCESS in 51m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4b3b165e5bc7499489c5e88226433386 : SUCCESS in 1h 56m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/682914e9c6cf4f1fba59e3a2759bbbb8 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d982bb122199402a9cc2d8b79d44311f : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c756ec172c34260bb87b30156d049c9 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":3},{"id":"dbcab0dd8d03b93aaa33b6066d1b7974f247ffb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:56:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ecd3bd6a1434007052adf0b31edcf334fc2e0868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 18:45:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89d9ec3f03b349e98ceb78bf9d5b534c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5e699f9f416541d18811c2baa44bc532 : SUCCESS in 36m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2bf2595a70bd4dea805a4e7d2bdfb443 : SUCCESS in 1h 42m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1fa88dea2cd4dec81d1fb9f4aed7d10 : SUCCESS in 17m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/beef972911c6471fb56edf91304ec9a4 : SUCCESS in 24m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/39540c3b5f754f2b97e37648e44439fd : SUCCESS in 1h 06m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de461d045bf14582a4916411093e3929 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfc045894ec44f51825c52df6b539234 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77dd2b5d4cc24197901d315c86e83891 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7d6c556084b46e5a0901e3fc58cc1d9 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe6885cc2b9c49ea81d715ce063d9245 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac3209006dfa456e9674f1ac78379958 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c9465a9d2184898a293f18f0d6eb204 : SUCCESS in 4m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/017709a9867a4a50ba3f1a08031f0885 : SUCCESS in 2m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1cd823cddfcc467a9814d27b5b90225d : SUCCESS in 25m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7475d4e962894a149cb4844f6d773b4f : SUCCESS in 35m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f37129459a784aa2b02e603d9386b1d4 : SUCCESS in 59m 00s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d822419aa664415a24d66993d835e0d : SUCCESS in 9m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffc2c4e10caf47c39879b494dae7b4d8 : SUCCESS in 18m 45s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/a004756424b847db8f89093fa67d2b49 : SUCCESS in 56m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e14482686e3249a5b8886e10cc8f0195 : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":4},{"id":"413046a3bbaf0e6b697d24c8907f24740a1357f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 18:46:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dfeed5020d612eaac09fa7e20570c565ea1d650d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 18:46:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e2f0a030de5096dd2fba2fdd45e34984c8e60a25","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-09 19:11:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"917d7b71fab9fb27445af16ec168d074bdd4c220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 19:55:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2b71b5d5698441b9c04ef9f1e4f5d27\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dfae3121aa0a40da9e5eeca466f22f25 : SUCCESS in 9m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c9685c5bd93b488cb6df5cb2aca1482f : SUCCESS in 14m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a018f8e20264ebf940891246e065671 : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7dbfe7744324c6f8961d89c0f76cd2f : SUCCESS in 1m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fd0d22859dd41eeacd80a1062c1ffc4 : SUCCESS in 1m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb43f2094e5843ffaaff8aa184a589ba : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dc9f5aa65643c2a225d94cb8587dc9 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/675433e9e3714418a66aa36d4148994d : SUCCESS in 3m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c183cf7d192c43c8a3de2767cda5bea7 : SUCCESS in 1m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/65bc14b7f19c4c2fb5e32a2caba8d0c8 : SUCCESS in 25m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/69ce6d39d95b49ff80d15f78d1913cd3 : SUCCESS in 33m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/07c9701b807e4a6d94c6cada61d0d184 : SUCCESS in 1h 06m 01s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d295b3172b6b4cc087f2b31286919257 : SUCCESS in 21m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b048645255af49d099725a9d56879291 : SUCCESS in 10m 13s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/5329ca50c3224846ade2f9261dfe9f33 : SUCCESS in 1h 03m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43eb77da36724c1ba0f395c2cf608e62 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":4},{"id":"76c962fda38bda84c68461f012de0f34cfa17153","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 19:55:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bbef2fc99aebd327b28d7ecf880422737e8cd7e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 19:56:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e13e0adba1649d2ac37cb1b89240bee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6defc9af4fce4c2493929ea9d38886f5 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa695d6b99748f38f17ea3a437a77de : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4},{"id":"42c3159efb8bfb496d3d70ffd1229887b3c62e86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 20:25:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af524c3fbf2e48dd9865da266a64c6dc\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1e498e1929e14aa6bd8dc947dacc9b33 : SUCCESS in 50m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3d99474b63034c18a699e892f6226927 : FAILURE in 1h 56m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/480533ee2c3840c28d9a1935b9345176 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/edfeb0504ff146918d2cee48c16490a7 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7bb910daa45d40c1b229c621233c2e95 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"07103ab0a0960e2a509fa349a25ad014ef675e38","revisions":{"e1fd7b3d6d09d241b7e4cc172341de36b8662486":{"kind":"REWORK","_number":1,"created":"2025-06-25 23:14:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/58/953358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/953358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/953358/1"}}},"commit":{"parents":[{"commit":"9f557299c64f4143f17b59d0dd4e443c052506d5","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f557299c64f4143f17b59d0dd4e443c052506d5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-06-25 23:14:18.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1fd7b3d6d09d241b7e4cc172341de36b8662486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1fd7b3d6d09d241b7e4cc172341de36b8662486"}]},"branch":"refs/heads/stable/2024.1"},"be6001ae158fce6e79a9cf4343513558082b86e9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-18 12:59:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/953358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/953358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/953358/2"}}},"commit":{"parents":[{"commit":"9f557299c64f4143f17b59d0dd4e443c052506d5","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f557299c64f4143f17b59d0dd4e443c052506d5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 12:59:00.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be6001ae158fce6e79a9cf4343513558082b86e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be6001ae158fce6e79a9cf4343513558082b86e9"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"},"6a392eebd552bc2440d91bd6021f0bd3d2c528e0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-18 12:59:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/953358/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/953358/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/953358/3"}}},"commit":{"parents":[{"commit":"1fe1364310becf189e3020cebe23342b7a3e2420","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fe1364310becf189e3020cebe23342b7a3e2420"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 12:59:05.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a392eebd552bc2440d91bd6021f0bd3d2c528e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a392eebd552bc2440d91bd6021f0bd3d2c528e0"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"},"07103ab0a0960e2a509fa349a25ad014ef675e38":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-09 16:56:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/953358/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/953358/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/953358/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/953358/4"}}},"commit":{"parents":[{"commit":"37d68ab91b3184919511fb8941373c14f0fe17cc","subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37d68ab91b3184919511fb8941373c14f0fe17cc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-09 16:56:14.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07103ab0a0960e2a509fa349a25ad014ef675e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07103ab0a0960e2a509fa349a25ad014ef675e38"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
