)]}'
{"id":"openstack%2Fkolla~953667","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I8e20c77a8f1538162fc1dac9dc92c837348a09f4","project":"openstack/kolla","branch":"stable/2025.1","topic":"2111828-lsscsi_in_glance_container-stable/2025.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-03 17:39:27.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-15 14:44:29.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-15 16:43:17.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-07-15 16:43:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e20c77a8f1538162fc1dac9dc92c837348a09f4","subject":"Fix glance-api: add missing lsscsi package","status":"MERGED","created":"2025-06-27 18:35:48.000000000","updated":"2025-07-15 16:44:37.000000000","submitted":"2025-07-15 16:43:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953667-2111828-lsscsi_in_glance_container-stable/2025.1","cherry_pick_of_change":951090,"cherry_pick_of_patch_set":5,"meta_rev_id":"c5e11e674789cd4d4476259552a58f8b17ba17ad","_number":953667,"virtual_id_number":953667,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-15 16:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"all":[{"value":1,"date":"2025-07-02 21:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2025-07-03 17:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-07-15 14:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-27 18:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-07-15 14:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"CC":[{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"}],"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 18:35:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"state":"CC"},{"updated":"2025-06-27 18:36:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-06-27 18:36:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-27 18:36:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-27 20:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 21:42:38.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"}],"messages":[{"id":"a405e33c5da4bf2fb3ce1e9add8ac1327375a036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-27 18:35:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ec6231ce9a7843fb06b28daa48452b8274861731","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-27 18:37:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2a88427a11fe956559852d5813ca270afe001064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:43:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a0d696f4e54770bb7d32f5be5622f4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/abbcd6a206ba4baf8a13126fb105dfc9 : SUCCESS in 27m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/82b1f9b8c5dd433587b35497cfb66974 : SUCCESS in 1h 14m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/994c4845e94f4c55b131cee7f46c257f : SUCCESS in 22m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/090941ebdc074da7b18df9739bea3e7a : SUCCESS in 31m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a729d98f7a80467c97cab2abe3cd0914 : SUCCESS in 59m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6534b980fe4443c08514cef15f6a9515 : SUCCESS in 1h 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655ea49171c24367b6265e753839aefd : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0125eee970cb45bf91d47be80b4609b3 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3123cf8bc518495a9080805e6d27f958 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45ed68a8eaa4444d95ebb17183486330 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8238e08e1e4a1aa9b45f7d7c289b1a : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13dd70a375f44725898910c739503f2b : SUCCESS in 1m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/31d997db58a74570b72adf1898ce3759 : SUCCESS in 2m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cb26decfe51a4ef5b3c45026ecc70e46 : SUCCESS in 15m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9c0e103a409b4ea1b1706951e8eeb0a1 : SUCCESS in 1h 13m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0669b89bfd584732b5d7993d57a3c497 : SUCCESS in 1h 02m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/12e900c9c2824e3e877a810840c47452 : SUCCESS in 1h 40m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/91b3f2af177a41648fd5c532a534b093 : SUCCESS in 1h 51m 27s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a6c3a8d902cb4cd2aa6ab33324b4424b : FAILURE in 58m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4199191ebe84163ae63700bf8b49d1c : SUCCESS in 18m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0931e0fd51124d94ad0338b759b68d62 : SUCCESS in 24m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/64c9bdc2ae614fd9867705c7eef2cf3f : SUCCESS in 35m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cf75287e3b3342278f846a5e60f3d6ac : SUCCESS in 1h 34m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/49cfcc98005f46be862538610b1965f1 : SUCCESS in 2h 04m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9943c53c382a4e49b473c595a5953914 : FAILURE in 1h 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4fed8c777ffaf552e5ffedce48a9f6991c54ddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:52:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4c381c67e244dbae49eab263b55b48\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9e7b95a2d792428b8605cfbf871c7f26 : FAILURE in 1h 08m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e68bc7063de64b21acbccd8da3087034 : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/082bcd537ac14a0c862043f3cefd5ec4 : SUCCESS in 2h 14m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf853184a77142fa9e037859ecdc4929 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/821f9b4923e9467397499e40b1392655 : SUCCESS in 5m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a39507176f7649d5a99fa72cd22d73a4 : FAILURE in 59m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/77d1301677df46b68dfc042266fa7b9d : SUCCESS in 1h 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fc23987f2ba2ce55d7145225f83c965b6316714","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2025-07-02 21:42:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b6ed9582e853d21ef3b32d231112dc92d330c7a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-03 17:39:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09e27b9eacbf70bd7c87c7ad971a676d80c09c38","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-15 14:44:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc779710c937faa8d0afe5f77b41983cd8efbb2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 14:44:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ac12a5809de9816c892624976c09861177bddf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:43:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18abcf2535ba487981982beffe4fb349\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/52d737e4b4c9490baeb67489524b239a : SUCCESS in 18m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/463abd5edc944b888ad2981b63443c0c : SUCCESS in 24m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6eb604ca5ecb41f9959c9aece52f823b : SUCCESS in 1h 02m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7c7eae50f3da457aa6a158a803f8faf0 : SUCCESS in 1h 28m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964748c9e5f149c5bc5bad829bdb2b7b : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c00f45e4404b45bb53569af430ad39 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2dab55fbd2b4778bd73a90a0696f1bd : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de42e9c046084407901bafc420bc2c11 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee95380203c545b2a6bf1665c10ba17c : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1ebc45809bf04c5bb8af1b84e0745075 : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/39510f70261c4d1abea9c1ffb73944fb : SUCCESS in 28m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9a3936414ebb46889aace2c10e9169a6 : SUCCESS in 1h 20m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/19d1ed8a5d66443f9783f845ebb2f64c : SUCCESS in 1h 07m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9b15eb096472430ab08ad7a4904d2ccf : SUCCESS in 1h 53m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/58b1d668df2147878e720e0e9317d664 : SUCCESS in 20m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/60c0280f8300498ca366e6c4e8a26992 : SUCCESS in 29m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/80f0938a46ec421ab50dd6add99d55eb : SUCCESS in 1h 06m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/374b187813774dd39666881545c784c8 : SUCCESS in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d4a0422bfd2b88c68378dba7127aecc787b376b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:43:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c5e11e674789cd4d4476259552a58f8b17ba17ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:44:37.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39e30dd85d7e4a89a07fb005423edef1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1005b5abeca421c86450d32f1afc3ec : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d78d34b6ccf475d9888cccbd51b3165 : FAILURE in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f12fd3438d827dd79254bbdcb88a29011d639d59","revisions":{"f12fd3438d827dd79254bbdcb88a29011d639d59":{"kind":"REWORK","_number":1,"created":"2025-06-27 18:35:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/953667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/953667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/953667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/953667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/953667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/953667/1"}}},"commit":{"parents":[{"commit":"c445597590bea2c5b36515ff266c076f4ab26170","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c445597590bea2c5b36515ff266c076f4ab26170"}]}],"author":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-05-28 07:42:21.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-27 18:35:48.000000000","tz":0},"subject":"Fix glance-api: add missing lsscsi package","message":"Fix glance-api: add missing lsscsi package\n\nCloses-Bug: #2111828\nChange-Id: I8e20c77a8f1538162fc1dac9dc92c837348a09f4\n(cherry picked from commit aa2e504a709bb186d271736b6042b68c42ad440d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f12fd3438d827dd79254bbdcb88a29011d639d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f12fd3438d827dd79254bbdcb88a29011d639d59"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
