)]}'
{"id":"openstack%2Fkolla~953859","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-18 15:32:36.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-07-21 09:59:30.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-07-21 11:31:30.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-07-21 11:31:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9","subject":"Build RabbitMQ 4.1 in stable releases","status":"MERGED","created":"2025-07-01 08:59:09.000000000","updated":"2025-07-21 11:32:23.000000000","submitted":"2025-07-21 11:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953859","cherry_pick_of_change":951938,"cherry_pick_of_patch_set":7,"meta_rev_id":"1dc49e5df53f86e67b1a53e6b878b2b934535db1","_number":953859,"virtual_id_number":953859,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 11:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-18 14:35:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-07-18 15:32:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-21 09:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 09:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-01 09:16:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-01 09:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 15:09:24.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-07-18 11:11:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-07-18 14:35:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ac0c74e2bb1d4145701c9753fb07f6cb0121bffe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-01 08:59:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.\n\nThe following files contain Git conflicts:\n* docker/rabbitmq/Dockerfile.j2\n* docker/rabbitmq/apt_preferences_rabbitmq.debian\n* docker/rabbitmq/apt_preferences_rabbitmq.ubuntu\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"460d99bdab2e0719be07a0aec56cd945a80526a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 09:27:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a123c5a5bf1d4bcdac1c16c0f25116fe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9d52a9117cd34d63983418bb24248e52 : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/351deec233b84e32be7dcd2f7757bc5d : FAILURE in 15m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b50fe4e6825147cf86277ceb1a3a48b7 : FAILURE in 3m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fa32673916a04b809bdb6a5c6db2a7b1 : FAILURE in 4m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4e96048bb4c1477e81e57a8f2ce57fae : FAILURE in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/213c8d718b3142c6a44e176f1ce8eaee : FAILURE in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17d09cd240fd4b3a948a284e9f1b6b86 : FAILURE in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b38fda9a967f468588744920c5f74a8a : FAILURE in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e3bb5a9f20f4bcd8f36556aa38712ff : FAILURE in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b571c1237be942cf966649e7621bc84c : FAILURE in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2064db2c994dde98fc1d689ae3c9e4 : SUCCESS in 4m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf6cf106b67d4c249c1863597f125aaf : SUCCESS in 2m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7fe5d28a1ea842bf8a6fe9bb44e5922b : FAILURE in 4m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/60f4c0d39d0c4a37be0052272dde7d34 : FAILURE in 5m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a74a76ec53a4fef9a7bf5094831b122 : FAILURE in 15m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a2121cbd516a48469933d7049d8f4647 : FAILURE in 16m 19s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bcd6dbe86bc426094f6a8c6c174da16 : FAILURE in 4m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e53973c524774c139855300dcec15300 : FAILURE in 3m 59s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/2c942e170f59489e973c9b7c4859efb9 : FAILURE in 2m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fa664cbfb5341d8a53bc59c630f1d6a : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b09010637c9706dbc92be546934ac36ccb8354f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 11:24:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee4b714c56c4cc2b62ba8001ff4fdee\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eabcc544491945f9ac13eb00d55b1e94 : FAILURE in 3m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06a2b043c0a0478fb953888b2e757596 : FAILURE in 18m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec2cfc3445624e069068e81d6a6e5844 : FAILURE in 3m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a6355ac34a774bafaf6df10f04e8110a : FAILURE in 4m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbcc5b641646406691359fd420f8f59c : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"25b658b72b974136d1441f8ee14e5f66bbd81aac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-03 12:44:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"48456009637a55ebfad20e9d3cdec6b9b791e91d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-03 12:46:07.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"7b0e4332fe87c755a7246d4b7d29cd8ad6b4ba75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:58:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a687ecc41e754982bc6c66496c0a572d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3da5d09577714bcab0a296583fa645d8 : FAILURE in 24m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e437b92cebfa40a493e05c453b08c67d : FAILURE in 20m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41e9117ed6ce4331b7bcdbf4857d84e7 : SUCCESS in 21m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/409722e5143c4cd09ffe85a12239fee3 : SUCCESS in 31m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dcb7391eea874d0dbdbe6255b491edf3 : SUCCESS in 1h 08m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477d6e287c794da181a11acb1c8d4f20 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddac85bdea514b6ca7c612d0a85d5e02 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87048998baa249b48d5b5c6ca0ac1ea0 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d540f1bb57b4bd3b3c95fb74aa0400f : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e11e7f0392454ebbc30a745575d612 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2e17a768c347678328589891a5bcac : SUCCESS in 4m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e1a3e634a074454180ed7e73a16e58df : SUCCESS in 2m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c5940b9e6c144dea091f012c76a5d1a : FAILURE in 19m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd2e79f1aebb412e9ccdaf678891ce28 : FAILURE in 23m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0810b26526c14474a2e005b23c4cd52c : FAILURE in 23m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c638917c6afe45bc9ffab298b59a2ee7 : FAILURE in 23m 41s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e2ab16b7d0543c3b3bc26420005e151 : SUCCESS in 18m 11s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51a763c029f6435e85f8d30acf84bd2b : SUCCESS in 20m 34s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/ef66642105524fe8a6e25572e43cf174 : SUCCESS in 52m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eccf7543ca80448a8f1a088d01e295c4 : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4011e01ba7bae26540434d4f6efdc56cfa429b99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:56:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3aef883c50c48868d4add5885440a52\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2fdf4341e7b34ee4bef8224c6d7a44dd : SUCCESS in 1h 03m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90ec59172cd14630a8e03805d922dc9a : SUCCESS in 2h 07m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcbbd0638519496995822f64a5b59495 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74822e2fdc094bc4b4fba577ceb6092b : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1a6b219dc134931b54f76bbe77930f2 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1c6609dfcf0ba04ed526aa98631d0af996b22489","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-17 13:24:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"782b2174d3f5d85ed73a8846eec5cc94e0724316","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-07-17 15:09:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4058cb768506c007bc49728202406c7fd992e6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 15:20:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53c52b2f538b4c699007e4ec927a9e9b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4b96d81b999d4739b2042b39c5351990 : SUCCESS in 19m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e96f273c6c884fc99464143b7e38bf4c : SUCCESS in 30m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3545f5d8d1654862a7093462b722577d : SUCCESS in 16m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9f7423a48bdc4e3bb32f2f757178d0b7 : SUCCESS in 28m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3a6f160cdfc8411b8ed56d3c790eacac : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd3caf32f1b48d6943c99a63dc4644d : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9c6cbbbb08f4f19b53e19a6202fab50 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff69f818bce9416f8ef02fb17bdcaefe : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84315fdb4abb455bbbd5052fb83e7b18 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65e7ae3aa25483783abcbe359253392 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770d5c370d7b4eafaf222afcb723bf33 : SUCCESS in 5m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e88fbb75d4434fe68bd0a73ec68d29bb : SUCCESS in 2m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f8073c5c74b44559a26d2f418e5fdf3e : SUCCESS in 20m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9cfd5f6c744644679a4046cf0dbda8ff : SUCCESS in 24m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e02881dfe5c14054b3d0d0447a2347e5 : SUCCESS in 1h 09m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/252fd821947e4c45834c7924ff91814b : SUCCESS in 1h 48m 36s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/181693379b9042bea8668c647a67b7d4 : SUCCESS in 17m 52s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/262f6976b46942d2b4c99ef726609899 : SUCCESS in 18m 49s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/4821e0c8eca742c4b858a054da8528c9 : SUCCESS in 1h 01m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8aaee3ac48e24f949aa658c904e63f03 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"97e2ae6e04bf662c7a541117a6a3cb6e5a8eacb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 15:52:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72ec2d0eb4341ba8cc46465099d4a79\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/000301ac3d374a009ca66a0212fb3a2d : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/103512f1644a4a9f9e1fb0a47f1f1442 : FAILURE in 2h 06m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d0a717d8a5d45a2ab58649e5eb1e4f4 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75b0b55a1d974045993938cc727cf47a : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25be567cd17a4720855bff9965f26c5c : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"88c422ccb0fb97b5cb50a9427aaf709d61480655","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-18 10:56:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"083523e9bc4c39c024f91e5881a5b3e3bf86de9b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-18 11:11:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3f58b8afd0ad50c3f5229d8a3aafdeb1770aad5f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-18 13:24:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"962850745d9bbdf16fa6d19d8680ccd19d3716c4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-18 13:31:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28f11638c74bcadf284c319070a38236502b74a2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-18 13:35:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f622d9df54f9aa4ffce0a337c71da1cf90553c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-07-18 13:51:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"66307efc5220a26d206d45049364f38d861b1aa6","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-07-18 13:53:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d6cb09f3294e36e2ba3772897538d6469f70bed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-18 14:35:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"718689c36d63cb5f798c6061bfe59a2a3ea1550f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 15:31:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01cad30b95594138a9ebd11e4fd17a9e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bc5a3555c8904729b4e297c28baafa15 : SUCCESS in 11m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dc225fe4cb244b0481ea251d74d53c0e : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41421120e05b4905b07cfe4820546cc3 : SUCCESS in 18m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4d4b500dcbfa4ba79fefd0e5cee4e720 : SUCCESS in 27m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff60a20b7977463bb21a8ed54d133ad1 : SUCCESS in 1h 01m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec40196f85774ea3af5ce61b1f60a149 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6fb6dd807804bb78cba599f7845ca1c : SUCCESS in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5603286afc1745ddadbad946ecbfe091 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb763e33e46f49c09d3a8957395a706c : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2556ac5069f5407b9cbd54d83a3c26fc : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8b5c93c7c84016a805c3be2f684947 : SUCCESS in 5m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/aec179d414e8454aabf6fea55db908fa : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b2aab0d093e84980a1e33acd0f22c76c : SUCCESS in 13m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fe5265d031ad4a1db0e271d0af2380a6 : SUCCESS in 26m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c8e89212cd094f0da1b1798b53d44641 : SUCCESS in 31m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/521e921de57846d1b14df8c0631f2c55 : FAILURE in 1h 33m 12s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d43d4bfba0b4043ba7abc19b8660724 : SUCCESS in 16m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5967a52574344c1bed8ff7c5ad732a0 : SUCCESS in 20m 51s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/5d26996646cc4f93ba9610af38157f77 : SUCCESS in 56m 25s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/67e3fffd169d48ebb2603d9b3acf7d39 : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":3},{"id":"f27a7b8f7c50410b4eca4dd13208280e0df8a173","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-18 15:32:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dc1cf5d1a439c00312a3b96c21fdde4367d36ede","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 16:22:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6dc78f39ad24febbf523a32d50e4ed5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a21d1df80cff47fc8d26c9de47cdcca4 : SUCCESS in 1h 27m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdaad59bb0714a80b21f95f8be107876 : SUCCESS in 2h 19m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f372ff0ec71446e79adee494ece077bd : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd632009b47d42449a49f4f74c334f54 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e943a00ceb148cea6347bfa3a9ba367 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"ce4fffb84f6673a6b4f3e1fc30c33a530a094074","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-21 09:59:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3767b5347e7a92210d3cd204969743cbeffacd85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:59:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3fff3606e2facff586d3b55881fd620a6eec02e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 11:31:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20eaf3d038f8499f81c8af96c4b0484c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4ac3caa8181042f9b209fad629b9f0e2 : SUCCESS in 17m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3e48fd71c91b418d8508801d044ec23d : SUCCESS in 25m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0cc1bd55c7004dcd8aecfc096723c09a : SUCCESS in 56m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98e8bfb0ca149328f407ed805bd2d90 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fb003c105774698a0056a28b2aed400 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f7c42589d194e0a9cb897c2c1d5c2c3 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311ad16260e74a66afc632fe1b77b700 : SUCCESS in 5m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f9c89716723a4a79913794f7fb25adcf : SUCCESS in 3m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9a0ed2b3b2a2499bb324db5103815eed : SUCCESS in 19m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0a5d41ab218544f9ab70cfb356296fbf : SUCCESS in 23m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/89b622754670406daaac4bae57b83a27 : SUCCESS in 47m 45s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30f0c8fa13ec440bae71fdb2732b4189 : SUCCESS in 14m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01fcbf2ee9e94f7cb806d615db22cbd7 : SUCCESS in 18m 38s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/340c656584f64c64b7badbf2c561ec8e : SUCCESS in 59m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fec19c900d5a497e85f39277a01d42da : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":3},{"id":"696fd95377509d89b8c2d963d426218b79e73545","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 11:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1dc49e5df53f86e67b1a53e6b878b2b934535db1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 11:32:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/387cfca3fbe24636b8c08bf17775f828\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b424b7e9873b4b0cb930b277aff37ba7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"23a40140e946317f68b8a72ddb055de32c20f409","revisions":{"d5d7a6c89d4d68273d22b80514a761f038a38e91":{"kind":"REWORK","_number":1,"created":"2025-07-01 08:59:09.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/59/953859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/953859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/953859/1"}}},"commit":{"parents":[{"commit":"9f557299c64f4143f17b59d0dd4e443c052506d5","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f557299c64f4143f17b59d0dd4e443c052506d5"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-07-01 08:59:09.000000000","tz":0},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\n(cherry picked from commit af9f7cf13f22b1075c078e50cb40648771d20762)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5d7a6c89d4d68273d22b80514a761f038a38e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5d7a6c89d4d68273d22b80514a761f038a38e91"}]},"branch":"refs/heads/stable/2024.1"},"d0c05ccf30e2560923b4990e5b21bf1f5c1e8bb6":{"kind":"REWORK","_number":2,"created":"2025-07-03 12:44:50.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/59/953859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/953859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/953859/2"}}},"commit":{"parents":[{"commit":"9f557299c64f4143f17b59d0dd4e443c052506d5","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f557299c64f4143f17b59d0dd4e443c052506d5"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-07-03 12:43:48.000000000","tz":60},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n(cherry picked from commit af9f7cf13f22b1075c078e50cb40648771d20762)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0c05ccf30e2560923b4990e5b21bf1f5c1e8bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0c05ccf30e2560923b4990e5b21bf1f5c1e8bb6"}]},"branch":"refs/heads/stable/2024.1"},"23a40140e946317f68b8a72ddb055de32c20f409":{"kind":"REWORK","_number":3,"created":"2025-07-18 13:51:30.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/953859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/953859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/953859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/953859/3"}}},"commit":{"parents":[{"commit":"9f557299c64f4143f17b59d0dd4e443c052506d5","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f557299c64f4143f17b59d0dd4e443c052506d5"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-06-06 09:23:05.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-07-18 13:51:21.000000000","tz":60},"subject":"Build RabbitMQ 4.1 in stable releases","message":"Build RabbitMQ 4.1 in stable releases\n\nRabbitMQ 4.1 is released, meaning 4.0 and prior are out of community\nsupport. This patch allow for building 4.1 in the `rabbitmq-4-1`\ncontainer, while keeping the existing version in the standard `rabbitmq`\ncontainer.\n\nWe do not need to build 4.0 in past releases, as 4.1 supports upgrades\nfrom 4.0.x and 3.13.x. [1].\n\nSame method as in: https://review.opendev.org/c/openstack/kolla/+/918974\n\n1. https://github.com/rabbitmq/rabbitmq-server/releases/tag/v4.1.0\n\nChange-Id: I4a4fe7244ea0b92cf12b96f4dee8a7abd34028d9\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\nCo-Authored-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n(cherry picked from commit af9f7cf13f22b1075c078e50cb40648771d20762)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23a40140e946317f68b8a72ddb055de32c20f409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23a40140e946317f68b8a72ddb055de32c20f409"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
