)]}'
{"id":"openstack%2Fkolla~954027","triplet_id":"openstack%2Fkolla~master~I81ad9c5bc404adfdc31eb50fd98e636ae5312eef","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-04 08:49:46.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-07-04 12:06:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81ad9c5bc404adfdc31eb50fd98e636ae5312eef","subject":"zun: add wsgi script to use module paths","status":"MERGED","created":"2025-07-03 11:01:55.000000000","updated":"2025-07-04 12:07:32.000000000","submitted":"2025-07-04 12:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954027","meta_rev_id":"fbd4ecd9b34a55fa00dc05ded48ef04b61c62ac0","_number":954027,"virtual_id_number":954027,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-04 12:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-04 10:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-07-04 08:49:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-07-04 10:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-07-04 10:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-04 10:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-03 11:42:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-03 12:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-03 12:50:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-04 10:44:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"77e5e34278352700657fa90c67f30b58f077bfbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-03 11:01:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c78587cb0132818a20fb76029eefef56fcd0f35","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-03 11:42:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"44c442a0e053891ddcb75127b508bec86c83eb42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 12:36:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3081e62c6e642569965078f33531f72\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff118bd1333d4dc1a7f526f725de67e1 : SUCCESS in 1h 00m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fdda3ae131644c1182c2f0cf44f1b3bc : SUCCESS in 1h 17m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b7c13b812bd4de49ae2d1ac598cc41b : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f21313e35064b96904d3ec51fe82f7b : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ff9da80f67f4b4495c5ce81a0294d45 : SUCCESS in 7m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aad30ba456844fbc82ab6b2dc77b52cf : SUCCESS in 56m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbef59f652c014729d4a2892223eb78b7a0fac41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 12:50:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63e2bf735c18428b871ce1cf50527dd6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b56b102c1274f86999ed07ff11d3a7c : SUCCESS in 15m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4b8403746c144de49fabfa2c1c1b74f5 : SUCCESS in 18m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2671c208d57a40c98f4ff457a4f3cb86 : SUCCESS in 1h 05m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/feb904e4e1b74ce4a7287fbabd8fed8f : SUCCESS in 52m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d708d268e049d69318a756fd869c37 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b49e97da4e34c4ba8e22fadaf515765 : SUCCESS in 2m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d928ac25ad84928b4d70065d7c06bcf : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f2ea07dd5d545f29c2e5ae5561383bd : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042395f8cf9b44668b3595bec043b559 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32d37ed4a0646a9804572c996c7fc20 : SUCCESS in 6m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a3151acddd9a4a43a8274b1b7ed463d2 : SUCCESS in 1m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0546c2d96484253bd4a6cca6418ac45 : SUCCESS in 15m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0e53c3558168482d8a3bb47200ff28c0 : SUCCESS in 19m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/827faad8a99d401498aa69559fa25a07 : SUCCESS in 1h 04m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/37e39dbecf89449284d832876dd7acef : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2d448a00ff07a76fdc8ca7e00f9004db2f7799e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-04 07:14:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1d28405394bd2172d2c96237c1406a0ef64588f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-04 07:57:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d5743932c09ed83e71fe634687c23f0c699779ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-04 08:01:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a0ac0da78f1cb48db09bd19644d026a086f2eaae","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-04 08:49:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"49b12aa2a86deb60991159b8f9453e9231dcfc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 09:54:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fbddad70f341a3b946f9551a59619b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7150bbe4a9d2460b9fe6a7aa03ba0bf7 : SUCCESS in 25m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d9f120201654d7089db249778e9b469 : SUCCESS in 39m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22239991e2464155bdc4638d0221fcb5 : SUCCESS in 1h 04m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0a3f5376c11d41dc84f16e43898c11d6 : FAILURE in 1h 46m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8228f9f66d9d4da2b3ea86c4ae4ba7dc : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74b02eb2094e49d5ac04a86befecc28c : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/023b7647d0414e1e8b3e15a8a27b169c : SUCCESS in 1m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2364e5e7468423eb91f466587c94d49 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/474c9af60e9d4455bfdd38ba96ec967e : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5c983ff4794fc19c2ec0fcc553c2af : SUCCESS in 6m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d643fbc9ff94b128d02cb440b3d1b7a : SUCCESS in 1m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f9afe2303cbb4c138be9a74f60b190b5 : SUCCESS in 24m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/30a7515f9cf14a5098191b043da1ed73 : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/981b27cfc5fd4c77a06a5cc889c43a68 : SUCCESS in 1h 12m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7754e0d4b1724aa8978ae2c08b7bf064 : SUCCESS in 1h 42m 44s","accounts_in_message":[],"_revision_number":4},{"id":"3447e11d8d18e54a4ac6321332cb1ce91ba5829a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:37:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383a91db04dc4a88bbe43c4ba1b5ef13\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9aba7193ac0f42799060e11f0d68a65b : SUCCESS in 47m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1764c1e2b814c9eae7a288bfbb76f2a : SUCCESS in 1h 40m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1d2f9409bb954a25a90961cbfe5c2890 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d668d188f337460aab35e709bc4f3a8f : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/00ec3890fa4c4c86a2eba7e3ea2d1430 : SUCCESS in 9m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d0849727bac642bab73ceeba2bd260d7 : SUCCESS in 1h 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3dacdd8ef1c498561beb547809dc18ec37fcbfde","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-04 10:44:35.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d5338b0a3ed85f4ff8847771982c8f224227f5a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:44:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"02cf308582637e2b465cf5bc579c034ef581785e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 12:06:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/463a4ff48c2d4f968777d244d504360a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/37c7f56d03434b5bb1d783c9e0cf16c2 : SUCCESS in 20m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3205f48319e84c3fbff37ae478e18478 : SUCCESS in 15m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e5159aaf45c14eb795cd1bf58277e002 : SUCCESS in 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8caa4f899f514b08813746ecde4d32e3 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f0eb315df8a4dea8828c4af530ed31b : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98dac369e83848d5ab1f8fdce47afc20 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8625992f058c4f4e9150e289dd419a51 : SUCCESS in 3m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5d467584f10c4af49c75fa74c3e60970 : SUCCESS in 4m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b0f7c042bd884f02aa4233ed21fc4af0 : SUCCESS in 21m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1b79d7f18c274863aab00eb5c60daaf7 : SUCCESS in 23m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d1353436f81a437c945b27d940763476 : SUCCESS in 1h 06m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aaa6cf54094f4776aef9256cafab3350 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":4},{"id":"6ed967b14012b844a8e0c712a9c95b33ed170926","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 12:06:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fbd4ecd9b34a55fa00dc05ded48ef04b61c62ac0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 12:07:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef2a0a2744347fe91910873b622a470\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff6cd5506f0d42cba61109f491db9588 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bba9c3dd553abdfeb6017fa393007f6fbee13b3c","revisions":{"801855ec09f61216bef2ccf5e883769bc7b03d12":{"kind":"REWORK","_number":1,"created":"2025-07-03 11:01:55.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/27/954027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/954027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/954027/1"}}},"commit":{"parents":[{"commit":"b56331ca8e5fce8120cd3a8590f3e79021be3499","subject":"Merge \"Fix glance-api: add missing lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b56331ca8e5fce8120cd3a8590f3e79021be3499"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 10:56:34.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 11:01:54.000000000","tz":120},"subject":"zun: remove copying of wsgi scripts","message":"zun: remove copying of wsgi scripts\n\nzun has moved to module paths [1].\n\n[1] https://review.opendev.org/c/openstack/zun/+/953551\n\nChange-Id: I81ad9c5bc404adfdc31eb50fd98e636ae5312eef\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/801855ec09f61216bef2ccf5e883769bc7b03d12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/801855ec09f61216bef2ccf5e883769bc7b03d12"}]},"branch":"refs/heads/master"},"526735553f185372421fa63152e2b1aab84e559c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-04 07:14:26.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/27/954027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/954027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/954027/2"}}},"commit":{"parents":[{"commit":"09884de24da643208f867169a7be0a9ecd0a7e27","subject":"CI: add kolla-ansible zun job to tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09884de24da643208f867169a7be0a9ecd0a7e27"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 10:56:34.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-04 07:14:26.000000000","tz":0},"subject":"zun: remove copying of wsgi scripts","message":"zun: remove copying of wsgi scripts\n\nzun has moved to module paths [1].\n\n[1] https://review.opendev.org/c/openstack/zun/+/953551\n\nChange-Id: I81ad9c5bc404adfdc31eb50fd98e636ae5312eef\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/526735553f185372421fa63152e2b1aab84e559c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/526735553f185372421fa63152e2b1aab84e559c"}]},"branch":"refs/heads/master","description":"Rebase"},"522a4fffd32fb63cdeaad4083fa299edc3957c4b":{"kind":"REWORK","_number":3,"created":"2025-07-04 07:57:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/27/954027/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/954027/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/954027/3"}}},"commit":{"parents":[{"commit":"b56331ca8e5fce8120cd3a8590f3e79021be3499","subject":"Merge \"Fix glance-api: add missing lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b56331ca8e5fce8120cd3a8590f3e79021be3499"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 10:56:34.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-04 07:45:59.000000000","tz":120},"subject":"zun: add wsgi script to use module paths","message":"zun: add wsgi script to use module paths\n\nzun has moved to module paths [1].\n\n[1] https://review.opendev.org/c/openstack/zun/+/953551\n\nChange-Id: I81ad9c5bc404adfdc31eb50fd98e636ae5312eef\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/522a4fffd32fb63cdeaad4083fa299edc3957c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/522a4fffd32fb63cdeaad4083fa299edc3957c4b"}]},"branch":"refs/heads/master"},"bba9c3dd553abdfeb6017fa393007f6fbee13b3c":{"kind":"REWORK","_number":4,"created":"2025-07-04 08:01:18.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/27/954027/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/954027/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/954027/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/954027/4"}}},"commit":{"parents":[{"commit":"b56331ca8e5fce8120cd3a8590f3e79021be3499","subject":"Merge \"Fix glance-api: add missing lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b56331ca8e5fce8120cd3a8590f3e79021be3499"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 10:56:34.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-04 08:01:14.000000000","tz":120},"subject":"zun: add wsgi script to use module paths","message":"zun: add wsgi script to use module paths\n\nzun has moved to module paths [1].\n\n[1] https://review.opendev.org/c/openstack/zun/+/953551\n\nChange-Id: I81ad9c5bc404adfdc31eb50fd98e636ae5312eef\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bba9c3dd553abdfeb6017fa393007f6fbee13b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bba9c3dd553abdfeb6017fa393007f6fbee13b3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
