)]}'
{"id":"openstack%2Fkolla~954147","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I87e4135425e721f2f750402b3fcc6404643a3bd3","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-07 14:03:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87e4135425e721f2f750402b3fcc6404643a3bd3","subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","status":"MERGED","created":"2025-07-04 14:50:14.000000000","updated":"2025-07-07 14:04:44.000000000","submitted":"2025-07-07 14:03:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954147","cherry_pick_of_change":954019,"cherry_pick_of_patch_set":10,"meta_rev_id":"0a48f0fb1ebf030d535a868b505cc892a9732f5e","_number":954147,"virtual_id_number":954147,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-07 14:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"all":[{"value":1,"date":"2025-07-04 15:37:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2025-07-04 15:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-07-04 14:50:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-07-07 07:11:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-07-07 07:11:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-04 14:50:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-07-04 15:11:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-07-04 15:37:11.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"},{"updated":"2025-07-04 17:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 07:11:56.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"79cc2fed04f7120351690f4f77bd3f6ba4ec5f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-04 14:50:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f14feeceb1905a4c61b7332ca757bbf7492ce913","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-04 14:50:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"36606ac102a896cab6387ad372741a1247c689d1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-04 15:11:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"644d6c179254fe49e3d0c3b73ec7a98122432e71","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2025-07-04 15:37:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bbe6e2ca56f40dd7334fca119b6a0eedf499b540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 17:18:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe9402b1a8fa4739af588850faf1445f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/92a90abfb7784ce2ada85114ffdd3255 : SUCCESS in 20m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4fa23f982f304c6aaec6fecf6bb45eb7 : SUCCESS in 1h 05m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b2ef3631b3d4f4eb4719b8b552e914e : SUCCESS in 17m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1d39469512924282b530c1b6e0399e8e : SUCCESS in 30m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/72f255c567504fb59ea8c6d9d7707f45 : SUCCESS in 1h 00m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c4907fbd0f54467fa0b88c9ecb8a44e7 : SUCCESS in 1h 36m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4b7c1041f7409abe1e1383ddd61572 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09264c3d9b514adb93cddc335812c0f3 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bd895ad38e741859c31fc78f64a31e8 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/399f009c7e294eafa0ecef991367314e : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff11ccfae71b4aa0a9d80f0b0c78437d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7f30374b7843a596192085215a1e94 : SUCCESS in 4m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8da2f47fd42c4aefa55a264780a2dd6a : SUCCESS in 2m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3d4a8e673606451488cb48aae04a88a2 : SUCCESS in 19m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fc4dcf9aa2934f23a3eaf62b1110ff3b : SUCCESS in 28m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ecd171acf36d485e8dea6b8fea911670 : SUCCESS in 1h 03m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dfb0037711954b948dd72f57c2e8f4c2 : SUCCESS in 1h 25m 23s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/63e416e9dea14aa29c3278266f33ba6d : SUCCESS in 1h 38m 57s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5fd1db0f39b4a13adb6365c8791fc7c : FAILURE in 1h 23m 18s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/88774bb3139d408aaee72321d178d1af : SUCCESS in 1h 18m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7224a924662e4a70860d9c23af63d161 : SUCCESS in 22m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a5620fe281724f6e9c7c7aa4f8f9836f : SUCCESS in 27m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/58a863fabd5b4a0f9f2330c756c68588 : SUCCESS in 1h 00m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/da1d3962cc284569b5cb032c7584c78a : SUCCESS in 1h 22m 46s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ad7b579002cd4f2ca4c7ea58b5919451 : SUCCESS in 1h 39m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bf67777762fc4cfe8ff25ddeaf4cb919 : SUCCESS in 2h 12m 26s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/507be64a0b4947f18ed642ae6e2d1022 : FAILURE in 2h 08m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf655a847b2d50206d5e8b5fbe2b1defe6857dc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 18:32:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9ed967e94f04b66b70790135c407099\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7ede994c36e2448aa4e62655e81a98ea : FAILURE in 46m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2c2a71e080d1487a923966cd51b08dab : SUCCESS in 45m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ccdd128f97fd43e09212ce442a6a2b60 : SUCCESS in 1h 46m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/709473718eb74e4594082de314628c72 : NODE_FAILURE Node(set) request 300-0027359563 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74e54da898ab4f6eb915eb844a65c48b : SUCCESS in 4m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c3fa8e46cf6a43149a0d1c7552d64d1e : FAILURE in 45m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6cc4066a1de34d308109469345f7eb2a : SUCCESS in 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c45d807c0ab18c1614f1d551e09a23a21dab1f2a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-07 07:11:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"daa242a0e568ccb1b244b26de8c399aa9fd81a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 07:12:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f962682745abc0a427e719e6885afe4b8e3f2fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:03:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbad1107d0cb40599d8525135e236d75\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d2647a9eadf44f3aaf5810e3b3e4f9fe : SUCCESS in 14m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/688e4a066e8e403281b7c37c3d386cc3 : SUCCESS in 20m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b25838777a964883920c14de1c67c48f : SUCCESS in 59m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0b656e94dfcb48d08b66e62269adbddf : SUCCESS in 1h 35m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfe2751205eb4613b05e730e40dfec06 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2c420a893084a1b92f07d86fbb0df55 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cfaa9447abc430c9a613d1d1ac88f0b : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcfde985453c42238f62fc33889c5ea2 : SUCCESS in 4m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1d8ee186f0a742e58d56fd263ee3f26e : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3e64fa5c98b045f49b8b3bdf6be86206 : SUCCESS in 18m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5fa7454ca0544cfcbf40b16ccbb4a1eb : SUCCESS in 25m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4cf01eb67f3d478986f0426c8d48c21a : SUCCESS in 1h 02m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e3275623d2e443ebcac0dade98b9a38 : SUCCESS in 1h 28m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b82caf6daa3d4344be93e2439480b814 : SUCCESS in 21m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fe19b45fb4d54377bcab850023c2361d : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fe29bebaa5ca4434827c7c359d7d4207 : SUCCESS in 1h 01m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/16db4dc43d7e43bf8a20154c6a1a42d1 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"67f7504c91773fd452dc2765a5d4d20b9bf46e99","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-07 09:40:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3de0b4fd62e78ddba339088c0a315182c0197d83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 12:10:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80dead25952642abb3fb75151a793fd7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0aa7bfa34bd9402e924260d1ccb3c757 : FAILURE in 45m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36d64443dabf4266bd58561f0ce3d980 : SUCCESS in 33m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cbab46988a6a4b96a4088f2ef655faef : SUCCESS in 1h 30m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c65beaf4876a4d7d87c893f8d3c797f5 : NODE_FAILURE Node(set) request 300-0027361488 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d80cb8d67b6d462cb5b27a75511b4697 : SUCCESS in 3m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4e27034b328f44ac9d8f75d642a763ab : FAILURE in 1h 18m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9ccd64555a4d4aefb525699090c6fb88 : SUCCESS in 1h 38m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d00fd8ea1b8d1bd5c5512692544ad42706094bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 12:21:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53bc79b9c0174e0ca07f314d2af370bc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2cd952c5a53d4f5ea8f72db5f94684e2 : SUCCESS in 20m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/854937499c794b17a660492255278c92 : SUCCESS in 1h 00m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cc1e61c389204aca8092f324ffb278e2 : SUCCESS in 9m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0ca062ab724043ea8f071ac009636cd9 : SUCCESS in 13m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bce65c7e5e5242a1b27023738eb60869 : SUCCESS in 29m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/29d84f33b8ac40729c8a6f4aa4e788d7 : SUCCESS in 1h 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c310d7ff3d43c8ae86b6afa04132fe : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3087e44381814962a1d15ac1753b500c : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7dddb64f21f141e6aaecb7027762ead9 : SUCCESS in 1m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bd65b4e536a4eeb9a25f17d0ebadcee : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75776c2133964f2da88bbd5140002bc8 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6356ddd1f06b46fa9ec1ed692d0a39cf : SUCCESS in 4m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/841491bd47c049b993c76a944c9734cc : SUCCESS in 2m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f40bcf536f2241ceab809805ef916302 : SUCCESS in 20m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/192c435835904a368ab4f2e69671a472 : SUCCESS in 25m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fb122f986af94020b07db46d08aaae80 : SUCCESS in 1h 05m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9d0f1ca7744d47a68fda7dc6eaacfc41 : SUCCESS in 1h 21m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bfafb3b2fe784004999305c7b56ac093 : FAILURE in 1h 21m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6a38d20401554f609986967e4aca2f7f : SUCCESS in 2h 33m 26s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a22b4260ea354a59a66547eb8245c235 : SUCCESS in 1h 21m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/28142abc885d448c89eace6ae090bc7a : SUCCESS in 18m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1fb3c16290c843d5be438d13d5f4d97c : SUCCESS in 22m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9413ebb1528c47bdafdebc1aec45447b : SUCCESS in 1h 06m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0863201cdf9041e794b25f228fad078a : SUCCESS in 1h 15m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/32b36b65671c4a7b946562e90ef99aa7 : SUCCESS in 1h 39m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cddfa45da8994625a247168b304234a7 : SUCCESS in 2h 10m 45s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b6645ea9d82f46fc8dd657b910856ca8 : SUCCESS in 1h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae876d0b8d86d05733455a4eeb4ab9e370ec755f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 12:21:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"729969531c04271a9b0385b17ca7e29cca712203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:03:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aec4d5dc2244b748af245732ec7e26f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1979a7fa27146cab4a43f6e927c9d8e : SUCCESS in 20m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0d9aeccb181b47b98285d1745138a23e : SUCCESS in 31m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1be41042540a47e9a7d567f644cd6049 : SUCCESS in 55m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7567ec0987114913a1c1810160b1d292 : SUCCESS in 1h 36m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fda9dc003fa4ba68d1a090c91481871 : SUCCESS in 1m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/516bf5944c994d3bb0e5284741a43dd0 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/546d7f415ebd48f6a836d31afc1d1eec : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f94a68bc9614eb9b16652b5d996e6c9 : SUCCESS in 4m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/389abb589d59473f8cb3c62bd422e6ea : SUCCESS in 2m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b332da70b7174e319fb8d31c39ef5ece : SUCCESS in 18m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3a3a7272900b45389615220abcbf168c : SUCCESS in 31m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c3c7dc7c2e2345fc816139be1f8bcfe8 : SUCCESS in 31m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ef7df53b6fc9436da8b9542b2e1adce5 : SUCCESS in 1h 23m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1a7601d2766a41a4948af9a2c8c31754 : SUCCESS in 12m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5cddd86d5a9d408da299daf93ad85c8a : SUCCESS in 25m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/de0850f7e5814e3ba8dca0f690d22592 : SUCCESS in 1h 06m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/093913aea4cc41bf8a3bbf426297f1ed : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1c3413b5e5c271dda6cf210e3501862e432d4d72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:03:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a48f0fb1ebf030d535a868b505cc892a9732f5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 14:04:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518beb6afa504b26825595932292487b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f1dba53ab947bb8ea2baae815efae0 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08637636770ef35982e3577790014a5a492a9aaf","revisions":{"08637636770ef35982e3577790014a5a492a9aaf":{"kind":"REWORK","_number":1,"created":"2025-07-04 14:50:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/47/954147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/954147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/954147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/954147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/954147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/954147/1"}}},"commit":{"parents":[{"commit":"cb204aa173b364d3d5021afaa7995a215645791b","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb204aa173b364d3d5021afaa7995a215645791b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 07:18:45.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-07-04 14:50:14.000000000","tz":0},"subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","message":"Enable EPEL repo for Ceph deps in glance and gnocchi\n\nCeph Reef client dependencies now require EPEL repo on EL9.\nSee thread here [1].\n\n[1] https://lists.centos.org/hyperkitty/list/devel%40lists.centos.org/thread/BZM3RO2LNAMB47FBT2QO3ZOSGBHPCO3V/\n\nCloses-Bug: #2115448\n\nChange-Id: I87e4135425e721f2f750402b3fcc6404643a3bd3\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 62a69190cbc125f74e2f72802f98f9d574609d49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08637636770ef35982e3577790014a5a492a9aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08637636770ef35982e3577790014a5a492a9aaf"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
