)]}'
{"id":"openstack%2Fkolla~954148","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I87e4135425e721f2f750402b3fcc6404643a3bd3","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-07 09:03:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87e4135425e721f2f750402b3fcc6404643a3bd3","subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","status":"MERGED","created":"2025-07-04 14:50:31.000000000","updated":"2025-07-07 09:04:20.000000000","submitted":"2025-07-07 09:03:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954148","cherry_pick_of_change":954019,"cherry_pick_of_patch_set":10,"meta_rev_id":"0a2b08cfc44b8a60ce4db0138ea3c9532e2890ff","_number":954148,"virtual_id_number":954148,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-07 09:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"all":[{"value":1,"date":"2025-07-04 15:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2025-07-04 15:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-07-04 14:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-07-07 07:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-07-07 07:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-04 14:50:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-07-04 15:11:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-07-04 15:35:37.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"},{"updated":"2025-07-04 16:45:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 07:11:51.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"900f72fb97e27a751f7f2cb8197cdbdf99496dc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-04 14:50:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"30348f30a95283a207992ad80de079b13e48f6ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-04 14:50:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"06e563487de86f80f2c2cb1ed67c24260029bb9d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-04 15:11:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2028c04ee23bdafa62b033744e4b744fe6354a7","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2025-07-04 15:35:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ded59e9b823ba752bb90b363b6d56dc39591608d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 16:45:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04583c07cb1b4c41a40858f0c84a06ce\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b95ae3683a464af9994f8ba082774d9f : SUCCESS in 27m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/60d5542753d847659344bdada02884eb : SUCCESS in 1h 07m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41592877057c432aadf96578b934a2ba : SUCCESS in 18m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/08feb39e2b3047a2a5012c7ddc0f458b : SUCCESS in 36m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7cb4f6ec852d49e48a8feaf5c9772166 : SUCCESS in 56m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9826b06bf92c44ca91977155d01ba8e9 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc52e81dbf144f54987d8d4c69722589 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1d229fe03b0466e9c69ab4572fec31a : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f11d5512121b41459fd8ccf41702799a : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/754076ff590c45a59fefd5453d5e0cd8 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9ccaaade164b43b0c53fb167a72498 : SUCCESS in 4m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8e594e3e910f43d68ae90f2ba63dc831 : SUCCESS in 2m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f4054c9412e04b32ac99d0d881c1d665 : SUCCESS in 21m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aa523db817494ff0a9f5ebe70332b41a : SUCCESS in 26m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0a4a8cdebd9144b4b453bb7dd9752865 : SUCCESS in 1h 02m 08s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/38e77fa1b29141c5aa96f153d11ca52a : SUCCESS in 1h 49m 28s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/928246da328b47328bf2f986ec91b3a7 : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fba670d9011c4c729e8efb8a95e1d69d : SUCCESS in 9m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc78aa6c53464896a52f61e62b0a52f5 : SUCCESS in 14m 04s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/2bb82f27877b49b9bebfd4b0bb6a8a79 : SUCCESS in 1h 05m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c03e5f898e046cdaa887519716d0ac5 : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f252e283bf58011254c4e3c47b51ed0f0993898e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 17:59:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da9a8c638fe4b679a53159384af4910\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a121e7d3c0494d4cbdd91439e06945a7 : SUCCESS in 43m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5e4c17d5ae6e4820910504d52dee681b : SUCCESS in 1h 41m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f2df90c6331406e96c53c1cf8e57903 : NODE_FAILURE Node(set) request 300-0027359565 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f033491319e9433aad81fb3451f6cefe : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f923f35d9b5b454eb2f262bd92764a64 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7b4fdd1f9f0fb70c2363ba138e42239ace9e4a79","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-07 07:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b33f287806d192fe160beb7f43984299d9be689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 07:12:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e832b3bf82df57bc006b2873b38b9ad4bd2b91d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:03:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c2cb5366c24ebfb03603c17cdfe230\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/020cd269778d4d34aaae930265890ee8 : SUCCESS in 18m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/06c712116e2643c9b95b4b117ce1256c : SUCCESS in 15m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b6a18fc23f3e4704b235e99bb0dedcf5 : SUCCESS in 33m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02004f72e4494f149aec92d46f362fc9 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65a3d0214b604354b1be5698b2bad58e : SUCCESS in 1m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/080cde171d3b4250b2b70520e8bf8a71 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fbe2bfd2594ceda84a5be524159876 : SUCCESS in 3m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a100a5f4bada4dbfb17b48ea050299ad : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a4e8aedbd3845989c945b7c0ccea441 : SUCCESS in 16m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/669df76f66ba4407a9af25a34e175067 : SUCCESS in 15m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0463305bd1cc4ca198526e89649de3e9 : SUCCESS in 1h 04m 01s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/802adc0b9a244b4b90ccb633cfe0d575 : SUCCESS in 19m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cda8b03118143fbbaab32a7325f6e4a : SUCCESS in 19m 47s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/b40ef61f5820413cbd452a8ab5acb541 : SUCCESS in 44m 12s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a3998012037e42358beaf97438e2bbcd : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2efdc481c36fb864f5911b0079afab1f611e1b1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:03:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a2b08cfc44b8a60ce4db0138ea3c9532e2890ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24049d79d0b64bfc852539d9643e2706\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656d8f89465a466d8c59b48e5d0c4cab : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92c9782e83f5364907d4093482d350099c7de488","revisions":{"92c9782e83f5364907d4093482d350099c7de488":{"kind":"REWORK","_number":1,"created":"2025-07-04 14:50:31.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/48/954148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/954148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/954148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/954148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/954148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/954148/1"}}},"commit":{"parents":[{"commit":"9f557299c64f4143f17b59d0dd4e443c052506d5","subject":"[2024.2 and earlier]: horizon: Work around pyscss bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f557299c64f4143f17b59d0dd4e443c052506d5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-03 07:18:45.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-07-04 14:50:31.000000000","tz":0},"subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","message":"Enable EPEL repo for Ceph deps in glance and gnocchi\n\nCeph Reef client dependencies now require EPEL repo on EL9.\nSee thread here [1].\n\n[1] https://lists.centos.org/hyperkitty/list/devel%40lists.centos.org/thread/BZM3RO2LNAMB47FBT2QO3ZOSGBHPCO3V/\n\nCloses-Bug: #2115448\n\nChange-Id: I87e4135425e721f2f750402b3fcc6404643a3bd3\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 62a69190cbc125f74e2f72802f98f9d574609d49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92c9782e83f5364907d4093482d350099c7de488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92c9782e83f5364907d4093482d350099c7de488"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
