)]}'
{"id":"openstack%2Fkolla~954693","triplet_id":"openstack%2Fkolla~master~I9d1f24b98f1d7c8f29b0dba45105b0263f961ca5","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-11 11:21:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d1f24b98f1d7c8f29b0dba45105b0263f961ca5","subject":"CI: Drop publishing to DockerHub","status":"MERGED","created":"2025-07-11 07:43:04.000000000","updated":"2025-07-11 11:23:32.000000000","submitted":"2025-07-11 11:21:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954693","meta_rev_id":"871507ce4a8a45409b4358fa8e3afc0fbe39714b","_number":954693,"virtual_id_number":954693,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 11:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-11 09:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-07-11 09:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-07-11 09:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-11 09:26:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-11 09:26:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 09:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-11 09:26:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-07-11 09:29:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"909dd06ec65fc3d43d895480eee5e9aeb960630c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-11 07:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f19a3f7c70574ced118e2be262deb7d4d5d90c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 09:21:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973ded7f5afc4a4889f3acdb58fc744a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b92c101facd947e89fee345c40927644 : SUCCESS in 21m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/18f5ca99c13c47018a05a8984f973bb7 : SUCCESS in 25m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e7940116d9cd4b6287b87778b34d7242 : SUCCESS in 58m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8a1f7622ddff4e23bfe5f90438ee4d5a : SUCCESS in 1h 28m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e30ee2ce9d4eb8bb03732b09a3b35e : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/675e9a501819494da3beea6de869d0f6 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da7f93e258e5406da74e22e9abdb60e1 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08040496283f4af08457791fb3a2b213 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7a5a9d4bd5744339f96e974fe24db11 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d54e072e6e4adf9919b54fe4de8321 : SUCCESS in 7m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/54436dea402e499da8a3ba349c6c0fc1 : SUCCESS in 1m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b53d5abd5f7458ba9c40a704638af06 : SUCCESS in 10m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ec65eb7dd8fc4313ac553a94a10071e7 : SUCCESS in 24m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eb373e021f15484daf92c5080afd4fd1 : SUCCESS in 1h 11m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c049a08549de4c3483aebb3bc9438421 : SUCCESS in 1h 33m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d3ee33648eb71f7f12c3f9bfbfbd3518083c5c78","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-11 09:26:28.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b8a5d47eda9aaee22d563af445de1c1f0ce398bc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-11 09:29:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bade02eaf5d7f5134ad3809f22883ef4ab8e61a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 09:29:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dc2ce092f03fdb6dbfc16bea2bea408326187ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 09:59:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9192682af5644b9f8272f4e91c503c5a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3b1a033ac7bd4fc5b9b9a7eb19199319 : SUCCESS in 1h 15m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cbcea817ed74475c9a466c5cfee3473d : SUCCESS in 2h 13m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52f7d6f75ea446a4a81873d73505c4e3 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa1d638237804acba4ecb5d74cc7b06d : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d5fe34708314fb280e96d4b8227f4e3 : SUCCESS in 8m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1a9795b539aa4d778beb6bc7f840d08d : SUCCESS in 1h 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a2bbdfb536da5b5babf4c02bc826e61582a9264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:21:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b86092f2537416aa2b8465cecd8c3b3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4b4d5905c43d4339a3d6616ef4f7a788 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7f1728f632ca486c91a1400902f34565 : SUCCESS in 26m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0ec81b746d664689b90d8ad43bc94b76 : SUCCESS in 1h 08m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5790c6b810084accbea45ab127e7e5db : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbd5b55231d140148a42e7967ec18e7a : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51823a8faf224d999c375cf839596681 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b9ccb485f2431b9d6dd362a589c5e8 : SUCCESS in 5m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a3db2855a351495aab54c6600eecef11 : SUCCESS in 2m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30800ef3d8f34d51b3bf78df47783800 : SUCCESS in 21m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/78b9c2034a954640a4157f889428b2a6 : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1aee5a53fd1a4871895af3ac07e455a7 : SUCCESS in 1h 12m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7117bb97e3684b83a58bc89e8e4dcbb7 : SUCCESS in 1h 38m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a4fab89eafd2959fdbfc532db8276521d9231429","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:21:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"871507ce4a8a45409b4358fa8e3afc0fbe39714b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:23:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e927dbe5c5b441d82d7b8d4848a597a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9292db2aec344a7c9081f6024b2de395 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba2b4bb56415b9bd7b690dcf077bb59cf4083f05","revisions":{"ba2b4bb56415b9bd7b690dcf077bb59cf4083f05":{"kind":"REWORK","_number":1,"created":"2025-07-11 07:43:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/954693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/954693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/954693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/954693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/954693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/954693/1"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 07:42:15.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 07:42:58.000000000","tz":120},"subject":"CI: Drop publishing to DockerHub","message":"CI: Drop publishing to DockerHub\n\nChange-Id: I9d1f24b98f1d7c8f29b0dba45105b0263f961ca5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba2b4bb56415b9bd7b690dcf077bb59cf4083f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba2b4bb56415b9bd7b690dcf077bb59cf4083f05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
