)]}'
{"id":"openstack%2Fkolla~954707","triplet_id":"openstack%2Fkolla~master~Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d","project":"openstack/kolla","branch":"master","topic":"kolla_uwsgi","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-06 16:27:31.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-24 12:15:56.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-06 16:03:35.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d","subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","status":"MERGED","created":"2025-07-11 09:57:17.000000000","updated":"2025-11-06 17:43:27.000000000","submitted":"2025-11-06 17:42:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954707-kolla_uwsgi","meta_rev_id":"d9f19249241f21b02e9652a9f4bda8e312d474fc","_number":954707,"virtual_id_number":954707,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-06 17:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-11-06 16:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-10-24 12:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-06 16:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-11-06 16:27:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-11-06 16:27:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 11:43:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 20:47:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-19 08:35:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"4d22fab05a17b36fa7129fc09f4238c4f76ecfe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-11 09:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc36306f33b23fd1f97f10cfe1efbfa084327bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-11 10:00:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d50f85f29a38a2c09bae818360abd24cd06410a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-11 10:16:39.000000000","message":"Topic set to kolla_uwsgi","accounts_in_message":[],"_revision_number":2},{"id":"23e93dfb9613e93c30d18886194702d94a80dfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:43:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b64d4f830ba94eedbf4fe9ed188d529f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4cfdfbf8289f416699b56db62c5cbe08 : SUCCESS in 18m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c71c80540243418395731511c892e4a8 : SUCCESS in 32m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a7637dc60e0e4fb59792dfcf8d9d5d07 : SUCCESS in 1h 05m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6455c425e96432a99afb4e6c4c8adb5 : FAILURE in 1h 31m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f84c87029e7c4542b1d8390701eb3255 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff4256855c33422ca11893dee66325e8 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5abb7547b47411ab7bb2e28116cc76d : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dadce18aaf214ead8087954497f77871 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f54777e690704c29a961d8270cd795e8 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9487cd11065a4e5c8424fe45103a0d17 : SUCCESS in 7m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0fd2c2560f9247fdb562649c97a7c384 : SUCCESS in 2m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ce2f392a38294600a98959d627a176c2 : SUCCESS in 10m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/19512b472dad46a1949f2541fae7c38c : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f7d6bc4dc0f14c4f877a18f4b61a0054 : SUCCESS in 37m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6b87e2a8fb16435197334b379982af73 : SUCCESS in 1h 38m 07s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/f74f5c8a57ba4042a0e3946d018fcb7b : SUCCESS in 1h 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"917233078026acb4ad7c0ae2b2ce6065c1f69bbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 14:20:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8cfb66ff6745a1a7dad38cede9e37e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f0e4f095310401db9e1883b5bdbbc5c : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42210b85890e4923988e9cce9f523964 : SUCCESS in 1h 14m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a01daec697a46a8bb78a9b58390f59a : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1abe58d9f11846198e395bdefb3c42aa : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96170c01021d457a9ae2e3a3bdc88f71 : SUCCESS in 8m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7479f11542fb43ad9651e72f5baa0586 : SUCCESS in 58m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"780b220a03f554f88a424402cfdaad32d7542f37","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-16 20:48:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"80830998af893c01ab6c68325e59797736d206fe","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-19 08:35:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74d453aac5481f2143ce8eaf5ad23c3cc5a41118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 08:40:54.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3ed31a34a21b7b634a507a3a2fe7e97d665b6813","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 08:41:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a89a14c10b2a6d52c1f3415e3b0d8456b187851d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 09:57:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239091c31ba14f21a5607d2b3eea8736\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/c9abc40897084e29ae16ea24b66302f1 : FAILURE in 10m 42s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/88e1e1b6a3a74a31969858a06d3f64ac : FAILURE in 9m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0ad4cb10d873451894599e84403fd11c : SUCCESS in 23m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3e9093f8a0114a4ca3e88260c0c39f71 : SUCCESS in 34m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a019fd2e192b487f826bad370efedc88 : SUCCESS in 44m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1293c89a7e2f460fb4f91f819740541e : SUCCESS in 1h 04m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39e3465d54d42a8826fe6380681a681 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5812e2f647a4418b9c8c944efca6cf57 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f0f6dc2d7d2424094649bf61957a5c9 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7fa22bd0c4f4dcbbd33e3933d9bdd8c : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79c678b8c11946949e3d32e3574c7c3e : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a26254c91d464fafc07f63474429d3 : SUCCESS in 4m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1bf23c91d5a84b0187a68b7342d8a181 : SUCCESS in 2m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/97d846eaa8bc43268e6894648f45cf18 : SUCCESS in 21m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8fb0b07d7a714415b185c5267012e213 : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/13c7bda22ed04f50980279487fc94158 : SUCCESS in 50m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e7bae9b9a50d4b959613c4c0e250c02a : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f6d8de7b5bf9827ecc7a1155c5b085095b0f1197","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 12:11:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d18307a60b548f9b3dd3d66049b6937\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/47e63672d0dd4af28577202f177ffcd5 : FAILURE in 23m 20s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/202ab5324bde44e09d391d61520da24e : FAILURE in 27m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/43497b8d95974fabb5f9bff56dabedb9 : SUCCESS in 21m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a55d794ec03448a812081d8df22c6cd : SUCCESS in 2h 01m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1520494f5ec9496cb302d780f3f6a641 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fff8ad5eb0e44f469b26fe0436ee22ff : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2841d2267b64cab99920fcfafabe7ff : SUCCESS in 9m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bcd296bc23f243cb9584a2ff9332d05d : SUCCESS in 50m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e21e768e026d757d41719aca2e21597958fc8a31","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 12:12:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ffbf4e9e7c36b728bf3503ffaa649a43ca09afa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-24 14:03:12.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6bc49145c5a6d166234dfbfbc9cd9f5640f41b5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-24 14:29:37.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7d835969c7a568cf6bd4643c5c0fbfdc46171058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 15:52:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4920d363ce94c0bafb7400cec5cb3f7\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/9e82e724bf514d67951c65a24642b080 : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/84c17e2dd4224635b8cdfef9f02d9ef4 : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5f7b190ead314d228aaa4513aa10c287 : SUCCESS in 14m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/60fee394372a4f75a7d255cd44c3f13b : SUCCESS in 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881e310f2bf94722b7c2a7716cc4717c : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eb4f316b57141f8bd8e209db4cb08a7 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1666147c045f4e358fcbd362a9577570 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8d83c1dc3834d3682c18e8cfd4a3545 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2cd823bd3204d37b33ba44fe60fefa6 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ecd2f2ec7af45b19cb17574514c540d : SUCCESS in 7m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f6b886fffb5c44c49a587cbba3fa3c2f : SUCCESS in 2m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/77c52c00fb8343859286d0ad6d4f5185 : SUCCESS in 21m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/144f0b83f79b494d9ebc3e95715d5f6f : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":5},{"id":"c7f198a881527539267d276d898f7544570a40df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 16:42:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b00cdeefaffc43f0808ef8da7ae8b2f1\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c5e397b8680e4f2583ff703cc730969c : SUCCESS in 33m 35s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7430e57818c44604905a660587c7d94f : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2566e078eb494d4980bae69fefa0a1f1 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1048e771e2d14760ab139883b13eb10e : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e51b17b58e0148428a8aff748afa2630 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5ad44afdfb1b4ef6aa08dc066f384897 : SUCCESS in 8m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/22f8bd3dad974ccab61d0a5143a68ead : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"97c846238c2b761ec4becd22f01b0e802003b4ce","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-24 18:14:08.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f574302e23b689995ccdfb07ec2e80caf1937690","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-22 13:26:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"33a18a5435b298e84b0a9891fc80683e30e3f001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-22 14:20:39.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a5e7627529e4219edf7a33d92aff00404a6b5104","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-22 14:53:30.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45a26526139d49f22e5b9c71bf6c8bfaa74d16b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 15:06:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07c8ef2ed98418c9ed706a3f5914dad\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/63e2eced0e7841babbf190d27b410d9e : SUCCESS in 12m 44s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/5dbfb0c18ed14a93831cdd9a867b5de0 : SUCCESS in 18m 53s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/a14916cddcff4786ba27fed8bde56ea2 : SUCCESS in 26m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2e4419ba34f54f19b6c27feef2b1617b : SUCCESS in 15m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0fd5b01f84b841579d23c06bf9096446 : SUCCESS in 44m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6af6eff0b5324e4c952415db5db39eb0 : SUCCESS in 24m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3a3c8514669949d4b0563d6619098582 : SUCCESS in 30m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0a139652dc4f4c3b8517cb584277ad4b : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd0c8d2052a4dc5b19926720f52b364 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a61997020bba4955a69fb5aa16442c16 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4f03d3d55f3401a9a4e7cd541bab025 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1809659ed394d81aa5bd38e0b1d89ce : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ec2254819e40efaaca90434956ef96 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":6},{"id":"28f26298572700866cbc41ccbb39ea6aa6e6cfe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 15:54:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56974d22176642c3a21fe2bd3f04aafb\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7214d2361d7945b09267e141e3b60567 : SUCCESS in 1h 26m 13s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/a8005cd1ddcd4a88b9e70bab9e5eaa89 : SUCCESS in 44m 28s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e52ea5d6c7c6405e955d5632030d2ae0 : SUCCESS in 40m 41s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3f27178d7b2641a0952cc8560bdda823 : SUCCESS in 1h 28m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b700928029eb4ad6a346b8fc28ab1f5b : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9fe702dd87f4f4cb762ec573f3d090a : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/715cdbd97bc24c0993451eae6511a9de : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2440aaf462823eb741f26ff00f683e4b8f27b904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-23 14:17:29.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c04a54293b411efefb4dfecb10c47b79e7980b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 15:37:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e3359caf32541acbd4aae177ff5322c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/28d31b55fe54482992571e9ae0e246cd : SUCCESS in 15m 41s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/074704ceb7854616bc83dca8a92216cc : SUCCESS in 17m 13s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/272b29e576a74d8496580d022a2145be : SUCCESS in 15m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5780f958b78e4965aec86cb2614ad50b : SUCCESS in 11m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9a12c11c1ad6402c9f7f1391e44c41c2 : SUCCESS in 53m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57c0739ed2894bfba2ee071a1c0bb861 : SUCCESS in 22m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc730c048dbe403689198293a9fefd54 : SUCCESS in 30m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/07b4b4dcf2d64e099f0271e2b7ee791b : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e75acab914c4280ab3ac1871e47aa92 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5bf145229114396af6a578d3fe3a618 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97a3dc4e7c93437db018e10e99daafc7 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/809ab93a389c495f88ef4b196af976eb : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8892b5e7260344a9950fd4d8d2f868a8 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":7},{"id":"bcb6008e40c45c42f33a938b661111207291125d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 16:59:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc9036bcfe374d4598d5bdae5116aa64\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1e25b47325b244f2b5fe448670d41cbc : SUCCESS in 1h 31m 44s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/2a85ebc8c4494d818d64670cf218f187 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/3bedd9ef1c4e42fdace38b2b5429a3ff : SUCCESS in 1h 27m 18s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/88b08f7809bf4436bff6cbade8c4f870 : SUCCESS in 34m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24613c23e3e94a4f81c89875764cb1b2 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96b031461d20411b8d19e36f12f197cb : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d5508058c734634b2b27302a7fd6312 : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"15ff981b7527afc892e5a68f81a4a7658b8a4365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-24 06:33:56.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"93c11250e0c88239b1ebfe6c443ae61b5572dbe5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 07:21:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af312cb963f84d83853894095d35bc94\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/04bf729a75b54144a6249c68759ad4fd : SUCCESS in 25m 48s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/7c8189660a114350b09fc65bd52c1bf7 : SUCCESS in 40m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/977fb52876af455db44cb25dadb7fce5 : SUCCESS in 42m 04s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ddf9fbce385f46b68b23a069bf252458 : SUCCESS in 45m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f5b0123379a482f9047dceb50228427 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edd7b076f95240a8adb2797b806d1be3 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f34e00dae8b643fc84f3f0da33063cb6 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e93391ee7cbe1edd4426e92e017e3c11ea5e683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 07:23:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d058d97bee0546349792d2d266020aff\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/eb12093f3dbc4e519667114aa8c0e39c : SUCCESS in 18m 31s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/7e11c1ccbf144d949ce1a5fd637575b1 : SUCCESS in 19m 51s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/90e2b2cec80b49899cf3cbc09035311b : SUCCESS in 35m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bfb9d307b1494d81b9185a1c5b35791b : SUCCESS in 17m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4ca3d6e7b392400bad5a9870f30b0f27 : SUCCESS in 48m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bbaba4a1c2f4fc397c63b850a32c47a : SUCCESS in 22m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1a97b0e842044783b890756c3ea53edf : SUCCESS in 24m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe8011e7359b423580aa58b052d96b8b : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0eeb99b4de423d89446b4426747f13 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53dc76c7b05043338dd9a9bc123ac28e : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1afabab802a41cf88a1c816ff481d34 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de0d8942bad4614b64f7819f53eb7dd : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11d9493911b4fa980ef84c24242dcd2 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":8},{"id":"61a5802741f5214c3486114989cd37ea6a4f8e77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-24 08:43:41.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"23d749149ec36431115908f86e7eb210b372d060","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 09:30:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/982b177ce0f44b99980c972a4c0a1a36\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/dff50c72462b4f66845e9876e40d4412 : SUCCESS in 35m 48s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/7e3206c175c44ca4ace662567db6dc99 : FAILURE in 34m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/47347d6b303c4fdc8a0fddcbb5e70cb9 : SUCCESS in 32m 47s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/87e668b1323c4840bdecdc2870ab53b6 : FAILURE in 41m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0270ffccd81e40839be23a7ea7896cb3 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba63103194a14e9e8a2062441d52b7ee : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/503b4946b2414fdd9783922cc2f563a7 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ab49b2b16d999f801de94c6c72304dac24b34c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 09:49:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/109f520cde16470e8fc67432371a1de4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/26bb3316ef1b402ea164e878afafe1fb : SUCCESS in 22m 36s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/ba364ee5a32349a6b2fc760c6236e7bb : FAILURE in 15m 03s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/93848adfe9e74e26b56583cd20a7cdaf : FAILURE in 31m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e8c38e05a80e4f0ba94acf4d4a304d71 : SUCCESS in 19m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2c113b49798749e5b203502e8433820f : SUCCESS in 1h 03m 53s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b06a95a64e8341d1a8ec5804bfa1b5fd : FAILURE in 10m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/074b242040d24d16842cd737eb69e23b : FAILURE in 22m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/da02bfd2feab4b7dafb2aaaceb576b75 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91490a44508544d4b738cdca41fa91e1 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a19f3260ca9f4c8e9f788a965994a578 : SUCCESS in 1m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3248eae833cb46e4a59cf3e1ef9a4f0d : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e6f8a12faa24e37907e86aadec755b6 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f47e0cf553584719960c039812303440 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":9},{"id":"8d93bf3e21a775a1a070a4d959ca0b6cd8690fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-24 10:22:30.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"f4a8f3be19300364e4163d0450b647ca03db437b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 11:07:52.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f818d8c41ee4ab8b1414b09f77abf8f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/58afb58d24db4c1980c5e3bdd0eeddee : SUCCESS in 29m 54s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/2b7d360cf01043679fb3a784df15f855 : SUCCESS in 37m 49s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/18d9a21d8aee49b192c900c90ae4bbae : SUCCESS in 31m 11s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/36539950e8574c2fbb2269ae0500b43e : SUCCESS in 42m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3168fdc701ef481a822d5eb25992ad4f : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3129c6d548b4ef4848c1cccf6c0e58e : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f2930d75e6f4afdb7535d74dc6e9840 : SUCCESS in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7e89adf792b6de63535a403fb6170fe1b08557d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 11:12:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db149013057405fac2b9da9b1597df8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8325843001984299b28efe50be5c46e7 : SUCCESS in 19m 14s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c2bf148d8c5d4cfe99b3e0ec6e60ebee : SUCCESS in 9m 45s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/baaa576640cf4e78ba364b14c251e265 : SUCCESS in 31m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/26a8194bb8614eca91aa9c4c33f622af : SUCCESS in 24m 24s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/23d6ea68114f40f9a56bf2fefe229c4a : SUCCESS in 49m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efa01a1f712f4dce89e5b4252f6a80de : SUCCESS in 13m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b49f90f183224d078507076a28583fbc : SUCCESS in 15m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a09fd77d2c204c0d98ae85a4a0b0749f : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87013775c583455ea4cf164f528a2672 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8d0e1c051984b48bb0a38352f550335 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8c4e65aa7f34e41ace827bd358433b9 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d06e494436bf4da58a7246851c348642 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5d624e84014492b6f427a095d89d0b : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":10},{"id":"93207c7f990487a3f4094b22d32959b2bc5a14bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-24 12:15:56.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"9337cebc7ee961dfda053dba22086d188ad5b8f8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-06 16:03:35.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"340354381ca7cd7dc97c7099392d198a3f55a153","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-06 16:27:31.000000000","message":"Patch Set 10: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2d28b5c11536be41788e2fefa0aee796e1a6dfdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 16:27:42.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"be7de890b5d4d31120af8ed81401373376648eac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 17:42:21.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36ba5f7551dc4fe2b30e34b52b15caa7\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/7655e0883c784028b752c8ab7e6fa783 : SUCCESS in 19m 58s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/787c4fd74a2e4d52846b30e44fe3e7e2 : SUCCESS in 27m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/668e64d59ef24a8faaf9b198fbcad402 : SUCCESS in 22m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8c45ad6ba6194b8eb610b4d27391d4d1 : SUCCESS in 1h 03m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdc53dc0221c4d829ce017a887a79317 : SUCCESS in 22m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5a668a1d5bc4308b755cfb8efd3c07c : SUCCESS in 32m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed1e7a64216d4073a1d7a6293615a776 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15904f39b104499e9282dea84dca3346 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51c4b36024134d4684f293830addc06b : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c825b1b4daa34c938e2d7302546ac13b : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd04116250c4c07be8229155e10ee05 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":10},{"id":"3b85c6d590bff1f360ca1589711cfba509ea9d96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 17:42:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"d9f19249241f21b02e9652a9f4bda8e312d474fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 17:43:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959d81ca73664318a4d430d3ef2aa226\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef7ec17c4d343aa874970e3ceb445c9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8f827e021bd1c3d070763a96bd19cc9d2516f91f","revisions":{"2a0a7150d526f983d926da2fcbef2f9e09323c2d":{"kind":"REWORK","_number":1,"created":"2025-07-11 09:57:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/1"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:57:11.000000000","tz":120},"subject":"Add apache image for OIDC/SAML frontend in uWSGI case","message":"Add apache image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a0a7150d526f983d926da2fcbef2f9e09323c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a0a7150d526f983d926da2fcbef2f9e09323c2d"}]},"branch":"refs/heads/master"},"b7be47e71b09464739c5a86d3aecc1f18440b4fa":{"kind":"REWORK","_number":2,"created":"2025-07-11 10:00:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/2"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:59:54.000000000","tz":120},"subject":"Add apache image for OIDC/SAML frontend in uWSGI case","message":"Add apache image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7be47e71b09464739c5a86d3aecc1f18440b4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7be47e71b09464739c5a86d3aecc1f18440b4fa"}]},"branch":"refs/heads/master"},"e15c4229517d00261e6578aebb208d98f3bbe89d":{"kind":"REWORK","_number":3,"created":"2025-09-19 08:40:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/3"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-19 08:40:54.000000000","tz":0},"subject":"Add apache image for OIDC/SAML frontend in uWSGI case","message":"Add apache image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e15c4229517d00261e6578aebb208d98f3bbe89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e15c4229517d00261e6578aebb208d98f3bbe89d"}]},"branch":"refs/heads/master"},"86ae65fa0785ba1a8680ef66b99e0ca3ddc3d745":{"kind":"REWORK","_number":4,"created":"2025-09-24 14:03:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/4"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-24 14:03:12.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86ae65fa0785ba1a8680ef66b99e0ca3ddc3d745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86ae65fa0785ba1a8680ef66b99e0ca3ddc3d745"}]},"branch":"refs/heads/master"},"52309174c5fa55f99d6c69210793ff96def00b3d":{"kind":"REWORK","_number":5,"created":"2025-09-24 14:29:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/5"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-24 14:29:37.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52309174c5fa55f99d6c69210793ff96def00b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52309174c5fa55f99d6c69210793ff96def00b3d"}]},"branch":"refs/heads/master"},"0d1538ab02fe36435f4e0c3ec43a0261bceef0be":{"kind":"REWORK","_number":6,"created":"2025-10-22 14:20:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/6"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-22 14:20:39.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d1538ab02fe36435f4e0c3ec43a0261bceef0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d1538ab02fe36435f4e0c3ec43a0261bceef0be"}]},"branch":"refs/heads/master"},"943be3555d59e8331f2f310a3feedd8eeeb743f0":{"kind":"REWORK","_number":7,"created":"2025-10-23 14:17:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/7"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-23 14:17:29.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/943be3555d59e8331f2f310a3feedd8eeeb743f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/943be3555d59e8331f2f310a3feedd8eeeb743f0"}]},"branch":"refs/heads/master"},"fb31e4e384f6e7a35b8e4cf86e3e2609ca5a1bee":{"kind":"REWORK","_number":8,"created":"2025-10-24 06:33:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/8"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 06:33:56.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb31e4e384f6e7a35b8e4cf86e3e2609ca5a1bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb31e4e384f6e7a35b8e4cf86e3e2609ca5a1bee"}]},"branch":"refs/heads/master"},"fcb0a005ff7f17f0601d84b9a50b875229115340":{"kind":"REWORK","_number":9,"created":"2025-10-24 08:43:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/9"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 08:43:41.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcb0a005ff7f17f0601d84b9a50b875229115340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcb0a005ff7f17f0601d84b9a50b875229115340"}]},"branch":"refs/heads/master"},"8f827e021bd1c3d070763a96bd19cc9d2516f91f":{"kind":"REWORK","_number":10,"created":"2025-10-24 10:22:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/954707/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/954707/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/954707/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/954707/10"}}},"commit":{"parents":[{"commit":"7aef10849b9cf3d1137a18ad7424a6292f802b7d","subject":"Merge \"ovn: Fix log permissions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aef10849b9cf3d1137a18ad7424a6292f802b7d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-11 09:56:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-24 10:22:30.000000000","tz":0},"subject":"Add httpd image for OIDC/SAML frontend in uWSGI case","message":"Add httpd image for OIDC/SAML frontend in uWSGI case\n\nChange-Id: Ia85f253ecaeaef63c17e75dc4a223e030dbbc58d\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f827e021bd1c3d070763a96bd19cc9d2516f91f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f827e021bd1c3d070763a96bd19cc9d2516f91f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
