)]}'
{"id":"openstack%2Fkolla~954898","triplet_id":"openstack%2Fkolla~stable%2F2024.2~Icb0e47cf95a67cb4585c71f863e542bc59ff5009","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-14 11:01:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb0e47cf95a67cb4585c71f863e542bc59ff5009","subject":"ovn: Fix log permissions","status":"MERGED","created":"2025-07-14 06:47:03.000000000","updated":"2025-07-14 11:02:31.000000000","submitted":"2025-07-14 11:01:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954898","cherry_pick_of_change":954897,"cherry_pick_of_patch_set":1,"meta_rev_id":"33559b38a6d51e938a55f6af4bf5b8daa91098f1","_number":954898,"virtual_id_number":954898,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-14 11:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-14 09:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-07-14 09:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-07-14 09:13:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-14 08:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-14 08:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-14 09:13:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"18a2ed82ec14fa91b5e524f1fb72a24e1cdf5410","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 06:47:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.\n\nThe following files contain Git conflicts:\n* docker/ovn/ovn-controller/Dockerfile.j2\n* docker/ovn/ovn-northd/Dockerfile.j2\n* docker/ovn/ovn-sb-db-relay/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"46e7576db00e05846b3095538bee11a14ac59321","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 06:48:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"69cff217a5b70ad1760875e9333b0c2b4fa5cfa0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 06:48:25.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"aed3497ae9a94adc12cc2b7200cbf8ef885c2898","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 08:18:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d50bb94c20d4c40abf684df032714e1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1c3a0463f08f4408b9a992aed9ab4ee5 : FAILURE in 43m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6fb79e0aa2604a619fcc1acaac619f3a : SUCCESS in 40m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e95c0cd0991c4e3e8a0f0cd6d7727271 : FAILURE in 44m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/018ce157605d4e41a58281745439400d : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac1190c9ed8d420ca1af9d4a77767f9c : SUCCESS in 3m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e99ae4901bd94a2b91cc5a4befa2394d : FAILURE in 54m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/07130b12aa1c439bb504e8126ef57d28 : SUCCESS in 49m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8fd364ac0bdd79e796d7df28da5602d3c87b5e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 08:28:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/256627f4bc4b4a39ab6e61806b8d9cdd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c753827f83804f798e4309735a259dee : SUCCESS in 16m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4c9fe98eee0d472aa561c89c6d11293b : SUCCESS in 31m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/46dc2b65bc1b4a87a9447e3040622ed5 : SUCCESS in 22m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/83d7f316e3d44630bca78b59e00cba36 : SUCCESS in 33m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d04eb3ef8c0042a986ff35a953d4fa87 : SUCCESS in 30m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/88989c439e424b83adb1a205315bf279 : SUCCESS in 1h 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d89af6a3e6241889978914495efb9a9 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c865a0bd877e40a48601ecd313e960e1 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09ac7eaad1b64b9e8760f4b223f2feff : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b33da8b6b6b4483e8a08e494197b7251 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a8bc49e8eb742c49f0742b14ac16bda : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcfb12b5d5444f58e31aa8fc9058591 : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ad448feb8d8461fbff4c8936232c9a8 : SUCCESS in 2m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7f1d5cf33ae74950a44e545b1f49b7d9 : SUCCESS in 21m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3dd6e2f802004aa691cde0e0f7a31287 : SUCCESS in 33m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/302c3b73dd7648b58e0ba2910a8cd0b5 : SUCCESS in 1h 06m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ff03299758914ce8b19b6ecb255e34d3 : SUCCESS in 1h 28m 43s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d9b9cfee1399433f8aaccea7816b7fa8 : SUCCESS in 1h 04m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/27c237d6a92c44c6a2007ff877925364 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/807155ec8f9e43389f83f9c07405389c : SUCCESS in 25m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6e520f494e6a482f86c6c241c57d92e4 : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8045981bb91347ad91c22d6f024d7987 : SUCCESS in 1h 10m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/13cb7519ab2c4aca9d9c9996026f9582 : SUCCESS in 1h 16m 54s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7b19b3646fa9403ca9cd7d560adfaa65 : FAILURE in 57m 11s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c485510a29aa446694d22555d3d755c5 : SUCCESS in 1h 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46ee66822406d957f95ed44338dc1604ff739e46","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 09:12:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"567c968b4285c6b7c5e7709e43d02d5f738cb5c8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-14 09:13:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d2054a699df1005602abc58bf8bed693a76b1ee9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-14 09:13:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03cfcf160e68c47f25e3764addf18338cc85ff80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 09:14:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"303230e4c397a7cde21e3bc38e27558b8064ab81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:01:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adbfa290c0624747815c6e5a645a6121\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fe6dc3aa390f44c9a4374350273223a5 : SUCCESS in 15m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7fd69e4cc0504bc5a661085131967a8f : SUCCESS in 13m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d5335b44373845bfa1ea04eac6cc3d2a : SUCCESS in 1h 08m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/42b00ffc954e46338d8a486da0567627 : SUCCESS in 1h 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66ec8e9367840c58cf5a29272315b9a : SUCCESS in 1m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a99a59c03ff44ffac7089a96917b0b6 : SUCCESS in 1m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90b6553def4f4cf4adad013a84229614 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d94e158a3246e3953b1fbd6332c4f0 : SUCCESS in 4m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b76f9d718754d79a79dc7de08c3e1da : SUCCESS in 1m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fa54ceed0e5c486792266b670fdb7c91 : SUCCESS in 18m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c94b494a644e466a98417a6028aaa8d3 : SUCCESS in 21m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4f8b7cbd65664756add79f6e165b385b : SUCCESS in 31m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/67b7719ae7484f758ee201830040becb : SUCCESS in 1h 24m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/124c4109cb714713bc8c95b2a82b214c : SUCCESS in 15m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/865701540fee42beb05ad6a1beda3984 : SUCCESS in 26m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4b0596af3f3a47298a36bd57a2383106 : SUCCESS in 1h 04m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6e2e8342cbd84116bc8a77b996859acc : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5d9f3ed86436715c6e4afa6dd3d6715f42db0adb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:01:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"33559b38a6d51e938a55f6af4bf5b8daa91098f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:02:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e5388aac3f4f348d30a4d0aa06487e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f47207892ad45d787edfed8fc2f3c3d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8653142051a98e6c7c0609ed1d793937f8e6620c","revisions":{"7e4743e46f0fa37139231827c4f94372d11617ac":{"kind":"REWORK","_number":1,"created":"2025-07-14 06:47:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/954898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/954898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/954898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/954898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/954898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/954898/1"}}},"commit":{"parents":[{"commit":"08637636770ef35982e3577790014a5a492a9aaf","subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08637636770ef35982e3577790014a5a492a9aaf"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-14 06:47:03.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\novn is running as root, and fluentd can\u0027t read it\u0027s log files,\nso we need to precreate these.\n\nCloses-Bug: #2115244\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e4743e46f0fa37139231827c4f94372d11617ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e4743e46f0fa37139231827c4f94372d11617ac"}]},"branch":"refs/heads/stable/2024.2"},"8653142051a98e6c7c0609ed1d793937f8e6620c":{"kind":"REWORK","_number":2,"created":"2025-07-14 06:48:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/954898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/954898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/954898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/954898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/954898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/954898/2"}}},"commit":{"parents":[{"commit":"08637636770ef35982e3577790014a5a492a9aaf","subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08637636770ef35982e3577790014a5a492a9aaf"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-14 06:48:06.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\novn is running as root, and fluentd can\u0027t read it\u0027s log files,\nso we need to precreate these.\n\nCloses-Bug: #2115244\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8653142051a98e6c7c0609ed1d793937f8e6620c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8653142051a98e6c7c0609ed1d793937f8e6620c"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
