)]}'
{"id":"openstack%2Fkolla~954899","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Icb0e47cf95a67cb4585c71f863e542bc59ff5009","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-07-14 09:13:56.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Icb0e47cf95a67cb4585c71f863e542bc59ff5009","subject":"ovn: Fix log permissions","status":"MERGED","created":"2025-07-14 06:48:37.000000000","updated":"2025-07-14 11:02:34.000000000","submitted":"2025-07-14 11:01:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954899","cherry_pick_of_change":954898,"cherry_pick_of_patch_set":2,"meta_rev_id":"e0ddadd5c77098701858a5a8cafe1e13dbfdc677","_number":954899,"virtual_id_number":954899,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-14 11:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-07-14 09:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-07-14 09:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-07-14 09:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-07-14 09:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-14 08:53:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-14 09:13:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"3f673988816dab1baa21c153dbf97fcf2621f131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 06:48:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"8053b91bcc7ae68bdbf7044729d1eca02dc86491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 08:53:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a79da4569f5464eb70e7bb8f2ef3224\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fe5419f255424555a0703cd80bdd07ca : SUCCESS in 21m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5dcdca1e72eb4a818faa1ed1fb73714f : SUCCESS in 1h 11m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb3b8828679740c282f0fbb750af873b : SUCCESS in 21m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ffb2ed74fb984b14a25771fb38acca43 : SUCCESS in 31m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a05b50d0aefe4ed6ada18538cb2337ec : SUCCESS in 1h 04m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47388c6c03f481a88002c70106fa1b4 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21c5cddc47b4459083c1e2e48dfb71ed : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f90c575dd62e47c0b290238dcbc9c593 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c26a4912a68c4fe5b547bbc444cb89c7 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7885e2b0e340412396b93da6b3660bd0 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb717aacf8764e2782c87840d8f3db70 : SUCCESS in 4m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1e0b18aca03248878b98a2053057c171 : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2544bc9e1e264667818cea8000011655 : SUCCESS in 20m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/33b93ecbd05c4ccfb7b8eee84243bfc2 : SUCCESS in 26m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f322bb589f934afa89f662647b3d34f6 : SUCCESS in 1h 06m 59s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e9eca3dae55649b4915bc7aa8a93c008 : SUCCESS in 1h 14m 22s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c4b4e0f810cb4378973c5814d619b73f : SUCCESS in 1h 44m 51s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b4e064f8ebb460b8090be24baaeff7e : SUCCESS in 16m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f28bdd11be0488e8ba48d5d731f3eec : SUCCESS in 20m 29s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/85ab0a8d9e6c4eb9943ae64abcf903e8 : SUCCESS in 56m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b88269cb84d4127819839dee87d8d28 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bb0a5116ec44ef898b253a095702ebd12b397b23","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-14 09:13:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"840219829c199b0166d51b5163a9dd4bda5db4c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-14 09:13:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"765211ed0eedcc97e4bf2b00aae4224333636df8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-14 09:13:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be9ab8f3812c8ecfbc01fa79d593f3bc5d5c8fbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 09:14:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"133b85bbc78621c89025f26be0ded6e41b0a7671","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 09:25:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5f8a92d1bb412e96e36f5e2e263656\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a95aec17ea4a4d18b8a06042ae7e0bf3 : SUCCESS in 38m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1458581375aa4cff968969e4dad4d9a1 : SUCCESS in 1h 39m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e997793afad34dce864a538c75219c2e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0ea22b8f68945f4b52d305081d24bff : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3ff2de31df04df0bfcf8d58feeea526 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"10c50cf5df013276a420c3bf1e6a0cd5b4fc8cba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:01:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c633310ca0b4791b482bc43f6da5ccd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dc04519f68394052817a789d44108a17 : SUCCESS in 15m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/92a786a9115740258fe9f9f639becf4d : SUCCESS in 34m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/53adce86f7464409848d3189ea651666 : SUCCESS in 1h 08m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f155c17e90747f099fd7763d40e06aa : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0816f444435f45f6b23b24fae72b2083 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3dfa3aa49bc45009f95c763e2443fe4 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14ad62420594c79b6d447ab42dfc7b7 : SUCCESS in 4m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f0a6cd66190d4bf9aa19101dbcb8c3fc : SUCCESS in 2m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fa096242bb7c47c097bc25c6bcfa36f3 : SUCCESS in 19m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ba1e456e8ed74e76b1076a1f38152bd6 : SUCCESS in 25m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f443ad870fee4b0f96e1faf6b8a9f449 : SUCCESS in 1h 04m 32s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c6f4a6baf004bb78de11429bce95ad3 : SUCCESS in 18m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/369d7d7350364075953b8ce5782d39a1 : SUCCESS in 25m 15s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/1a3242662c9d46e8a2cb5cb63efa52fa : SUCCESS in 56m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a8015c3ecfb84f74924c966740217c06 : SUCCESS in 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6be3230fe41c026250778501a7ee3135e97f2506","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:01:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e0ddadd5c77098701858a5a8cafe1e13dbfdc677","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 11:02:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95dc43018ba44302b340128d8c79e862\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e47b397d4e4a3c9e903ce3e78368ec : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f9055397aae92e6c056adc8d3352767e07a5ab2","revisions":{"5f9055397aae92e6c056adc8d3352767e07a5ab2":{"kind":"REWORK","_number":1,"created":"2025-07-14 06:48:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/954899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/954899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/954899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/954899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/954899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/954899/1"}}},"commit":{"parents":[{"commit":"92c9782e83f5364907d4093482d350099c7de488","subject":"Enable EPEL repo for Ceph deps in glance and gnocchi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92c9782e83f5364907d4093482d350099c7de488"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-13 10:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-14 06:48:37.000000000","tz":0},"subject":"ovn: Fix log permissions","message":"ovn: Fix log permissions\n\novn is running as root, and fluentd can\u0027t read it\u0027s log files,\nso we need to precreate these.\n\nCloses-Bug: #2115244\n\nChange-Id: Icb0e47cf95a67cb4585c71f863e542bc59ff5009\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f9055397aae92e6c056adc8d3352767e07a5ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f9055397aae92e6c056adc8d3352767e07a5ab2"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
