)]}'
{"id":"openstack%2Fkolla~955174","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I00bcb30662bb12ed8711b28a64ef1941dcce893a","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-07-17 07:17:45.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I00bcb30662bb12ed8711b28a64ef1941dcce893a","subject":"[base] Ensure full control over APT sources for Ubuntu","status":"MERGED","created":"2025-07-16 09:27:12.000000000","updated":"2025-07-17 09:01:29.000000000","submitted":"2025-07-17 09:00:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955174","meta_rev_id":"8b1c74263776784183bb4ea8d8f57efed1bb5447","_number":955174,"virtual_id_number":955174,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-17 09:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-07-16 20:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-07-17 07:17:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-07-17 07:17:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-16 09:27:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-07-16 12:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 20:12:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"221875429e579d8a2b76581a82705ea8eb2d26b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-16 09:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"600b50c585d8cf456a66c4ec34779eb8893303a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 12:02:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92c75725120c46b68aa2f6fbd41fae0a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8be3daab27a34286bbd3dc7c5aa20bac : SUCCESS in 21m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fc0dbfc86a6d4f239b6db14d21d2cd6d : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01aa1abc150b40ee95598e6356e42a35 : SUCCESS in 17m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d8e9d5a3d96349709e238f15b3c5ce37 : SUCCESS in 17m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/97d0167003424bef8d91b2d567fa8593 : SUCCESS in 1h 07m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f7577fad22bb4a9387b3fde8c1ceb1b3 : SUCCESS in 1h 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c4a5f688ef4193865ccf02f07f3584 : SUCCESS in 1m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0b6cfd8d1da4797a75b88754c9b1cef : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6189a0eb026d4c979112f539dff72685 : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90ee4baf2e3b4690965148fbeae61a75 : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6100491aff814ff89ac6f6ef80075495 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf069316ff7c4a0ca7bf708608bf504c : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fffff98e7aa6451e84eceaef4cd2384f : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/17f628da27e14df8b9da076c068e2d8b : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/03e15017d65d4ab5997aa3e225e457d2 : SUCCESS in 18m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d89445660e7c42d5b5d427766043ab45 : SUCCESS in 20m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cedd5f45eb264972b80487e54779a791 : SUCCESS in 1h 08m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b04ceb46e0ce40cf9d927accb46fad79 : SUCCESS in 45m 33s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d9d3d3bc80084c1996e91ae2c04ece85 : SUCCESS in 1h 36m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d29b3b8198704374abe273a6a1433e50 : FAILURE in 1h 17m 21s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a9b4f59ac8904325a4926d3129204e8a : SUCCESS in 1h 08m 06s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c3cc85ebbbf84465b9c5996d3d026333 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/75c09d2e5b7e4496b1fbeb4a255750e5 : SUCCESS in 25m 53s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/18f39d867b9c48269805ea2a8626fea4 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ed8e75084f434786ac3efbfdb1a55a07 : SUCCESS in 1h 00m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cb3c6db4e04a47659f09b6d059684002 : SUCCESS in 1h 20m 11s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fc4742d0dce54c46bff29ad5df9bfe86 : SUCCESS in 25m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7512945cc17e47b9ac515e5ee93557a1 : SUCCESS in 16m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f9d99a3f2aab4e67b3ba9c6a1695987e : SUCCESS in 16m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0f68665c17844b2caa4ccd2070cf2e44 : SUCCESS in 58m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fe3de60bb73b418397ee85015d76976d : SUCCESS in 42m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e7f127a8e1644d7fac30ed1e07a1d0e9 : SUCCESS in 53m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/82582450510e49e597b2b5dd518c130b : SUCCESS in 2h 30m 43s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/02a54d2a7e0948149e0b51389587deea : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/89d62da7319843198d5186319d3eab2d : SUCCESS in 1h 21m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/052b95e7e24349b4ac5277ef7c879c50 : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/535de2d91abf46389aae79a64ef76a0a : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/08819e5d93ca4fd69a6aba47161c492a : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4c171779b53442c5bab68f3eb03c3592 : SUCCESS in 1h 21m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f712e9795e3a4964ad2169c097bead51 : SUCCESS in 30m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a499662947cef2b8139332264cc00ce5ed1c41f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 12:36:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfcc854b3c1c4a2ab404242a66d09f2b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/41bafe444d3740eeb1c5b8b064e0a62e : FAILURE in 1h 29m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d788f8b798a045d89e6f1b473f5deb6b : SUCCESS in 1h 43m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb5db6c040814babb98bf000ac72901f : SUCCESS in 2h 33m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9e0e210d96349f0b9bce1e82a783e18 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0535160d7c17401f835f038b3c9b4273 : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4cb97f7466044ab28ee9b9b493b2af9a : FAILURE in 1h 19m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/deacd6e0d0ae4d1286a8494109372d56 : SUCCESS in 1h 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d34a4159a02d4d4f4e8fc08b101cdd485910a2a6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-16 20:12:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"33deec1cd1a1f948df227f017e5d826fa8008d74","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-17 07:17:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3ba77bc7303e53276b9ce040522e76e60968764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 07:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cc2b7c2579805a45aab86367367e0fe381bc8e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 09:00:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d8a2706cb8c4878beb0737fc7198f5f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/13742b34c9bf4d71aaca743187572240 : SUCCESS in 20m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/13e0b44cbbb841029c8cc7b12ab14a57 : SUCCESS in 25m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/80487fa7389545ef8878a2ad5d7ee8ab : SUCCESS in 29m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6d74c89633a04487a777dd37e32d4f83 : SUCCESS in 1h 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34d14d72a5c452d836256c669ab2b08 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/418cca930a57471a99164943c9dbb144 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b3da4a1bf174c5eae13d757b0de37e9 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718801aa8d494e09a941e12f65c56f13 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06fa94db7c444e09b452ee9c2f1c1add : SUCCESS in 3m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bc1b48233f19446fb0b64366d9e65ab9 : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55d3138b4a2f44c182e060c1f999466e : SUCCESS in 21m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/68f1d36be7c644cd84a02137923b102e : SUCCESS in 28m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/51fdf1e83aa242d080a1340ac50fbbbb : SUCCESS in 1h 04m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5d2c65b455a74325aa6bee13cbab5a29 : SUCCESS in 1h 23m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/afb3ca6c4e92407c92dfabf1c47f65a1 : SUCCESS in 13m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/87aa6cc621dc4a11b6e129bbd77105d4 : SUCCESS in 11m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/510486d3453149e183feb3b77f2a6a5c : SUCCESS in 1h 06m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6834e00306784450a174fd78a8226001 : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":1},{"id":"362ec43daa05e414efa631b24a742cdb3eb76f4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 09:00:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b1c74263776784183bb4ea8d8f57efed1bb5447","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 09:01:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1e3f639adec462c8e7a28a200d78c98\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dae637512e44e1cb00a5bf0077dd522 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87f6356dd0d54a0282b7179712d1261f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cae3cd6192140acf70bc44b6a37b136bec078916","revisions":{"cae3cd6192140acf70bc44b6a37b136bec078916":{"kind":"REWORK","_number":1,"created":"2025-07-16 09:27:12.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/74/955174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/955174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/955174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/955174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/955174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/955174/1"}}},"commit":{"parents":[{"commit":"2abe7b1d6ef511c72be7d89c5277c10364bdb11d","subject":"Fix glance-api: add missing lsscsi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2abe7b1d6ef511c72be7d89c5277c10364bdb11d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 17:53:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-16 09:27:02.000000000","tz":120},"subject":"[base] Ensure full control over APT sources for Ubuntu","message":"[base] Ensure full control over APT sources for Ubuntu\n\nUbuntu images were occasionally failing to build due to\nunavailability or slowness of the official Ubuntu security\nrepositories. Although Kolla uses `mirror://` URLs to\nleverage Ubuntu mirror infrastructure, the original\n`/etc/apt/sources.list.d/ubuntu.sources` file was still\npresent, causing the system to fallback to default sources.\n\nThis commit aligns Ubuntu behavior with Debian by\nexplicitly removing the distro `.sources` file and relying\nsolely on the sources configured by Kolla. This ensures\nconsistent and reproducible builds both in CI and\non local environments.\n\nCloses-Bug: #2112140\n\nChange-Id: I00bcb30662bb12ed8711b28a64ef1941dcce893a\n(cherry picked from commit b5849adc45e3a39e0837dea3ee8e1fb3cd057dd8)\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cae3cd6192140acf70bc44b6a37b136bec078916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cae3cd6192140acf70bc44b6a37b136bec078916"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
