)]}'
{"id":"openstack%2Fkolla~955347","triplet_id":"openstack%2Fkolla~stable%2F2024.2~Ifd664de82a81830eeda8ccc019b22ee93768635e","project":"openstack/kolla","branch":"stable/2024.2","topic":"mariadb-fix-stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-10 10:10:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd664de82a81830eeda8ccc019b22ee93768635e","subject":"debian: Use bookworkm suite for rabbitmq APT repo","status":"MERGED","created":"2025-07-18 08:51:22.000000000","updated":"2025-09-10 10:11:30.000000000","submitted":"2025-09-10 10:10:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955347-mariadb-fix-stable/2024.2","cherry_pick_of_change":950531,"cherry_pick_of_patch_set":10,"meta_rev_id":"4d0c979d3b2c6aae7cdbf8c61541df7c2084eb31","_number":955347,"virtual_id_number":955347,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 10:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-09-09 18:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-09-09 16:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-09-09 18:28:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-18 08:51:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-07-18 08:53:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-18 10:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-10 06:56:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"215dedc1d7fbac4a00dcef4f8841ea0ed521dd7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-18 08:51:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"20d6e304d9fc9ad6204568c07857d334370b7143","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-18 08:53:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"13722d354604719559d80330ae80f06efb64c44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 10:11:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd9302d2306047a38ace99dc2ad0a8c5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/087afdffac0c4c0ca2187768d7f3f093 : SUCCESS in 18m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f6264a0806f24545a9db57286c24cbdd : SUCCESS in 1h 10m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f18494f408884d2391c0e3779d7e86a9 : SUCCESS in 17m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/26adceabc6a94045bfc80f4f57c8d52e : SUCCESS in 27m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/38ba1c315a3941009acd8cd4d773a79b : SUCCESS in 52m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4dc6b5cf1d0c4cfe8394e4097ff98a08 : SUCCESS in 1h 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fdc083c87c24db6a4e004cdb31e37fc : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1badc271b9a4426a8baf032f85ab02d6 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8c09a2f4776428ca5ae62241bb114c3 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c410484ebd3949948e09018c69a3fe1f : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d614cb0915d249a6b875501890724b1f : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c933d76c1e25425eb7c9db7c821ec9a8 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47012495504645e09e25f230f1603a59 : SUCCESS in 3m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8c80cf6ffbb747f8b5e2f393018946b7 : SUCCESS in 2m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/652d5a3be7b14fc2a372290fba27c9b7 : SUCCESS in 17m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a4647a8361a2426d875d57f0536359dd : SUCCESS in 20m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/478621e33fd44e388b43bcc563352218 : SUCCESS in 59m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b931fd7d45484494bfa0ef21dc26a6f5 : SUCCESS in 1h 15m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a7128515b8b04210a9365455c882e35b : FAILURE in 23m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d5927ab55e9f42978fb2f1238e38c110 : FAILURE in 26m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9193555f834f4edd8b3cb5f60d9ca643 : FAILURE in 15m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/912ac64fc6dd4845805b8d667ca661e2 : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fc74d048c84b306346216c54224c0894d395b17f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 11:13:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07619951693d47e89b046888bcb16e76\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/528636b498ed44ca98f260e44b97f0a7 : FAILURE in 54m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/56d3be0aeba04bc3b51c423225469e13 : SUCCESS in 58m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/82ca1a4c5016488d8950be5632ed9797 : SUCCESS in 1h 58m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89bb7df3eb534f0e9f50e3bb7f5eccd8 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f186388a0a44b0db72582e03a408b0d : SUCCESS in 5m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3e6277e706ee49959f03a58c09e380fe : FAILURE in 47m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3290b4b0cf1a4965975b8ddf24a3097b : FAILURE in 55m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0e0199ea80d05d83b7a8f59d68b6f290a518dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 12:58:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c6b0d5c8d380709f3dafcd58e7fdf1578bc14a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:11:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ecb049bfd7042198363d33dd9cf5f9e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f8d5b7745ca643eb967d4bb3052585b7 : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5cbc169a28be4962abea7e4e2178766b : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/906cb74b556e40e782bc70d146da2290 : SUCCESS in 18m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b7f88072f7ca42e682110508b7223f9a : SUCCESS in 30m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/48fd29d8502e4611bafabf93144d2679 : SUCCESS in 50m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/031ef5c86f98439aa97fc83fa1064257 : SUCCESS in 1h 34m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513af6a18dab4dbf970f1bc27aa80d7a : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfdf86ff9c0143249e81f33f1fceca48 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08c414370a224d4cb037f541ed85f106 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dab4d144714342d2b3f8a994887c8ee1 : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c424eed00e17466c9c07dc11b46c4dad : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725c725561d94710b7ab62bdbcdda9ea : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1efb54def0e478ba1fa62917663ebe3 : SUCCESS in 5m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a6d876fabd8f485695ef031b70115807 : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d18ee87bfdb34700bc1bed847b562130 : SUCCESS in 13m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd4126d4f0654241ad8e8c359f9016c7 : SUCCESS in 17m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8cb2cc9a6e0a448e9b2270d332e55dfb : SUCCESS in 1h 09m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0a293e819be4404cb8efee2f806605d8 : SUCCESS in 1h 16m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bce596fa99d749e3a6c64402efdd4c29 : SUCCESS in 21m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c90948ada8334accb3d50a6a760bf0c8 : SUCCESS in 26m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/73f1bb5667bd4449a6e18931eed192d9 : SUCCESS in 36m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1fb765e8ead4deebcc7fe51697f2c5a : FAILURE in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b7fffea33d4b8f11b55ff5ce4bcd4cbfc2aae9b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:00:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd072c0267d42a7bfb378ca24d6da13\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6e6bfdfacf9e40fa85ba963620612c35 : SUCCESS in 54m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac380632b4d14488a3eb18090cbae40c : SUCCESS in 51m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d01dbb5411114e3eaa96802bb74f3bc4 : SUCCESS in 1h 56m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/312135b7111841d2979832415b8091b6 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01fc5d21324e4572827e2c57967967a1 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5730e27e5ff04f00a43b00fc3c0bb166 : SUCCESS in 50m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f19f0cc40ae5417f95eefc389047856f : SUCCESS in 49m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e52235372925df06c9dead0d5487c5695755e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:56:08.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"36cfe017aad23e732296259fe06c13a149949481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 18:26:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8727abb9f584b9e8fec3ec3726b2895\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b68cae7052824e7cb4ea69a85902ffb0 : SUCCESS in 23m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b269b8064e4a4d82936e42eec7ab03d7 : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/de823e0d87f34c5caba72e9e85843197 : SUCCESS in 9m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ce8afc25b46c434c9ae364813309d6b7 : SUCCESS in 26m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2ac3dae437cf4ddfbe7fea55952a7656 : SUCCESS in 1h 14m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c59d443de8084c1ba7a409ff26657ed8 : SUCCESS in 1h 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cab9b3da724cfdb1766d07f21ea725 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3a726a95a7342ed96617e76e92ee571 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0fef145b10a40ee879d541513ec2ad1 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e25457e69ba4856b0995269d1a4fdf2 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0dc635806f044d2b0a1a8fa7ba2588b : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d901f6b51f4c7db230bd2754f88c85 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e34e34c2ece246f08c45312b6d4dd0ae : SUCCESS in 3m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b6f54fdbf3be4972b7ec60c32cd96f12 : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/46281afbb122486dac9df1956a44ee7e : SUCCESS in 24m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d85856d52b8f4ec6bd48f8b83214a866 : SUCCESS in 19m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1e17823608d24bf49dc829f070adba1d : SUCCESS in 1h 07m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1a5cdd3a08644f31b512fd5cde8a12ad : SUCCESS in 1h 18m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/078750f2ec5e49debfc99fb7d96f42e8 : SUCCESS in 21m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1e9df0b5a0c24f7b930d9c953f41cd18 : SUCCESS in 24m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/366724042f1e492b9dc1cf362829e864 : SUCCESS in 1h 04m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c9f7947f4f2a44cca87fa78609dd17eb : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":3},{"id":"cf4f92554e64b3d08da8acb460819617ecdfc107","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 18:28:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e8fc696eb398c847dbf2401f9fddf3fbd0ff154d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 18:28:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e69addc4a57151e2fd0e9d90a5bb519cf374e0f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 20:12:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/632873dc055f4d26af90018cf5ae3621\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb013c1a51e440a08094adaee00d9d10 : SUCCESS in 10m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c0b9d6765a974522a2e8a92e9eb3f77d : SUCCESS in 27m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1d3690bd64dc41beae6619200525d70f : SUCCESS in 53m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b1d4977ea660458ab6c1f276624144a9 : SUCCESS in 1h 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69721f0ea3fc49adb5496ac87da0f593 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e28440d3bec4bc7b0eafb9ba5f028a0 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c9b7101da30466d99609fad380587f9 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a983d29a824e318e80399d85817a73 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13739ba90577446aac91fb64e5edae05 : SUCCESS in 3m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a5192e15cdaa47b7b71f5a4fab80367a : SUCCESS in 2m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c1a81d0463cb4d44a7a1e75647367c51 : SUCCESS in 13m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a66203e3ef454645b7916f4a3644f27b : SUCCESS in 15m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/07b6711abb2041458ee0f068565c4da0 : SUCCESS in 1h 06m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9356676ebbea4c50b5a5653f1172b0f4 : FAILURE in 7m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e560c294654a46bb9dfe198f56882dfe : SUCCESS in 11m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/496aed3191e647f4976f250304d22885 : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4ccd64e63be945c69de2b258fbeeaa61 : SUCCESS in 55m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6f2d4b5875934d56becbd06e9699a573 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a346241ce2e88e209dbedef9274d047380efda52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 20:31:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c572e0fe1b4a6dbd6c771cee93945f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/de0fbd473a904e63a89bfe25305af759 : SUCCESS in 25m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c53e6b5479e0419cac7b2a5621d02c16 : SUCCESS in 52m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/997c034eae5e45eda8c41fca52019d77 : FAILURE in 2h 21m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53817595d02b48dc9b446fe34f294b34 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/997f406b93644ca2a4fbc5b1fbe116d7 : SUCCESS in 5m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9c28860c98cb448eb99d1decd024d2c3 : SUCCESS in 27m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9d2ae976a10c4a01ac7587b9f0c308a3 : SUCCESS in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49451893a68eeefbf4bebfc02e6e5b47391ffe37","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-09-10 06:56:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"479af61f82086c8ad385905a86fa3c1b26e9e410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:02:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26cce4a8080a4851b72798a74ddde905\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1ad176e000b24308a1f8cd86b0adb9b9 : SUCCESS in 23m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d4f8ba9c0624bdab57f13195cc1780b : SUCCESS in 32m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f0ce37a3d344de39684347807000139 : SUCCESS in 1h 02m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be32812611fd486b863e1f3b2671d356 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/33f15e3f83c54bb1b6492609d397b920 : SUCCESS in 4m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d9ae22ad56d24e719db0a4e30888a2cd : SUCCESS in 35m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9f34182efdb643779dad1b17d225b81c : SUCCESS in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e877ae941b6244a1e33f965572e195c9baa9274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:27:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90679b3773e41298af8af9b36a758cf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cba00cb3f6524c3e812a4c3d040b43f4 : SUCCESS in 12m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9afd438ffb114830bbc966bb46814cbe : SUCCESS in 49m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d5d3cf691c04f3cb5466c3d1a319d01 : SUCCESS in 9m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2f50c70a2ea147c5981e53b531b94e00 : SUCCESS in 21m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5ac0f2dc015c4e6e85f25f54089060df : SUCCESS in 1h 01m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/19c226526c884c71bd171069447140a8 : SUCCESS in 1h 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866f41139ea34c7aa9b6beebc03e70c3 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f622a30fc140404a9bd3cde7255f5906 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63507e816ab24bbfa5cb2db1675f75c9 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac324999f8894b53a063f52c10e13abc : SUCCESS in 1m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0db5b8c3a92440c780de6d8cb276659c : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0270e75a6d456fa19beb1cb7b6d937 : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16468f64cf5e402abdf3b13974b9b00e : SUCCESS in 2m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/38430b2bf06a4fc6ac2fc3631d56c08f : SUCCESS in 2m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/65842ce98a664c7d8eef4047e683ae6e : SUCCESS in 11m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c58185272756499cb03d67c1ae02a3b3 : SUCCESS in 29m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6279ac01c1d64cb99fceccad71ae4330 : SUCCESS in 1h 02m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/93c3878c2da84d6aafc24a19e6ad20e6 : SUCCESS in 50m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3441008192264e9d9233f2a538bb0e90 : SUCCESS in 18m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/19e7347721bd4d10a98be47723125286 : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/457eabd555e64b1ab45f9f3cb2c91ab3 : SUCCESS in 1h 21m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e14a5e1a56454c87b20ad38a581deb8b : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":3},{"id":"9b604ae6da4db009a014b327b97a7e9db9738870","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:27:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3774b9339df08daa98d0feeaffb6dc7803acb0aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:10:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f5dad05e1a24e30b4b5be1e93298264\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/618def50f671450998bd8701ead02e31 : SUCCESS in 10m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/61cf377d492c45f0a40a4da520cca508 : SUCCESS in 27m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cfeb674cd4ce47b493bd95e171d3c67c : SUCCESS in 1h 08m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/de8b0e66f2b040a7b1abeda89f27370f : SUCCESS in 1h 37m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc07926f2284c8796db2148c2fdfb3e : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f1fe0cd2c54616a3475755dc811dc3 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f7b73101eb4443c8bda4e612cef0fd4 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ebeb07e24d470eb17e21d966775eba : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c65329d627e54281ba4e87bc543eb4d1 : SUCCESS in 3m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c9b8de5567f34e8d954e6f3cfd7fc841 : SUCCESS in 2m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a5029f291adf498fb9bc6a379f59b71e : SUCCESS in 17m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9a995c3ad7314b6ab99584c400286027 : SUCCESS in 24m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d5ed9853668746d7aaf737efd6a44b35 : SUCCESS in 54m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/75ea40b8b4534cb5aabe845f4b4d2481 : SUCCESS in 45m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/22e615173fe34c4ab7a6872fc0cbbe85 : SUCCESS in 21m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c75256bd5c8f42a1b324f83d06f76f6d : SUCCESS in 21m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/85355ae8d35646d1a2cadcf6bbf34662 : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aa8f37e05e994da281108c07d216f45a : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":3},{"id":"2813eb274027ed4cc9ac0c430216d282ca712654","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:10:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4d0c979d3b2c6aae7cdbf8c61541df7c2084eb31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:11:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693f3ca1b6914828adb62b115f06a558\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16640b773366420587128aaefc45f645 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9665f3a55c3e4824a9cd3d85a823425f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e89e069ba74ba297d9f72ac69129008f226ca7ca","revisions":{"d63ba4ac9f1c7fe714c62cac84f3b31f1e820811":{"kind":"REWORK","_number":1,"created":"2025-07-18 08:51:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/955347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/955347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/955347/1"}}},"commit":{"parents":[{"commit":"cae3cd6192140acf70bc44b6a37b136bec078916","subject":"[base] Ensure full control over APT sources for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cae3cd6192140acf70bc44b6a37b136bec078916"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-18 08:51:22.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d63ba4ac9f1c7fe714c62cac84f3b31f1e820811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d63ba4ac9f1c7fe714c62cac84f3b31f1e820811"}]},"branch":"refs/heads/stable/2024.2"},"914cb586a58416cbcec834dacb90fc8c91c74b23":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-18 12:58:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/955347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/955347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/955347/2"}}},"commit":{"parents":[{"commit":"ab3234bc2afdd16f82caae976eb2f904c3376d60","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab3234bc2afdd16f82caae976eb2f904c3376d60"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 12:58:38.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/914cb586a58416cbcec834dacb90fc8c91c74b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/914cb586a58416cbcec834dacb90fc8c91c74b23"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"},"e89e069ba74ba297d9f72ac69129008f226ca7ca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-09 16:56:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/955347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/955347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/955347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/955347/3"}}},"commit":{"parents":[{"commit":"c91108a1519a0f4c6f82d9da8df765fcc52cd689","subject":"[stable] Use versioned Erlang 26 and 27 repo on aarch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c91108a1519a0f4c6f82d9da8df765fcc52cd689"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-21 11:56:46.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-09 16:56:08.000000000","tz":0},"subject":"debian: Use bookworkm suite for rabbitmq APT repo","message":"debian: Use bookworkm suite for rabbitmq APT repo\n\nChange-Id: Ifd664de82a81830eeda8ccc019b22ee93768635e\n(cherry picked from commit c930c9fb9c435cf0681a15a84814ff91eb055b9f)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e89e069ba74ba297d9f72ac69129008f226ca7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e89e069ba74ba297d9f72ac69129008f226ca7ca"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
