)]}'
{"id":"openstack%2Fkolla~955680","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I3a0aa82d0903b341156c2fce68393cc4031e7728","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-07-28 14:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-07-28 16:38:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a0aa82d0903b341156c2fce68393cc4031e7728","subject":"change erlang arm64 repo to rhel9 based","status":"MERGED","created":"2025-07-23 10:37:59.000000000","updated":"2025-07-28 16:39:21.000000000","submitted":"2025-07-28 16:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955680","cherry_pick_of_change":955659,"cherry_pick_of_patch_set":1,"meta_rev_id":"aec3732d5095b038f674f8711ddac5ebff45bcab","_number":955680,"virtual_id_number":955680,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-28 16:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-07-28 14:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-07-28 14:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-07-28 14:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 14:37:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-07-23 15:09:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"291d3a882dc9d6447e53036feda3e5726ca2db4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-23 10:37:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02964f40c79b73f935bcb774f9d8611b259c3017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-23 10:47:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cb0b5d89d22247c3c5c1a62aa4022fcd1da4de01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-23 12:41:20.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4936597949c76ae122bd64134de9e28d1e54a17b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-23 12:54:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"09d58c379e93c3def331516e6f81f45db42c2b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 15:09:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56366cd9600f4d97ac4a1a74ec603457\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f49789c9d31a4e489bf39328cbd671ca : SUCCESS in 17m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/64832eb6fa5f4c4baff4f85ebb9ed839 : SUCCESS in 36m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1135f38984e744a681d912e2f9288273 : SUCCESS in 20m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5a043f8da1b14fa08c20f8a099ee4202 : SUCCESS in 18m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3edb076facfa43948ad8b3e55839992c : SUCCESS in 1h 06m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1751e45205bb448d892513a6e3312f78 : SUCCESS in 51m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69dd303d091c4de1995f320864bebd4c : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c90b5a5d563d441897dfebf389f487fe : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d13d77acac9d47589d6f70164e64c160 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f1bdde0616483caf847bdead8c471f : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9360228dab6f403fb13421ecb79e10da : SUCCESS in 7m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b44665718bd441d0b70881cbdde355df : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/772ba6db2c2b4b92beececb8b4325027 : SUCCESS in 24m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0eb5da77441943cdab8210f2bee407e5 : SUCCESS in 1h 20m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2a7b59256d75408db9ded803f9b8936c : SUCCESS in 1h 06m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da2e7f42ddce4e479c4ae156092d4891 : SUCCESS in 1h 36m 40s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/cc93ee227eac47159d531abf643df19e : SUCCESS in 2h 07m 39s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b03fa5f520774bd294fad60435a95fa9 : FAILURE in 34m 36s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8560ef731c7a4a8b85738c976cf7a4ab : FAILURE in 39m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2be3c0062b324793a96d648d628078ff : SUCCESS in 17m 48s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9184ca4b800b421e963baffcd8a378e7 : SUCCESS in 1h 00m 54s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/43478bd2af91408b97d1c95045fd6ee1 : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/069162508ec1408fbbbbc0dcfe3a2002 : SUCCESS in 1h 42m 11s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/120353345232413eb283d84844573a38 : FAILURE in 26m 48s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7594f86ff66c4f139d2e81cc037ab79d : SUCCESS in 18m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9d4fd6a7d2c840d3af5839a407ca7d0e : SUCCESS in 24m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18c1abe1ff844a648578b6917962e0fc : SUCCESS in 1h 05m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/11a952ae83c7430e81ec571c67bc6189 : SUCCESS in 1h 21m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cd351d980d82480288a2332488cba4f1 : SUCCESS in 1h 53m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/82dfe79afdbe4a9c89d92c595a75f1ac : FAILURE in 38m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ba8d32bc6325412d92b5989ae27a39de : FAILURE in 1h 04m 24s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/16a82f0c7e424844a515cfd5cc16fa6f : SUCCESS in 24m 13s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/daf393a4f32a4ebbbb4458a5069b101f : SUCCESS in 57m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/42058a4ecd134b019789d30266698c99 : FAILURE in 50m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f1144f9940d5415ea76d9b1e1f099cd8 : SUCCESS in 1h 34m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea53627e20584aa78a9ef8ada1c5d074 : FAILURE in 23m 51s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/450c3c12d2be43ea969fe585bb29c92d : SUCCESS in 1h 59m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"67a043d5930dafa50570eee14283a2c155c9fe51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 15:09:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd3c75126604338af80713c77264cbe\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/016ff310ff8445f3a5f9c46e256cb480 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f7fdccfc7a414044ba3bd5b1f2454939 : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a3419cbfe4a14603a457bdaf7a387ade : SUCCESS in 1h 22m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0b9fab6117043078a14a146199395a1 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49ced0eca9354a9c9576722754863bd7 : SUCCESS in 4m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/16ea3bd5092148cf94f883000de6960a : SUCCESS in 53m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9600327e2a744ade833ee46ecfe74722 : SUCCESS in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a5257aa0438de41bfb1b5a8d211f44422632500","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-23 16:00:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"85579a44d290657aa67ca7549f24423cffe53678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-28 12:17:07.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"abcc0624a5be03947e6869117d587b3e96ba84d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:16:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf04e809270843e9b365fb8915f6f657\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f92c3fedb7b9481f8ed310f31e37750e : SUCCESS in 58m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be863c24588646ed972492eb914c544c : SUCCESS in 48m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/89452585c02f42fa80faf4280a8f096b : SUCCESS in 1h 57m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/789e7a91f4a64b55a3e8ac5a2ea60bf0 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a22430003a9749dd87ba342ed8e2c78d : SUCCESS in 3m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4576d203c7474c92a3cf45ef665601f6 : SUCCESS in 46m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16d494f9325d4f61b7ce7e4367860b8e : SUCCESS in 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d822e89f3d05e3b470a8f1f9cc9d1326acdce22e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:39:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbf2ad06b8f143da87c657a8d9aeedfe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/014863dc87d842fba68701951e31012f : SUCCESS in 28m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/56cc48eb13e74905a55c3460075cd28e : SUCCESS in 1h 10m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6e605b08698241b494032a7bdc8b79d8 : SUCCESS in 22m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/39fbc51cd1424b27bc912a1d8868b0f9 : SUCCESS in 29m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c7570d2de554403c8c1b510981f4b608 : SUCCESS in 1h 05m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c44cddd9475f401fa6c4e98890a6727e : SUCCESS in 1h 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece41ef6ad7b4d82bcd10e5809b40a20 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9abd969bfd549d5aa96836a06d899f3 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d3905f63f8e413a9c4768b602e239e3 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8146d34186614c5eaa14b46ea1a8322e : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8d375ed0b14957a4ee9eaa1df2a278 : SUCCESS in 5m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b0bf797ea08f44de9d693cc04d36c070 : SUCCESS in 2m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/67acc887ac83475eb5bbcf94e0bca4d9 : SUCCESS in 24m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/89f1ef6f841648daa25ac9dabc917688 : SUCCESS in 1h 06m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6bbea4bf35eb4d2289395fc33c973a46 : SUCCESS in 1h 03m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c519ff16c54b47e487825b7922b0437b : SUCCESS in 1h 30m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d1eb776d7b764ff0a72a5d300ddeec0a : FAILURE in 1h 35m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01f957b7214c4483b3769bf98824dc87 : FAILURE in 1h 08m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9f212fc8169a40d391e80e2bccdeb1f8 : FAILURE in 58m 21s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ed11e570453241c69141885e1e267ee5 : SUCCESS in 27m 22s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1e40b2ba83a840728c510021f324e42a : SUCCESS in 56m 55s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/703b9698fa6346ffae351af49a985462 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9ac730c30eb24fcd9f2fd67490bf72f5 : SUCCESS in 1h 25m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/addb0f7a66f4431f925b27f1c0ddac72 : FAILURE in 31m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4c245fd034c34bf8ad79d0c9bf4fdb09 : SUCCESS in 24m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cbcccdbc4fe541ff87640b60cae3b913 : SUCCESS in 28m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3d136cba17a84058b76a6402179bac3a : SUCCESS in 1h 03m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/af514a8c858448e3a68632395283291b : SUCCESS in 1h 44m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a8bf8331022b42f6b4654004fdf19321 : SUCCESS in 1h 52m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6f71a3b552b2498d828a28a4d11bab90 : SUCCESS in 2h 21m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/23cff1ccacc642a8b9f3b9e2e58bea80 : FAILURE in 55m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/41a69599064e401987b0565886bef1c2 : SUCCESS in 27m 57s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/48927cbea312494caae991ed2ad27c90 : SUCCESS in 1h 01m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/47ba7d00d8c946b0bb1c0f7f6bcbf569 : FAILURE in 58m 53s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fa290371c68a4cc587e300c251c33f92 : SUCCESS in 1h 38m 20s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/357ea93d9b9e4b1a95f4ff899a36a036 : FAILURE in 28m 19s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7c29280456c940a28068a347c89a49f4 : SUCCESS in 2h 07m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e5c2353e7bfbce323f9f0ca56aae1ce486dff323","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-07-28 14:40:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2334179ec150b53dff6d9958b60d6d14951abb8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:41:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1e164e95b86169db8ec56e17f2fc777a586d9776","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-07-28 14:51:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e466562f7125febd21992a7a4302f48a4e6e66b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 16:38:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1034da4ce38b4447aa77f16de56dd7c2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0064e0f2ecba4cda82253f4be0e9baa1 : SUCCESS in 21m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/22de0815fd104d2586757df8268e7d2a : SUCCESS in 27m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1a82a03f57d43a0b20acedc9954db9f : SUCCESS in 1h 10m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/45514489589246bc86482272178e5972 : SUCCESS in 1h 43m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87a408abee04384be90a96d6fb7f0ec : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0db6accdb024b958024c7c348bf9391 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e63b35712dab4532a4ec0faae03a7f5c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86aae2142ba24865bca4ebf52a4bb039 : SUCCESS in 6m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/460a496d21294737bcf1e5c0813c1c13 : SUCCESS in 2m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1470402407254929b55c00f1b0e4a854 : SUCCESS in 20m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/146703e12a3f416d9ebf1410d1868c48 : SUCCESS in 1h 17m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0a611919061a4942addb0fcece45f1b5 : SUCCESS in 1h 04m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7367382cb685460eb5f1b2789dd2636c : SUCCESS in 1h 32m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/972fbc9535994d9fb67122fbe5571b1c : SUCCESS in 21m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/04ec048521d84c0a85a90c4c6975e89b : SUCCESS in 25m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a442656bf0e3477ab67d3a0a42aabac3 : SUCCESS in 1h 02m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bf37f78427364fd0b18f6f77048aeea1 : SUCCESS in 1h 56m 12s","accounts_in_message":[],"_revision_number":5},{"id":"f10d7cb18f2a6b6ee144d488ac27c5ee33f6725a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 16:38:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"aec3732d5095b038f674f8711ddac5ebff45bcab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 16:39:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d60dbce57f425f926160ba0376508a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c775a4c0a845428ca528bcd1717e3d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c8f4ac4ec6b4c2a6691f6a07294c2da8d51baa33","revisions":{"d99841a45f994bb1e501cb921d4c7b2a4a1d1cea":{"kind":"REWORK","_number":1,"created":"2025-07-23 10:37:59.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/955680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/955680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/955680/1"}}},"commit":{"parents":[{"commit":"6661f59278d2a269c34ccbbd3960ba4ed63da70e","subject":"Merge \"[base] Ensure full control over APT sources for Ubuntu\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6661f59278d2a269c34ccbbd3960ba4ed63da70e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 07:56:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 10:37:59.000000000","tz":0},"subject":"change erlang arm64 repo to rhel9 based","message":"change erlang arm64 repo to rhel9 based\n\ncentos stream 9 diverged too much\nnothing provides libcrypto.so.3(OPENSSL_3.4.0)(64bit)\nneeded by erlang-26.2.5.14-1.el9.aarch64\n\nChange-Id: I3a0aa82d0903b341156c2fce68393cc4031e7728\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d99841a45f994bb1e501cb921d4c7b2a4a1d1cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d99841a45f994bb1e501cb921d4c7b2a4a1d1cea"}]},"branch":"refs/heads/stable/2025.1"},"a32b26ab2103668923fecf658500b873aecf11d6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-23 10:47:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/955680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/955680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/955680/2"}}},"commit":{"parents":[{"commit":"f7bfab8bf9bc7d9b1b8f0e65b561d287b943fd9e","subject":"CI: Bump ansible-lint to \u003c26","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7bfab8bf9bc7d9b1b8f0e65b561d287b943fd9e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 07:56:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 10:47:50.000000000","tz":0},"subject":"change erlang arm64 repo to rhel9 based","message":"change erlang arm64 repo to rhel9 based\n\ncentos stream 9 diverged too much\nnothing provides libcrypto.so.3(OPENSSL_3.4.0)(64bit)\nneeded by erlang-26.2.5.14-1.el9.aarch64\n\nChange-Id: I3a0aa82d0903b341156c2fce68393cc4031e7728\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a32b26ab2103668923fecf658500b873aecf11d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a32b26ab2103668923fecf658500b873aecf11d6"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"7cc2166ef52742f57dbfa35df9c0e4be5e805440":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-07-23 12:41:20.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/955680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/955680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/955680/3"}}},"commit":{"parents":[{"commit":"6661f59278d2a269c34ccbbd3960ba4ed63da70e","subject":"Merge \"[base] Ensure full control over APT sources for Ubuntu\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6661f59278d2a269c34ccbbd3960ba4ed63da70e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 07:56:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 12:41:20.000000000","tz":0},"subject":"change erlang arm64 repo to rhel9 based","message":"change erlang arm64 repo to rhel9 based\n\ncentos stream 9 diverged too much\nnothing provides libcrypto.so.3(OPENSSL_3.4.0)(64bit)\nneeded by erlang-26.2.5.14-1.el9.aarch64\n\nChange-Id: I3a0aa82d0903b341156c2fce68393cc4031e7728\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc2166ef52742f57dbfa35df9c0e4be5e805440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cc2166ef52742f57dbfa35df9c0e4be5e805440"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"992b7a17c9164fbe041b13a4859b1905224a0dd8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-23 12:54:02.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/955680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/955680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/955680/4"}}},"commit":{"parents":[{"commit":"32201c261c144d236d8902367cc27e1aba389587","subject":"CI: Bump ansible-lint to \u003c26","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32201c261c144d236d8902367cc27e1aba389587"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 07:56:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 12:54:02.000000000","tz":0},"subject":"change erlang arm64 repo to rhel9 based","message":"change erlang arm64 repo to rhel9 based\n\ncentos stream 9 diverged too much\nnothing provides libcrypto.so.3(OPENSSL_3.4.0)(64bit)\nneeded by erlang-26.2.5.14-1.el9.aarch64\n\nChange-Id: I3a0aa82d0903b341156c2fce68393cc4031e7728\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/992b7a17c9164fbe041b13a4859b1905224a0dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/992b7a17c9164fbe041b13a4859b1905224a0dd8"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"c8f4ac4ec6b4c2a6691f6a07294c2da8d51baa33":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-07-28 12:17:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/955680/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/955680/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/955680/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/955680/5"}}},"commit":{"parents":[{"commit":"32201c261c144d236d8902367cc27e1aba389587","subject":"CI: Bump ansible-lint to \u003c26","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32201c261c144d236d8902367cc27e1aba389587"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-23 07:56:55.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-07-28 12:15:58.000000000","tz":120},"subject":"change erlang arm64 repo to rhel9 based","message":"change erlang arm64 repo to rhel9 based\n\ncentos stream 9 diverged too much\nnothing provides libcrypto.so.3(OPENSSL_3.4.0)(64bit)\nneeded by erlang-26.2.5.14-1.el9.aarch64\n\nChange-Id: I3a0aa82d0903b341156c2fce68393cc4031e7728\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit ac1969542df12362ac3218d9d7437ad1b1fbc387)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8f4ac4ec6b4c2a6691f6a07294c2da8d51baa33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8f4ac4ec6b4c2a6691f6a07294c2da8d51baa33"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
