)]}'
{"id":"openstack%2Fkolla~956456","triplet_id":"openstack%2Fkolla~master~I7a2f3fce223afd74741b40bf62836b325fca5b19","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-06 13:20:36.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-06 09:53:21.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-06 11:17:49.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-06 15:13:35.000000000","reason":"Change was submitted"},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2025-08-06 13:22:07.000000000","reason":"\u003cGERRIT_ACCOUNT_37855\u003e replied on the change","reason_account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}}},"hashtags":[],"change_id":"I7a2f3fce223afd74741b40bf62836b325fca5b19","subject":"send empty X-Registry-Auth for anonymous pushes","status":"MERGED","created":"2025-08-04 09:33:05.000000000","updated":"2025-08-06 15:14:30.000000000","submitted":"2025-08-06 15:13:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956456","meta_rev_id":"bbb1e295d71366b0f2169e709954c2c2cb59eb75","_number":956456,"virtual_id_number":956456,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 15:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-06 13:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-08-06 09:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-06 11:17:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-08-06 13:22:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-08-06 13:20:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-08-06 09:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-06 11:18:12.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 09:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-04 10:16:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-04 11:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-05 06:59:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-08-05 06:59:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-08-06 13:21:32.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"}],"messages":[{"id":"b4eeafb7b311381b3ae0a9f2e1a79f47b1118d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-04 09:33:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00405dae604586e015651964250fa610fc640a82","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-04 09:37:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2226fce226177cd379542b252891956b3b084eee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-04 10:24:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0060964faafed8bfcc8379032286cb034a872a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 11:33:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34bad793f4064d688a4721f4f1223610\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7ec2f065df524d77b8207b9c5e4013aa : SUCCESS in 15m 26s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/f211321b86674a819177e943a877b90b : SUCCESS in 1h 04m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f77cd5ba6be4bb0b211da0ef1005cb2 : FAILURE in 10m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8a87e9137cbd47b88f7e2e8fa5d6c506 : FAILURE in 12m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bd553b17a1bf4ce28325674909ad7b53 : FAILURE in 10m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0d8377cd5f7a4c7a9904e5e5a92dd9b6 : FAILURE in 12m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e699b19f64224d4bb518faae04ab604d : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc380b412e54e2494c4af638a56a754 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3888a1a67c6441b7bfc6ddb5f4fbdba2 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3fb5e830e2041e6be1d84435c8d8f69 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58ddb4d92b67477f9206537447b292b4 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5315a9f623864058b6483f420f6b57d3 : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a0f5356cfa74314bd3924b3a945252d : SUCCESS in 4m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4781ac24f12748cbae8adf2db9331f4e : FAILURE in 10m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2319bee40b3542d0ac1dd0ac4f06f41a : FAILURE in 23m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c152ee049e804467b5f70f7b16dcc6bb : FAILURE in 19m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cd9cee193000458db4f51dea1b5133a8 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ab4d14eacc5a15831cc9dbb118d03361d75fb9ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 13:06:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f58e7136d647f380eeb4aa9cda90f9\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/881c880ec19f4d7fa3285f0f1b3cce7d : SUCCESS in 1h 29m 47s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5f0af22a40be40bbb04281cfe504fe7c : TIMED_OUT in 2h 32m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9bdfd1b0db9745a98c9b0fae43cb1018 : FAILURE in 35m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/956cdd04e3354504b7d2feffb9ab18ba : FAILURE in 31m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16abd016d5a34af39a9316b161509c9d : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/957f3e9305c645d1948b0e9680641c9f : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2e5e32b1f084b3d9b29fa883c5f9ea9 : SUCCESS in 9m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0cd55c2a37f84996b3298bb7dc7b815b : FAILURE in 1h 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"699c571d70e63523720d25444b34b0d1c162ef75","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-05 06:31:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"cb54396e846e41a9febec209c7cfa56a0dfa8440","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-05 06:31:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bba3bfc8bdc4df467dcbabdffc8fb2dc4949cfd7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-05 06:59:17.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"a8ce30b25c36b7903f38e00136bb9c6601385245","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-05 07:08:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"66019f442e3e55b5ac909a12f690c92e674547d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 08:40:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32fe876812c1420b8d653ca3e7bcbee1\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/b0f6e6ac9883415f840a6c90d12a7a8d : SUCCESS in 15m 13s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/3c04221098c743b6b9dce204a3320ecb : SUCCESS in 1h 37m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/98126abaeec041fca6156024a91cd641 : SUCCESS in 16m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/406d4962f22549eead3056dfb62ebe8f : SUCCESS in 41m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e0673918bfa948219356da4ec5a54879 : SUCCESS in 1h 10m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6782a6a90ffc4e4b817a802e4d5a14d5 : FAILURE in 1h 13m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0142b48801e24819b880c4033604b97b : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ace276163a7546e9a4b7fc2037619b19 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9866e6785a444b3d9298498827d9b8d3 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4127a689ef1f4564a9033e6f1a05c867 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5264ed0513dd4171947442beac07d0b3 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f728233d0184ae1bc6f42e63f492dc3 : SUCCESS in 8m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/47c3ce3ddf4a44c49bac42f76d7c03b7 : SUCCESS in 2m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ddd14ef44ee54c1d8f78b530db3c97f0 : SUCCESS in 21m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c45930084a4b4ac1861d7479fefe741f : SUCCESS in 27m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1f4f52f3030145f4b55606eed30b3d74 : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3f682a59c0be49618caace84422425f1 : SUCCESS in 2h 02m 44s","accounts_in_message":[],"_revision_number":2},{"id":"463798d49b24d3ee9f05a5a80af0da33fc5706a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 09:09:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c098c68fb2374e8593483f2301dffdbd\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/07681195a1d64d558cd758fd3b3a390e : SUCCESS in 1h 31m 52s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/28779e3277914b2cb5ad58e09ac6da7c : TIMED_OUT in 2h 32m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c13497a3e4de45418da5530730d55869 : SUCCESS in 1h 32m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b57378ffcaaa4f02bedee366a41e336e : FAILURE in 2h 17m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0dcaef0498b045c2be12ba322a7c2d94 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f68131ea5c644ebd8e5ebe76ca03715c : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85f7e84f9a0145f5ba3c4fd08f6774a4 : SUCCESS in 10m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a18a35d8a6b3479d90f7091a849c872f : SUCCESS in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7f1592db9122a09add2674cee02c1956cf9e9e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-05 11:03:17.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bc0a33535ede129afd0792a2ebae869462f5b353","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-05 11:08:29.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9cc53f5565b2e66181b987022da58b2b9b17b47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 12:49:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e094705752044f5aa73f7fe4b66ed690\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/75389c14d57e4ad5b02e7cd610814d24 : SUCCESS in 14m 59s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/54bc08afff5a450fa5c3eda39d76f3c0 : FAILURE in 6m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/47b74089dfb146a1880feb8dd7a844d2 : SUCCESS in 11m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0a49818309d24896bb02923d913be797 : SUCCESS in 33m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d57193aae0ab4d77a569a5531a0b7c99 : SUCCESS in 1h 07m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dd8f2fbd9e174388b12a5d5d14aa542c : SUCCESS in 1h 36m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2245d43402476dab8ae0b3032d2844 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3af840d4e4449568d099d5a01d66741 : SUCCESS in 2m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8341bb72096c446787eafa14b7843f1f : SUCCESS in 1m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bfed5da5d4542718e34a311e9c06bb7 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2700a00bc9341708ffbceac1a55ad13 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f01d0ec72954120a0e5e40cfc027caf : SUCCESS in 3m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e750dc3007ca4f9c8f9b505e6594d29d : SUCCESS in 2m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e397f1649904782b2816546d7e74f4f : SUCCESS in 11m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fbe6953d8eba4039beb786c1e2faa35c : SUCCESS in 29m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d4988c2d9e364dd9a7104b7fc83f94e9 : FAILURE in 8m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d0ac422498d7423dbada715f39e2f624 : FAILURE in 5m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ad9fb5838e8a7e8793a6338a0ad4f8b8514fae13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 13:34:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0878e9bb548943bbb343c4d63a35eccf\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d238c9b317934d2e87fd0cb1f88e7249 : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c702588048824d1d8b27923ed572fab6 : FAILURE in 53m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b7b759acaa7f45288a1706ebd5796777 : SUCCESS in 1h 04m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f6a7a3220fa43b2a4ee4bb0383f5a4a : SUCCESS in 2h 24m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27607c0e78024005b0caf79d1ff8e46a : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab50fa69d16a44979fe0b9b6d031f70f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a010147160b74168a190d6761723b23b : SUCCESS in 9m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3f1f484af6614e6188cec7420454db8f : SUCCESS in 1h 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9114a7362adc283b1fb1ba7feafcac7af7f438e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-05 17:30:44.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"90af7db0a400b46ad67952cdcc0f1663e8af65ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-05 17:47:05.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"57d8baa8c9a4845f0454ecd220f0fe5e7d413756","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-05 17:48:08.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4ae3d9e5dcbfc0918cf8b7f4d3d85596b1ac3a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 19:40:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/522bc998987d4c2bb0ed8fc3175abed8\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/3a8aba7956124599924fc988b05777b8 : SUCCESS in 14m 51s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/0970733a6b3a4bd78b7b4c5f10f59a77 : SUCCESS in 41m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6175833497fb47a9a30ddba9ec195d45 : SUCCESS in 12m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/42b070f962a0438fa46abaf1d601ef0b : SUCCESS in 15m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ba237322a894763b5884f26b6bdd16c : SUCCESS in 38m 09s\n- kolla-ansible-debian-podman https://zuul.opendev.org/t/openstack/build/05f36cc5d7524f709e4e77ffa08ddb4b : SUCCESS in 49m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e21f56fb74ca4cf88571834ef996ca1d : FAILURE in 1h 41m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b32570594bb941739c4bd89528daaafe : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13ed74ea86c64dfc90ee732e45a93314 : FAILURE in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2e155355b594a9e885aa68b59f040e4 : FAILURE in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1137435913de4646ad194d939b20699a : FAILURE in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41beb8c9a79e480b9679b628fedb2c37 : FAILURE in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba712d9288c48e5bbed251c3e208e63 : SUCCESS in 5m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/40ad71381ed34da0b795c4ee01059ab6 : SUCCESS in 2m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e08effbe612240f8a6f7c35a9654180e : SUCCESS in 10m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bad20f12153d447895b9966b0ad99787 : SUCCESS in 22m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a8f5d3a00f64485cb6f7f843d0811472 : SUCCESS in 52m 03s\n- kolla-ansible-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e6a972c6862749b5819ff84daab01a39 : SUCCESS in 1h 18m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/75eff94a2d034ca9b246f3b7fbbf6830 : SUCCESS in 1h 46m 59s","accounts_in_message":[],"_revision_number":7},{"id":"3f28b480844cf1d4df7d48273faa95f2c1de2481","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 19:55:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ad574d1c2ed44da8adb45830c1175fe\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f6afeac0289b494e87b11ff54c9c1e60 : SUCCESS in 56m 22s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d7e69c26b1e042ffae30ae60b45609c7 : SUCCESS in 2h 04m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee389754fc244a6c9b0a5f11fd58d297 : SUCCESS in 1h 00m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/659603887f6e4dbdb031fcfccfea80f4 : SUCCESS in 1h 59m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10bf64c312564053a7d5da21a3627c56 : FAILURE in 3m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46ffbf1ac4aa41338fd0a7ed780adde3 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6b240e1ba70d4d13a946e2d9ceb26983 : FAILURE in 8m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/40c4d201665444d08fc28ec016236c60 : SUCCESS in 1h 00m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3113720532209ef69803820aa7449c057a5daec5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 06:39:48.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"116982cc7ef0928654829a6385232e158a70198b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 06:56:48.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":9},{"id":"39418881fc63e18631c13817a4d042f875a18500","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 07:57:38.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"cd28fac11c7adc5b79bad64aeec0bf4b4e79163a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 07:57:41.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"82c4de67919e385753592398915b21c63365da3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 07:59:18.000000000","message":"Patch Set 11: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":11},{"id":"1c02ac1a8e94e146b4b432b420d961243c49840b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 07:59:28.000000000","message":"Patch Set 12: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"caa1a0aa2d6b37af8814b5561c4b671eb7c0b615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 08:11:04.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"cd5a7c58ab26d63b85e95253c74fd3dbbcf838a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 08:11:54.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":14},{"id":"3d1b7c291b162cfaeb0f8f50718ac5825dbf1395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 08:30:53.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"20efbdd1212bc206d3ee1abe611070dd3d7d8b49","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 08:33:01.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"b1394c03bfa030fca4538f8324ad12d5572b0f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 09:11:24.000000000","message":"Patch Set 16: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":16},{"id":"fb165f01c87a92a6a58fc001abad07994f30cfcb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 09:38:48.000000000","message":"Patch Set 16: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"c243757b879380e826d5f04bf4adbf6c8f7f397e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 09:48:56.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"5a6ec3c43a2b7969f4ee565b157c4c408230dced","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 09:49:45.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"dca93d16463ba15af3249c1997ec61639a9e36aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 09:50:46.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"a3a4cd00acfb8bf36466c5a3633fdbebad117877","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 09:53:21.000000000","message":"Patch Set 17: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"99242eede2a03d7af07e9ee8c863cc38fb67d254","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 09:53:59.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"a2c9f59cdecf1069c0dcfd684c180e508043133c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 11:17:49.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"b63dc29855016a4a0fb47746d661983bfa2d6d99","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 11:18:12.000000000","message":"Patch Set 17: Backport-Candidate+1","accounts_in_message":[],"_revision_number":17},{"id":"f833f2addad0f61f985f7dd9ce6a213767475386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 11:28:29.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/692d2cf44fc1492088d1b1d23681cf76\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/a83d3f24be0a4074bb763a58975473bd : SUCCESS in 14m 19s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/c2237b6c474a4a3fb8578dfa84e3e952 : SUCCESS in 1h 07m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ddc37113a8304704a109e0262cd2cb36 : SUCCESS in 19m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f3ef182f8bb0423294bebb6277244df8 : SUCCESS in 28m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/242e1ce6a36340d38351b3499ba8a493 : SUCCESS in 1h 13m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ef5f06a5990d4b249c1d41b50e0fa73e : SUCCESS in 54m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d893e0df67434b0188dc9b2b5decd813 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49fccddbbf4b452f8aa2b2627ba32376 : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dfaa60acf6c433099a6e4500cb6a4a5 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac1bad9464534ea98cba77ce0b35ed88 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0eb8b2561964f60a548ac5f14b7b567 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa790fceed1349eabf3f1f5be9507c8d : SUCCESS in 7m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/39d95ee48cd04a26a65ed9a6c63b5083 : SUCCESS in 2m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3d7690381da2474684f0ce8c29e66e3c : SUCCESS in 17m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1ebcf9a6687c41c0abb4f35108eb7f27 : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dedd8052d55d455ab154d253466b79aa : SUCCESS in 1h 05m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e8136f1dd84c420983b262144cb21265 : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":17},{"id":"2b4df8217ac806a97bd3f12dc5ff425b9f87424b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 13:13:05.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2d82b2aa9bb4f888b84e77c782c56b2\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5e7fa18e70dc40838f4707793d7ab76a : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7eadfdb7561149899d269fb0fba54d45 : SUCCESS in 2h 23m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e3bd9b6265754523bfa8bc868a80b490 : SUCCESS in 1h 43m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/525082f1e5234f44a16e32d2a9158471 : FAILURE in 2h 25m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e169638bd1d142258e755f8bab0b9be9 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e76a48839fd1455095e75545c55f5259 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/711f8bc905e24251bb404bf0c69875e0 : SUCCESS in 9m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f9c2444cb12444bbbac235ebf8f9ed2c : SUCCESS in 1h 26m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"bcf9fd0168eff6f7b6d8781f6c22f633dd852131","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-06 13:20:36.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"4faca6c2da8c2be6675b7901f393efebf2856aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 13:20:48.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"6cc2fc7ec2a7045e7b6684902a11d3d11e46557c","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2025-08-06 13:22:07.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"f1c6fb25029235f8321d2bf98de5cf9530911cef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 15:13:34.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4134904bfd2a4270b79608ad0e6db980\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/561e778b53d84209854a6193ff96cb3a : SUCCESS in 9m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9fcbe35c843549e1b0ecfdf8989e7bd0 : SUCCESS in 28m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a79df378aabe4cd0abbc6bd5083b3024 : SUCCESS in 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fdd761abfcf40b280b89daf4a0bd679 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44240430fa8f41ef8712478b3207b49f : SUCCESS in 2m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f079dba8d1c544efbcfb5eb89dd8d520 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7946a3b041e42038b4f9f3b6defd6ec : SUCCESS in 7m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0b17e73436fd4d638ae281addaf8b382 : SUCCESS in 2m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/57955d9c82b1420bb171454bfe8ffda9 : SUCCESS in 21m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5666776a43e24415a4cae522dc11ff7a : SUCCESS in 31m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b9dc16c5fbbe499c8a4f3e96d12bddf7 : SUCCESS in 1h 00m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5962c085bc34445bad7bbc07ff6c7888 : SUCCESS in 1h 47m 57s","accounts_in_message":[],"_revision_number":17},{"id":"9e7fc2eabd5dc8d9e9535c3f8281f976ef6f2a31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 15:13:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"bbb1e295d71366b0f2169e709954c2c2cb59eb75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 15:14:30.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3482c80f3aa4c849bfd9da0ac9f064d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2574e1b7aee4091b76d4725989104d3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"ddac7ca1ed393609b7bda8753db3598f1ae02648","revisions":{"89344821d185b9e5b14f9e9a801f01b671fcc5d0":{"kind":"REWORK","_number":1,"created":"2025-08-04 09:33:05.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/956456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/1"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:33:03.000000000","tz":120},"subject":"always send empty X-Registry-Auth on push","message":"always send empty X-Registry-Auth on push\n\nDocker ≥28.3.3 errors when the header is absent. Passing\n`auth_config\u003d{}` makes the SDK emit \u0027{}\u0027 and keeps anonymous\npushes working.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89344821d185b9e5b14f9e9a801f01b671fcc5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89344821d185b9e5b14f9e9a801f01b671fcc5d0"}]},"branch":"refs/heads/master"},"ba4616593598d13469aaefd07c489146194f2188":{"kind":"REWORK","_number":2,"created":"2025-08-04 10:24:56.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/956456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/2"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:59:58.000000000","tz":120},"subject":"always send empty X-Registry-Auth on push","message":"always send empty X-Registry-Auth on push\n\nDocker ≥28.3.3 errors when the header is absent. Passing\n`auth_config\u003d{}` makes the SDK emit \u0027{}\u0027 and keeps anonymous\npushes working.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba4616593598d13469aaefd07c489146194f2188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba4616593598d13469aaefd07c489146194f2188"}]},"branch":"refs/heads/master"},"e0b29f3ee005804bdf0f2df5a718ce53f529e28c":{"kind":"REWORK","_number":3,"created":"2025-08-05 11:03:17.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/956456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/3"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-05 10:51:23.000000000","tz":120},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0b29f3ee005804bdf0f2df5a718ce53f529e28c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0b29f3ee005804bdf0f2df5a718ce53f529e28c"}]},"branch":"refs/heads/master"},"68e297be04f2107d8e592adb685f9f13ee52f0d6":{"kind":"REWORK","_number":4,"created":"2025-08-05 11:08:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/956456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/4"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-05 11:08:24.000000000","tz":120},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68e297be04f2107d8e592adb685f9f13ee52f0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68e297be04f2107d8e592adb685f9f13ee52f0d6"}]},"branch":"refs/heads/master"},"0a34a1f3e5815f314a58ff31955e63c0e587f378":{"kind":"REWORK","_number":5,"created":"2025-08-05 17:30:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/5"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 17:30:44.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a34a1f3e5815f314a58ff31955e63c0e587f378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a34a1f3e5815f314a58ff31955e63c0e587f378"}]},"branch":"refs/heads/master"},"799ee4d24fd765ab6df0a8705f842a2a09e30bf6":{"kind":"REWORK","_number":6,"created":"2025-08-05 17:47:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/6"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 17:47:05.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/799ee4d24fd765ab6df0a8705f842a2a09e30bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/799ee4d24fd765ab6df0a8705f842a2a09e30bf6"}]},"branch":"refs/heads/master"},"8aebccd244480e523be8c89a6e8c318fd637767a":{"kind":"REWORK","_number":7,"created":"2025-08-05 17:48:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/7"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-05 17:48:08.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aebccd244480e523be8c89a6e8c318fd637767a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aebccd244480e523be8c89a6e8c318fd637767a"}]},"branch":"refs/heads/master"},"27f9e1192e25356b9556efa982502b17b378ee69":{"kind":"REWORK","_number":8,"created":"2025-08-06 06:39:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/8"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 06:39:45.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f9e1192e25356b9556efa982502b17b378ee69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27f9e1192e25356b9556efa982502b17b378ee69"}]},"branch":"refs/heads/master"},"cb6dcbe5df736a7539bd37692d7c7663ff4c0704":{"kind":"REWORK","_number":9,"created":"2025-08-06 06:56:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/9"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 06:56:39.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb6dcbe5df736a7539bd37692d7c7663ff4c0704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb6dcbe5df736a7539bd37692d7c7663ff4c0704"}]},"branch":"refs/heads/master"},"5ffbffe11ea1ed331fbbd883fa23cd890efef3b4":{"kind":"REWORK","_number":10,"created":"2025-08-06 07:57:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/10"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 07:57:35.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if `self.engine_client.api._auth_configs`\nis empty, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ffbffe11ea1ed331fbbd883fa23cd890efef3b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ffbffe11ea1ed331fbbd883fa23cd890efef3b4"}]},"branch":"refs/heads/master"},"5c2a31b75c458a6a970e932622c2fa88a1014368":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-08-06 07:59:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/11"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 07:59:17.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if return from resolve_authconfig\nis empty/False, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\nUnit coverage\n-------------\n`TasksTest.test_push_image_with_credentials` proves nothing is added\n  when creds exist; existing push tests cover the anonymous path.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c2a31b75c458a6a970e932622c2fa88a1014368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c2a31b75c458a6a970e932622c2fa88a1014368"}]},"branch":"refs/heads/master","description":"Edit commit message"},"861d031b074bf458a8af408b5338b985e13d591c":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-08-06 07:59:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/12"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 07:59:28.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if return from resolve_authconfig\nis empty/False, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861d031b074bf458a8af408b5338b985e13d591c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861d031b074bf458a8af408b5338b985e13d591c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c8c6ac8fb0a0adeb4dc3bff0212bdef9e305466d":{"kind":"REWORK","_number":13,"created":"2025-08-06 08:11:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/13"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 08:10:56.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if return from resolve_authconfig\nis empty/False, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8c6ac8fb0a0adeb4dc3bff0212bdef9e305466d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8c6ac8fb0a0adeb4dc3bff0212bdef9e305466d"}]},"branch":"refs/heads/master"},"235d47899c8e40564b3d92fba1e3e475ecfdeb17":{"kind":"REWORK","_number":14,"created":"2025-08-06 08:11:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/14"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 08:11:51.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if return from resolve_authconfig\nis empty/False, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/235d47899c8e40564b3d92fba1e3e475ecfdeb17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/235d47899c8e40564b3d92fba1e3e475ecfdeb17"}]},"branch":"refs/heads/master"},"ec25faade1581f8bcc62f8b7ae22a9f8b8d64f5b":{"kind":"REWORK","_number":15,"created":"2025-08-06 08:30:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/956456/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/15"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 08:30:52.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if return from resolve_authconfig\nis empty/False, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec25faade1581f8bcc62f8b7ae22a9f8b8d64f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec25faade1581f8bcc62f8b7ae22a9f8b8d64f5b"}]},"branch":"refs/heads/master"},"d0f82bc8682db358bd484063c819ff60b3f37159":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2025-08-06 09:11:24.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/956456/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/16"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-06 09:11:24.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nDocker ≥ 28.3.3 rejects a push that lacks the `X-Registry-Auth` header [1].\nThe Docker SDK adds this header when credentials are present, so the\nproblem appears only for anonymous pushes.\n\nWe now append `auth_config\u003d{}` if return from resolve_authconfig\nis empty/False, which makes the SDK emit the minimal `\"{}\"` header required by\nthe daemon while leaving authenticated pushes untouched.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nCloses-Bug: #2119619\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0f82bc8682db358bd484063c819ff60b3f37159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0f82bc8682db358bd484063c819ff60b3f37159"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ddac7ca1ed393609b7bda8753db3598f1ae02648":{"kind":"REWORK","_number":17,"created":"2025-08-06 09:50:46.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/956456/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/956456/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/956456/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/956456/17"}}},"commit":{"parents":[{"commit":"9f24ce63d0ce245cc732a9c47a7dd4c39ee42066","subject":"Merge \"change erlang arm64 repo to rhel based\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f24ce63d0ce245cc732a9c47a7dd4c39ee42066"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-06 09:50:39.000000000","tz":120},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nSince Docker 28.3.3 the daemon rejects a push that carries no\n`X-Registry-Auth` header [1]. The SDK already sets this header when it\nfinds credentials, so the breakage happens only on anonymous pushes.\n\nDuring `PushTask.push_image()` we now check whether the SDK can resolve\ncredentials for the target registry; if it cannot, we inject\n`auth_config\u003d{}`, causing the SDK to send the minimal \"{}\" header that\nsatisfies the daemon while leaving authenticated pushes unchanged.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nCloses-Bug: #2119619\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddac7ca1ed393609b7bda8753db3598f1ae02648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddac7ca1ed393609b7bda8753db3598f1ae02648"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
