)]}'
{"id":"openstack%2Fkolla~956538","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I49015b8fc72c83449767bfce0737b0e44ed75808","project":"openstack/kolla","branch":"stable/2025.1","topic":"revert-bug/2110122","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-04 13:00:42.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-29 11:14:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"25277":{"account":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"last_update":"2025-09-04 13:00:42.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-09-03 13:25:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I49015b8fc72c83449767bfce0737b0e44ed75808","subject":"Revert \"mariadb: pin to 10.11.11\"","status":"MERGED","created":"2025-08-05 08:16:08.000000000","updated":"2025-09-04 13:02:02.000000000","submitted":"2025-09-04 13:00:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956538-revert-bug/2110122","meta_rev_id":"af5ffceb8f0e3b5e8d12f52bd3766d4f4811ec6d","_number":956538,"virtual_id_number":956538,"owner":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 13:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-29 11:00:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-29 11:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 09:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 09:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-05 08:23:32.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-08-05 08:23:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-08-05 08:26:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-05 09:58:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-07 20:24:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-09-04 09:00:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"5b7037607f22cd4ee0b579eb75f8754b540a5ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2025-08-05 08:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2423d12991c4834be54377802afa0c31c8fe3bb3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-05 08:23:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8afad6681081d9cb8d402bbe66f985ea9e46b8ad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-05 08:26:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9407535f984def9dc1cae27e254d982f36fda253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 09:58:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52332c887f4c4f9b82e4c2074cf628b9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/30e2751e7e694927a212c9f4a5564880 : SUCCESS in 25m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9ab01d5b2d2441fc86793570bbaef742 : FAILURE in 11m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1aa7a13ced36498a9a550df7c79a29ca : SUCCESS in 23m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9de57f92f3b54110902af7bba71c2ef0 : SUCCESS in 17m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/047020ce270642c7a8079482987ec31e : SUCCESS in 58m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4ec499bb583c41e480341150f768d974 : FAILURE in 1h 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4508886ceed4b249ff0a9b6e62c65e0 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a857b43fc6a4eda9f3e7e5757058250 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cb208c3fc7a4613854388fa357fa2bb : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c459fe731a143b580f279398569c9ee : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960a42efee69415da9b568fcc7e1586f : SUCCESS in 7m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3255e24cbea54be2b8fee850cc1ae4e6 : SUCCESS in 2m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/081cadfda6344ca1b6d3e029fafee068 : FAILURE in 29m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/552b271674464831868f33ffe738430f : SUCCESS in 1h 09m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/32217ed2ac2e4c80b18c2519fe64fd4d : FAILURE in 17m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8decf678af864adeafd74b0971b606ca : FAILURE in 10m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9765fcf9eae943e3a9c82f2b803fffad : FAILURE in 16m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c4cbcb2af0ef4738b2339f1e353b2835 : FAILURE in 15m 22s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ddc28f43c01949848a231ea9e4592c38 : FAILURE in 14m 30s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/53fc72685e29468e86df5d69b5bbbd38 : FAILURE in 10m 52s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/524c0b1ba95942ccbcae86e9aef76531 : FAILURE in 19m 32s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/47c5a75a16ef49ef802263b3a63e23d0 : FAILURE in 14m 28s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/34c1f23e2ce9454ea5f8e3860275e7f2 : FAILURE in 12m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/77ea86892edc487886602d4056e5f8f0 : FAILURE in 10m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/410bcd65a5564079bc467cb1ea7d3711 : SUCCESS in 22m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6c5177807ff742389a8936e3cd5befb3 : SUCCESS in 28m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6bc8726e287d4099b4ac8928d58730da : FAILURE in 16m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8ca4a4292b534c68879ecee813e247dd : FAILURE in 11m 44s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3b7c9709bf7e4d75bf5a818a5e606730 : FAILURE in 13m 18s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1199d21db3d04bff9200cdf0ed3c8089 : FAILURE in 13m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9f70cc2fbf1d4f1a89d7a9cb839879a6 : FAILURE in 15m 23s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ff914dfff76a4184b95f0cb7f5d872e2 : FAILURE in 11m 15s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f3e3356eaee04842a207320aa3a7a70c : FAILURE in 12m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ac1ad37eea4746a8ba859596ab646a9e : FAILURE in 5m 04s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/45e8a7539e7a4ad4b6d8f019186b09be : FAILURE in 12m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/59c185e9089b4014af2ed982be196f56 : FAILURE in 10m 29s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/428b405cb1694ca9b462d955a0e74afd : FAILURE in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d967b4f2bf469d18f90210137184ea7d8d2f9d5a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-05 10:10:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"973226b2df5a27659d15e47cd71a696c352d25ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 11:47:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f427626c3ac4499adfd11947491c604\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/89d19e1a01274bf3a2a37683aa48368c : FAILURE in 1h 15m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ea9ff25718bf49508525cb8dcef340ee : SUCCESS in 1h 21m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f4ed4caca9064a9f85fab53a6282db7e : FAILURE in 2h 35m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/679ba00bdafc4e678a2b9526c53a6db4 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85845d9611554c09815f205911513156 : SUCCESS in 5m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8c1380d280aa4aa0946c0934478f7ce8 : FAILURE in 1h 22m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8cd52652eaf643a089195fd917acc707 : SUCCESS in 1h 30m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6aeaa482d72a959de70a1d4480662b2474fca817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 12:15:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c34b6765822406cbb710dbc01343759\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6dc69ef557394bcb86393f64a1aa3539 : SUCCESS in 12m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/48614372a8cc43b4bc9cc214fc5b3c7a : FAILURE in 8m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a3ce394c9f34489180da0b96dde37e50 : SUCCESS in 19m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/276d2547eba54960b687a78f0bc5828c : SUCCESS in 31m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/63ed211ca67b4c3b9a3bfe6e91219ed3 : SUCCESS in 1h 09m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6d400bc49b247f8a0d122e77261fa9e : FAILURE in 1h 55m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb6e259902e4d009b51fc0d1571cdc3 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88753df8f36c43f6a1fc3eaeef66f96b : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d4e3a56458443a5a9617600d6f55c15 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce80263e65a426db7eb46a1deec7ae0 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c41b29faf045f2bceb2935735c0a16 : SUCCESS in 7m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f7149435d4a04387bc4d5addc9f4ece3 : SUCCESS in 1m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/98f2e9e0ecff4b86a7480171a549ad31 : FAILURE in 19m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/57e0ccdcc14546abbf5c269f8e7b1626 : FAILURE in 1h 09m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/46aa7d1b3a994b28858f801c35b5c5ff : FAILURE in 13m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3bfd6b91476d440090e2423459dcc36f : FAILURE in 20m 01s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8400bb140bba425ca0b2395dd7412953 : FAILURE in 14m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9d10224974514bedaad22bab79ccd3e3 : FAILURE in 15m 01s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a9c02c572df0423cbdbe3191de9317ba : FAILURE in 14m 09s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/fd81e443a3d14bb9adc3ecd0fd765d2b : FAILURE in 14m 10s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f16b97a5bb704c61b6ee9a8d5f86dd48 : FAILURE in 18m 27s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9e1ae40e28de4e8094b649c712c78d68 : FAILURE in 17m 22s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/059527a46d794bc494152492c2e206cf : FAILURE in 18m 46s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bfc1ec9fc4da4fd7bfa52ea45f8f10ba : FAILURE in 12m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8f03db870e40439ab24eea1884066648 : SUCCESS in 10m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/528db4927dba499facf7aee1c0418c7d : SUCCESS in 13m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bfe0f61ac6e04613a8d7b706620d654c : FAILURE in 5m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dc0e7f290ecb4cc895bcaaf315270955 : FAILURE in 11m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e0227f3fbc2b4a0494c6fe448432a5e0 : FAILURE in 7m 12s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c5108863d42a450a8178c64d5de6843a : FAILURE in 12m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/eeb8969554ea44b8bb925961b58ccf65 : FAILURE in 17m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9b9a47f6a92747f3a50538ace256e5dc : FAILURE in 11m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b0a071a01e1844f5b22ae799ba30ec1a : FAILURE in 16m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/eea899b601174b00ad53c83ce6ec1b7e : FAILURE in 16m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/600f424cfa2340f0ab6fd9276c44f8c0 : FAILURE in 8m 12s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/58e362e852de413bbf1ccb03328d5eb9 : FAILURE in 5m 57s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/01d2dffeaa76408eb9cb19513d263aae : FAILURE in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4879b2fb08f938935f553c421a6945b4aa0eff6d","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2025-08-06 07:53:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b955dc8d9f2f961ce018bae2ffc0b817704c446e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 07:55:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab50beb30bf96505c48e4e3c2df5c80f303c27a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 07:32:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb3f16ca8e5ccb50abf8419bd7f9ff1ad9783253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 10:25:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7e16bc37a4d4734a8db7b94fb2c6cca\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/53a48b8707c9453a89dfe46ff9ec6065 : SUCCESS in 28m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/461a03bb473b495d9c27a80974493f4a : SUCCESS in 1h 15m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e687e4593f3a43d3a7b934c8a23cf17a : SUCCESS in 19m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8c8a033cafe04551aacb423b145b759f : SUCCESS in 29m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/633162b6995543c19bc3a093821c784d : SUCCESS in 1h 03m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b44be228425b47269112a7e5ba943904 : FAILURE in 1h 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5919ac4d7c73478382d39f74fdf9e388 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1bdd910b71d471ca6c798d93d18ac07 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38b74469069b48a68f3e7a37d68fec84 : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65bf845572384fedbc65198e4f1415b0 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f812c5c4874531a812b875ddb480e6 : SUCCESS in 5m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0c81ecd573f94b9695cfd4c2cecd4949 : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a21fcf532d404773a94db01a94d5de8a : SUCCESS in 20m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0e36d129d19040d69a747a9dc352a4e1 : SUCCESS in 1h 00m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/97697184c0b342ca8426f47b1228eaa4 : SUCCESS in 1h 12m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/33240b80481a4cc983e257b100785c10 : SUCCESS in 1h 44m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/82d1fa74fdd944beaa48c978c257cd98 : SUCCESS in 2h 01m 04s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bedd4a2e325347d087f4e86717a97978 : FAILURE in 1h 13m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fd9282cc1274481987960e3ba6afefaf : FAILURE in 59m 46s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c65078fd3d6f4bc7a4463d42adb9010b : FAILURE in 21m 53s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/326930341c1a4878b06cd11983770f12 : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/cc93ef749e924efc9849201f3dc55d63 : SUCCESS in 1h 14m 46s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9a90171edaa0444cb1b886d1a479fd0e : SUCCESS in 1h 26m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4033e3a5db27409fafd1baa680db1efa : FAILURE in 29m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dd8323e18eb540e0bd1d43654e3870e8 : SUCCESS in 26m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/316014e150bc44d7ad6e8269571a7173 : SUCCESS in 33m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5e009b3f9a3c4c3995d0b8196e496126 : SUCCESS in 1h 07m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7333ae4062ce4ffcaa3650581f280497 : SUCCESS in 1h 46m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d24be5119af24cfda1b3357d6dce17f9 : SUCCESS in 1h 58m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5a06dab459a4158b9efd7ac07c891eb : SUCCESS in 2h 38m 10s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4ee246b27fee4844bb915e4bf5624cbd : FAILURE in 59m 56s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7913b69f5fda4b1eab0884e3fc43c658 : FAILURE in 23m 55s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2429cf7f12b94306a871d39d7828a690 : SUCCESS in 52m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/922eb0888ced42da95e6cfc88c21c5e1 : FAILURE in 1h 01m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8efa0d6fd9264dc98273fdae4e5c7492 : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a81a2dc00faf479db13d27f4573278c3 : SUCCESS in 32m 00s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/44323a6851b04f0590df6353315dc80d : SUCCESS in 2h 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a3e647d1446e0c03443a3fa2fceaf88bb82ec46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:39:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18219dfda9b4c35adbe23a4f1eb612c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8348b852caba45919da2abf6f35db42e : FAILURE in 1h 33m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a76fdee9bdd34085a08c3eb13b1bd77c : SUCCESS in 1h 37m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41b8bdecae2e4449988b1bd529d17365 : SUCCESS in 2h 37m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40dfac0218414c68a57adc00b3a3d563 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a18bfa042104ea69b6d23e3884777d0 : SUCCESS in 6m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/56b4d20736204dfc88eb86240f110075 : FAILURE in 1h 13m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a79a649ba5804ecc85e4c71417c80e7e : SUCCESS in 1h 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"866965e91a90ea8beabcbf0cf895b13a6b7f2501","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-07 20:24:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cba68bba87d868bca2ebd0a85d7c9fd49a11830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 22:30:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f737a36813e4053875e954d038720fc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f392bbd7139c4c02b8f05a500c6c90ad : SUCCESS in 14m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/19cd3936e9cb4fc38b014c66c808d930 : FAILURE in 17m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5dd79d26b6c94746a8931e681c5d8ddb : SUCCESS in 21m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a1eacfd090fd4ac0868f117ac0d58222 : SUCCESS in 29m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7a0d7a7fd6384e819e3cdc4b436fed5d : SUCCESS in 31m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5eeaaef8836e4643978658d4232b2c58 : FAILURE in 1h 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b77ff2a9564e4f88d110755692093d : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c21348f123642c8aaf269aa81da7213 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de4095624a8a4dd0858121cb34eac28d : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32ad423544bb4528b9065f0411391c76 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a260492c89343d7954a168a72b0f91e : SUCCESS in 6m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b8aa9d022924ab9a0a4330816680222 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5c67697be04b4a8fa5e9df867e8503df : SUCCESS in 20m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bbb4d2f5935f4b508f1f51dacc87eb2b : SUCCESS in 1h 11m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2ce183f9ca9a4eb3a07ce99640d5a17d : SUCCESS in 1h 09m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ab477451f3744c028d85e7dd7186db4a : SUCCESS in 1h 41m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f6547159b07b4024b88c70ef9c294b35 : SUCCESS in 1h 54m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/271d9acfd6a7424989037ca3a1f1e5ee : FAILURE in 1h 26m 56s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f6cc5d390961401fa7b5f43e6974d0f3 : FAILURE in 59m 48s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/065f01ebb665439d8ed1fb183d8dcabf : FAILURE in 13m 21s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/53806f2115d64edcb11e286bffa17a79 : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/73f3362129454254b823a488225c7128 : SUCCESS in 33m 57s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2c8fa61d4cc240c9aa3163129afdaac3 : SUCCESS in 1h 39m 02s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b5673e8ed4f6497482d4df12e28e7c0d : FAILURE in 22m 26s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff3da9ab307e48adafdbd3e836488f6d : SUCCESS in 24m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c30e3eca14db423982d23f0db30a433c : SUCCESS in 25m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9870e6bbf6da477f877abbe45c0a633c : SUCCESS in 1h 09m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/654d018d1177480dbbc124a0a9821c19 : SUCCESS in 49m 27s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6a934046946b4b809c4af5e87ccac782 : SUCCESS in 1h 58m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0406ecd1c2ae47fd884f08e734ce7085 : FAILURE in 1h 37m 59s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/729a0f588f9f461fb84a54beb27e3bfc : FAILURE in 29m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/13281a66d972485ab16b1f027396ac34 : FAILURE in 24m 16s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2017f9a4452d4d728bee81dfb7bc1652 : SUCCESS in 52m 39s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/00115c09211242cf9f68745dd940bbce : SUCCESS in 32m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0546683f52054500b3d00e1f9385009a : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cd33b3fa780d42ebb1e2e5f3601bb4fc : SUCCESS in 14m 10s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/40a6f0e15f2a4b97bb0ec1debd7d8412 : SUCCESS in 1h 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11cb9615320b27a3ce10f2174df1dcc50d504ae0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-08 03:29:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5888cedf2bd3807d0d857cc06f355510eab0f3e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-08 03:31:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7f08400744c9cf0b09a51f7c933c361514c4a53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 05:53:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13036a0ad7b54a90a71ec8c3fa22678f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6133ff8e19724ecfa63599859fbcda33 : FAILURE in 37m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ecabc0c9e924f2d9df06f11ed24239b : SUCCESS in 57m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aca3f78e60d54abfacd23cc57cf196e2 : SUCCESS in 2h 01m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a882d94aee0f464b93938e19aea142ee : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a33f557fb4b84f85a89c5dc2939c061f : SUCCESS in 5m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/66b6fd8cdb5a4047b89104fc8e095318 : FAILURE in 49m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0643d7dd0868498dbf6a995446def6a1 : SUCCESS in 1h 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"365179e8caa7d0ad41eddc13066fa5dac01f6475","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2025-08-08 14:04:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9456943eb7a00813e3635970b0fd15893e518818","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-08 17:36:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dab91b1c2c30270ed761ef5dc9b37d27c6d99bee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2025-08-12 06:49:31.000000000","message":"Abandoned\n\nAbandoned in favour of https://review.opendev.org/c/openstack/kolla/+/956792","accounts_in_message":[],"_revision_number":1},{"id":"14f749cc9acf43347cbc4277c82e0e9cdb5f1386","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 19:56:54.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9b51b8180e516acc41f739047577d111d21b5a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 22:22:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea1956f470ae408ca43992fe66982745\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/985e9739828c40ccb3a2aa644376b109 : SUCCESS in 25m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/98630c5f3d22430181931023a415cd44 : SUCCESS in 1h 10m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7a275cde03894701aec2f8252a84c426 : SUCCESS in 18m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6991e0005b324f6fa24a58803f21bd24 : SUCCESS in 28m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/885821e062514179987e8de53a69f28d : SUCCESS in 1h 06m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4940cf408ec9474e8818c90705d104bd : FAILURE in 1h 29m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3f0825ce954489a55c1cdcb21a3737 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7d37a00ec244a859483ea762aef03c6 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6142efc671f5444d966a87aeec81e139 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e93fde052b6473aa811a4d332502bf9 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819b44eca79b484d8c29f9f42362f0c0 : SUCCESS in 6m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/748bfc22c5b74908aac41b64070aba28 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5af2d0496a934c7192a8ad1ccfd9435b : SUCCESS in 25m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d8c0ea4293754beab5a6715595087fdc : SUCCESS in 1h 40m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2ba62974de91440886d7a04d094c7e5d : SUCCESS in 1h 07m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cd54a3ef735145d8a2420a04d8c4e69a : SUCCESS in 1h 37m 56s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/001485dbafcc458fa007fcebd8ed92ea : FAILURE in 1h 41m 01s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/57c36f6470a14f9098efd94ac70ee651 : FAILURE in 1h 34m 02s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/431c7ddd2f9f4e40a19e60195b8f8e7e : FAILURE in 1h 02m 12s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b4ce78333cbd4549a8743d0f6431178e : SUCCESS in 26m 12s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f5a5484eac984ba596a53d34838f5ee3 : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0c7d132e9b49401bb00ed871b282d178 : SUCCESS in 56m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9796ae534d884b5289be6ec31afd9311 : SUCCESS in 1h 38m 20s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d7eed9dd92e241b7b8dccf4beba335ba : FAILURE in 16m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7c5e105dac6541beba366ee2d4d586c4 : FAILURE in 10m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6ddee822caf4454a98b704d4163a5bc : SUCCESS in 22m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f2417da4877746189fe5d3e28600e29f : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/392e9e286e3e450189b427ed3a423bbd : SUCCESS in 1h 23m 41s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e8f23bce1d424e3b84a3127d78b38277 : SUCCESS in 1h 33m 43s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5074b8ec485c45d6ac5d640bf2a08562 : SUCCESS in 2h 23m 55s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/18633b1ebca845f8bb54e62a65cefa68 : FAILURE in 1h 03m 14s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/01b15c04cdf6408c96f11691a6bc79e0 : SUCCESS in 28m 17s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2f7e6ba6b933479abd4af6c18e59141a : SUCCESS in 54m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6ec8c04acc7549eb9e9bd31829014147 : SUCCESS in 58m 40s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3c4943b5f91244d6a3ad9f225766bf69 : SUCCESS in 1h 39m 11s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8ee4eaf5bbbe49cc8c245ea9fa519cb0 : FAILURE in 6m 11s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/482a11bbd6f747b3890a5af276207a88 : SUCCESS in 2h 05m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16ba8c3f3728692334f9db3bbbc6fb86800a7b29","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-13 06:24:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fcc074553304ecf053132b2d274179b73fb391e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 08:03:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7566ab52089b4f2cabfe7170646d8602\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a889fa6c001349f1b79e001eb3e818a2 : SUCCESS in 1h 36m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/85ebdd02c0434f54b5d820eaa862ee16 : SUCCESS in 1h 37m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f810f3c69dbd47688f63f0c801a23b3e : SUCCESS in 2h 27m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40acdea67d9a480ba2eb4b03a4043d78 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73cdaa6e97334fa38f34627c3ffda4ae : SUCCESS in 6m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/19de4431dacc4f918f3aea95e707b0ca : SUCCESS in 1h 29m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/36c09d4713f34b85a06ac2c3362a211b : SUCCESS in 1h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7aeed7ccfcaa1d2b0fd53b574d34a9c142d58f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-13 08:34:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_25277\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"}],"_revision_number":2},{"id":"ac828dc6d6114a35a8d02999fd474b100d327ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 11:37:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e33833dffcb84331b14d5496ea5d6db8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/747e1c55bb004b62a819fdb6ec9f9fe0 : SUCCESS in 32m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8e88288a692842d78951e44c21105baa : SUCCESS in 1h 08m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c95eaf1a63754f87a5696665e0a3e932 : FAILURE in 21m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/30a5b8f372b84ad480094dac8536087c : SUCCESS in 30m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/41a5935de72e4d0eb0e9d42edc81cdae : SUCCESS in 57m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/410847024b854374be408ca17460bf6e : FAILURE in 1h 26m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fef76fba7204850b7efabf3e9978d34 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4694de55036244b0b7bcba1b37c0bf10 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33dce33cbf154b0aa75059683a79e9ab : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2646905a1c5f4e4ab6d534bdef9c6ff2 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083300ded01f4a3fa7e455cc3691e188 : SUCCESS in 6m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eacbd9792ca5464daece48b79d8f9c6c : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/70d731cde03c43229e16979cbc8ac589 : SUCCESS in 23m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/968e816c095248579b91054a5852f31b : SUCCESS in 1h 08m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2d3340b89304da1b98b7ad13c21ca98 : SUCCESS in 1h 06m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c2a9645673ff43fb867fcd845756aa40 : SUCCESS in 2h 18m 19s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/82872b07dea24ed6b9e6cbc31b19ffe9 : FAILURE in 1h 34m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d6d0dce75e554a2f863082834eae2168 : FAILURE in 1h 25m 23s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e0a186d288474154bba1c1bd84977230 : FAILURE in 56m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9d86ccff468f481fa887c6490dbd3c5f : SUCCESS in 25m 33s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bba4aec4f0a74d2dbf7ceb5ddcd42c40 : SUCCESS in 53m 11s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c4692a275f544a9f9eba5964b36bddde : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9dbbcd2271734acb937f1787b731e3aa : SUCCESS in 1h 34m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/96089083a31a484780b319c7e7ec7848 : FAILURE in 23m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1a913e8d5f21410b85e99d0d43ee1905 : SUCCESS in 20m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9df7c827525b471c8850f74cd4074a17 : SUCCESS in 26m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/866d284be33549258c8526d39dcd8703 : SUCCESS in 57m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b75b260f4c19404fbe4a2cc6c79fed7f : SUCCESS in 1h 45m 30s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/24a80284da924368b511667a05358a2f : SUCCESS in 1h 45m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ef9f9279ee5c42daba7653422542b47b : SUCCESS in 2h 22m 34s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/541cf0d9b23842eba8a378f8bbb0456e : FAILURE in 52m 13s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6a132161effa4761a8d6752c0d8a7606 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/225d829eb044438794b273817e27c86f : SUCCESS in 57m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/61b42b643c7244379e4e4bb1f6a21a33 : POST_FAILURE in 1h 42m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/28f632ed1497497c85a4adca1b4f7946 : SUCCESS in 1h 36m 54s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/899383ed118f41d4bee481f17f754c15 : SUCCESS in 27m 49s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/96fe5fce2afc40afa9721c5538cb5a4d : SUCCESS in 1h 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68196a255ad71557c48ff0048832100b28325a17","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-13 11:56:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d0ea6a7a0392139afebf6e5eda2aa5d8b163c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 14:49:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/135677059bb4462eb0a2e071c233c5eb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/479358b4f963455e8948719bd6211bf7 : SUCCESS in 12m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9168ed8023734d28b2979705d1eb38a3 : SUCCESS in 32m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/978d03a16ad346169a01b6de9dd83de7 : SUCCESS in 17m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/35376e5668294455b3dc20dc9b9d8e42 : SUCCESS in 16m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5eec8f73f5f64be0b9af4b34f107b2a6 : SUCCESS in 30m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a316487a6d6d4c7386bc853cc73dc4db : FAILURE in 2h 45m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/742aef64716f4009be045d657acd6f2e : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a1197c5ced04883979c4902adc0adbd : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9a87f79e22147daa98727ec5f6fb110 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87fc9c5f85164d90969e786a9e321584 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f214ac7a7754f94bc1767c22a43cbe2 : SUCCESS in 7m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/69c1db9a95b14540b4c42cf1b0f7ef94 : SUCCESS in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7164918185024ecea3ff0afb3423ee21 : SUCCESS in 19m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/174fb8df2d7242a49e6253d1eaa915fd : SUCCESS in 1h 19m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0b42428eb43f480ba2c0c148df3ae4d8 : SUCCESS in 56m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4829624bcbb146b1a2154147f134a930 : SUCCESS in 1h 37m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/24a4abc1a2f547d0bcc23275336ca6e4 : FAILURE in 1h 48m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/28c9083da9a74aa684b92d36764a4895 : FAILURE in 1h 28m 21s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/515735079bae474fbaea3c43b5f6ee9f : FAILURE in 1h 01m 20s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4a9b07656bf64464b16785ac07d207f8 : SUCCESS in 26m 25s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/dd4b667578b6459c97eac09384f5428e : SUCCESS in 1h 11m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/53927ecbf95f491e82ff8c4b4611d723 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/78e437acba2e4306b90865323185c047 : SUCCESS in 1h 36m 30s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fe08cd5889dc43279aab760317d8b558 : SUCCESS in 27m 19s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2cd52f1bd4cc43d7b9ac6ced4862f72c : SUCCESS in 24m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e5bd880dd1564babb3c6fa961e59136d : SUCCESS in 34m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5010778c3dfc497e85de951b8992cb65 : SUCCESS in 1h 06m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/08c60a965136464ab02edd40f2152ef7 : SUCCESS in 2h 21m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4f8b9c7dc7e84b6098927ec796893208 : SUCCESS in 1h 53m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/252a3e66c8014b068d6a9bc68c695e6b : FAILURE in 1h 24m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/22c686be7538400d8efcdcaecbe412e4 : FAILURE in 1h 04m 16s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e44b7e2e862d47b2bbfa84f957df59c9 : SUCCESS in 23m 19s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/67b433563ba443e6961871227afb5d76 : SUCCESS in 51m 17s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2472d6b5e70d408e9df86e73a8bd253e : FAILURE in 1h 14m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/578074f8c65b4a62a962635ef57cbdd1 : SUCCESS in 1h 41m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ce0107733ee14f0598cf11be24cdc2ed : SUCCESS in 30m 38s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/3b298ac29ed8474a86cd7d48f1f4d384 : SUCCESS in 2h 03m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0ba4e1d76de2f69e6632eb3e6da202fd124133a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 20:17:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61af5a290cf408d9ffdef4917bf4389\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/726d8b112e7c499fb6905c355c3c790f : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ad04a1f1087434cb04ee94c312bbfb5 : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51282e37c1cd401fac5c014b5c9064ac : SUCCESS in 2h 15m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9770bd39c70546e2b462e6a1d23e0c47 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a2fc386e2e447b281a245b5fd68cdd6 : SUCCESS in 5m 59s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6db1908a6ac845f9beb72cd40e625dce : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ca10295ba73475d93b44559ce2c2137 : SUCCESS in 1h 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"763558f863ac2354608d6df2e67dae65a09c56c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 05:00:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb8a894c175b08060e1983cf5aac6425f94bad68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 07:11:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d10b0a8a724b479969aa319759d2a9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0e7d9175136a4ceda7bb2b80e76415c2 : SUCCESS in 1h 07m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2cb1f07e211942efbb6a73f057579298 : SUCCESS in 1h 07m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/84f66cfea0fb4ddea25caab615eb1a4d : SUCCESS in 2h 08m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9f2b67daf784ddea8cb666f32eb07b1 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1fa9e41e00464f0daf464a999bbc0fd6 : SUCCESS in 6m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0a46f2cb81f247ed86e410c084d986e2 : SUCCESS in 1h 07m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/109f5675a4e04ec29ca7e9d847fa1695 : SUCCESS in 1h 30m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89a80f928105039b070ce28c52c796c92918ec58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 08:01:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db524588eea3463eaa568fc1a0067a4b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3f5f4f5fea1143f6bb260d1a3ae674b3 : SUCCESS in 32m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f58f3f2bd2a84deaadccec7b364c9ddb : SUCCESS in 1h 13m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2416f8dca14e4fddb03fc23d41c47e26 : SUCCESS in 18m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4f9099d2206e41fdacf9ba0a08c723ca : SUCCESS in 23m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/357f64af5ad24294878f36afb087c570 : SUCCESS in 1h 10m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8ff2909c5e5045dd89e8cf698c29bcdc : FAILURE in 1h 48m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0df2287443e47a3b31e2247f166b336 : FAILURE in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0bca85dfeee4e3f85414074831b480b : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d60c05a1f50a4c09b5da16c52dc6287c : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bd1c6f08cf54ee68996600a843e57e1 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b5b9a1a01844bcb31e34977978eb5e : SUCCESS in 6m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/995a152039fc441c8b60d41808f1a3c8 : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e27addf9399b4969a97f8cc0c0ab8b1b : SUCCESS in 18m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1d48c9c4804c4438b7902d3a0321057a : SUCCESS in 1h 16m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b0a74018a8154797826e80d1feac19fa : SUCCESS in 1h 10m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/006953fafcb24d248af50bf560c6d4bf : SUCCESS in 1h 40m 49s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/272538e2ba0c4a558d7719a33f5f1202 : SUCCESS in 2h 11m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5f8c8d957a7444489bd29610bef9f988 : FAILURE in 1h 20m 19s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c44865a21e394701b10aa486de8ade6b : FAILURE in 1h 04m 13s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/620e23e268564c21b82da4b88c615333 : SUCCESS in 28m 01s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8025e5c352834a828acd49c9b9a12859 : SUCCESS in 57m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0fcb04aae6a044b19574d18b11127515 : SUCCESS in 1h 15m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7a7cb69eaf304c7d942ed924a63103a7 : SUCCESS in 1h 34m 38s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9379fc5cefdb473eb2003d00d8b361e4 : FAILURE in 26m 03s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/22f1838a78354932a07b768a6c94e8d3 : SUCCESS in 23m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d527e24856d74b72b031086f3bd45c6e : SUCCESS in 13m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e1f8872cfb6a4227847dad23325a5cd9 : SUCCESS in 58m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/001b7c1f041d4632874d508b609e94ba : SUCCESS in 1h 48m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f6235c04bfce40c5a04c5c61a1257fe2 : SUCCESS in 2h 03m 24s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5feb3fe8080484892006abf42aa7e30 : FAILURE in 2h 50m 56s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c5d57b5c6fab4c2eb2a2cec34c198f84 : FAILURE in 1h 06m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/1e5b75a5a1104509a398c5bfa5ce483e : SUCCESS in 24m 37s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/adab3f54e66046ffaa88ff89ca6860b8 : SUCCESS in 57m 50s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2c13f343edc4432e8f9ad3705df7788a : FAILURE in 55m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bb83a87621fd469698b9d6e1fc1ea434 : SUCCESS in 1h 27m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c925febece7a484abeae3ff6a2946d39 : SUCCESS in 14m 16s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/8e4c206ca41a45a1bd3976f2c2632092 : SUCCESS in 2h 08m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01bcfdd07c0369e4c7d3e7aed89f9b1938169eab","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2025-08-14 08:25:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39f162e43de24e14a34fb744627639142edc0d3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:43:54.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9028735f414343af2c6f9fe89ed9db50ea389153","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 13:23:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e02158e4f48462eaabe94b9dc58af7b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/de39db78acab4a1c90a42cd08e5620b9 : SUCCESS in 1h 00m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a8a9c1c633fe4884a778cfe6348a583f : SUCCESS in 54m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff98b24dd04847b19d443b0879b40feb : SUCCESS in 2h 08m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31e7729d43c941499bdf2320ccc00b5e : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1eb23f02fc1a45c7b50729f70da54b22 : SUCCESS in 7m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/065aa6552af746d3a6ca6a773addbe44 : SUCCESS in 56m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03f1d67d3d9f443093a54aaef895ddd0 : SUCCESS in 1h 06m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ac9f74740bd3b247050edebb7879cca2dfdd7a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 13:46:03.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f6a9249720167c0f3f05ea6c2017e777d2cf6056","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 17:35:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7099245584de4aaa861d9f7489d54082\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dc1561aac980431db5aa89bf89952732 : SUCCESS in 1h 18m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/987ba10fd8164a1eb2e398f65877a5fd : SUCCESS in 1h 21m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77700701343644aba1c7c159f0c13dd9 : SUCCESS in 2h 17m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/487dad3a06cd47fba1c040715deeb969 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61cd1af8f19e42c9ae584173c9f54775 : SUCCESS in 6m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/806ac538b88544cdb086c9afb590e66d : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/38eb07ec58b541e19650504e4e778e7d : SUCCESS in 1h 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"314fb662b96bc6c2bbec52f8e009e7b6cddc7fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-15 07:00:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"852346cc96d4014f33892ccd4a221cb2fd0a82ca","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2025-08-15 07:58:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4f554ca549ef9956b641e85102d1f259808f47a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 09:20:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c84c7f74083144cda90ccd47e51da58e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fa69c4d4436945deb194c2dca8f3c315 : SUCCESS in 18m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66fdec087aab450c8e0ca3c2f20ad477 : SUCCESS in 1h 16m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/050fef2b8ead4de2b8263f43b1b7fa06 : SUCCESS in 21m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bc472a471b744cf2840d53ce2453fb91 : SUCCESS in 26m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e0b370eb6c104f0ca43c1bd356bdf7b0 : SUCCESS in 51m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8dcca3ee82d24ef4a3d4e4ef3a9cf47a : FAILURE in 1h 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06193c67cb534a429d473bc9386c9a84 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/413d391100ec48f391444c71aa3fab7c : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c568e8566551440fb8cec4b12d2739e6 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aea74e51c35946c79ec6799844bb93c8 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8229b082beca4c0c95d7416c3c4768c1 : SUCCESS in 8m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b7fcc39a542249f08ae65f1564a6df80 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/83ec9f2a671e46ea8987341184c4f019 : SUCCESS in 20m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b99e4bfdb754480b587204453b3b437 : SUCCESS in 1h 11m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2f2d16bfd85b44aa91b09b6f855a1169 : SUCCESS in 1h 08m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/401b4efcdd7e46a8884058783478fc65 : SUCCESS in 1h 48m 42s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c9c4f0ea1b294f24a94440e3c4a9f0e0 : SUCCESS in 2h 09m 16s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e113d3265b104b13b7018f03c95021ba : FAILURE in 1h 00m 03s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1a85e6ec468b4c7e9be85ea613bedebd : FAILURE in 54m 36s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/6c5bff2e5e5c4b4485e016c19524851e : SUCCESS in 25m 29s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/800e7ca2c53b42eb86f381d25c21dd81 : SUCCESS in 35m 21s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/50d852eb589e4b75ab60d9af9b3912cb : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/557906e52e284c7c8a1b28cadbe5ef6d : FAILURE in 55m 04s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8283a69f3e3c4e08a2b593796877f1b9 : FAILURE in 28m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4361ec5493da4f71a819d3d67e63f5d8 : SUCCESS in 22m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/daca722c0bd44f73836f1d506c5f0896 : SUCCESS in 29m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b88256c5e9f04bcda38adeafba7db05a : SUCCESS in 1h 11m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/efe729dfe9b240e0b244ad8df25b8514 : SUCCESS in 1h 21m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fa734f7ce60642ee8f432d5197cffb01 : SUCCESS in 1h 15m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bebdf2d862624605bd7e425fa1582e68 : FAILURE in 1h 22m 15s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/05dde5824c6644939cb6b3f38223cd88 : FAILURE in 1h 05m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/911f814d6e654acaadaa34fc4f753b7a : SUCCESS in 15m 37s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/574cff1102c348278b6d66c3f424b035 : SUCCESS in 27m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/01065bddf0f94885baa645453fe97592 : FAILURE in 39m 59s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d594d82bb58a41c1b94d66b2cdcb61ea : FAILURE in 58m 16s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/293345733ec0403faa0ca8eda8938227 : FAILURE in 14m 40s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/a2eb13f62f3a4b5899bc73ef5a349be6 : SUCCESS in 1h 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2799c07f8ecdecf32532b908570072e7a1a49c19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 12:21:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da1af49a9244d8e8a25b8c3309891c1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/96e9baedf61243cc85777c67a98d7811 : SUCCESS in 1h 35m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e70a3f178f2943698af075855941cbf2 : SUCCESS in 1h 39m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59b6841ade9d4bf5b2a1eeca210995b2 : SUCCESS in 2h 24m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acaac7af29e64a3f8b0caa796e888941 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7460cf7cb9a642b99a91b54969f709a9 : SUCCESS in 6m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/88985a3f738b4443b9968e3f96d488fc : SUCCESS in 1h 26m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cbd9a72681e74187817617fb3daf26c9 : SUCCESS in 1h 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"97b8a98f332e14be2980a3c37fab65cab4af768c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-20 06:58:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4abfb283ef1ad05f98864313c4434bc8b2b3e5fe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-20 06:58:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b71692c175f911bf6033e465982b1d0e024e9eda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 09:41:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e1ebbea977b45b3ba2237c07c0adfa3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/91dcc056d7b94de0b57dc1f3084358d6 : SUCCESS in 43m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/13802f9173884b608c80c10ca79aa54d : SUCCESS in 1h 12m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42b0cba0fde0493dad0a62226790056d : SUCCESS in 2h 18m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae2568bc4cbd4a3c8b4ec64630bb21cf : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4077d3f5614442d9b1139977ead3832 : SUCCESS in 5m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8c67d44ca4f2498cb3749219bbfd7ad8 : SUCCESS in 1h 01m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/93e852a028ff4dccb12a697a2aff51ec : SUCCESS in 1h 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c0005570730b4b0a849f5c8a24b1480100ef9bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 09:41:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/729fa821372948d18fe6433eba1f7c5f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/99b352bd9e8e456dafafa94945369cb2 : SUCCESS in 12m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/258bf0dc51ee400180db3a9a32da41d5 : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/198834e679594b3298c42dce0c1b034d : SUCCESS in 18m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e419736d2d63441aa03e09c7ad9176b8 : SUCCESS in 31m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/85e7059b63df40e0b5969ac0f0ce0122 : SUCCESS in 1h 12m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0aa29c1953494d06ba49e22cf81653a2 : FAILURE in 2h 02m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d10d0902169f4e54a51065e82fa9c582 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdac4664e0664e22989dbd94dda2193e : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10e4e409b4634306b64b818cfb550d86 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44e87b7bd57f420986418cf391b5e5d8 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44205d5e5f9c4299ba79dc67d6f8576a : SUCCESS in 6m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1635f07d6ffb440d807dcb2cb7f90f51 : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/03cc88f3766c4130b06e7875c22f57d0 : SUCCESS in 21m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9436727bee9542e28674ba0f4a207fea : SUCCESS in 1h 18m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/910772e1be344872a4524533fce3cc2a : SUCCESS in 1h 06m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/40da9d7b89c14d759a8594fa254869f2 : SUCCESS in 1h 41m 17s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0a3a101bfe8a49079e75afb3c68c9d3f : SUCCESS in 2h 15m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4a269562be74439db11b2fb191835b6c : FAILURE in 1h 23m 47s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/87b6be38482b40aba535d0ecdc650168 : FAILURE in 52m 09s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0aa922a8ef414ad285605cc7b0be4f6b : SUCCESS in 30m 50s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ce52bf9eacd14f8990ec3d30a2253e0e : SUCCESS in 1h 19m 25s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/36b320394a9c4eabb74449d1b564ebfd : SUCCESS in 1h 15m 37s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7d0d1cf7642646a8be989d4d476ad607 : FAILURE in 1h 29m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3adec534eab04b64b76d3582c774ac31 : FAILURE in 23m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b14f0df581724f35992c8768ff1b0b08 : SUCCESS in 22m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/02bd0413bc2741f68eee0c6d2203eec5 : SUCCESS in 20m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d47730b493e0408183c899ea2b1b2d39 : SUCCESS in 1h 04m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b55781d14c474ba6b34859583f4f1a0d : SUCCESS in 1h 41m 35s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4b29d6cc82bc4d90be2152d2365c8a32 : SUCCESS in 1h 50m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b3e3ef5fa4ac4807aff3a97c460302f5 : FAILURE in 2h 22m 27s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8917ea9ea737478b8c0d9132b248a6bc : FAILURE in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/01abcba887ef4b668a18b08bfcd7a547 : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/aa6d804947624a9993b291328606acb2 : SUCCESS in 27m 49s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/412de24b3499449b8168d9eac930b0ef : FAILURE in 1h 24m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ff5968befcf74ea6af3ac96f31978599 : FAILURE in 1h 39m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b820d214203140aaaec470ce15fc228a : FAILURE in 35m 25s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/e884ebc0280241f1bfbd70c597ba2572 : SUCCESS in 1h 55m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e0956dbb06229fd85df8385c1afe015b142c48b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-27 15:21:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0f7dc602728843f63129c8813c08e11f501adb22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 06:33:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f419a066431c1cbfbd3acd2a6e1512061ccaced9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 06:34:15.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"c1b2c4214dda1de35178c9f99ece8069ded179bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:51:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21efd1c5d4c14547815052ab19c9de0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f0da2f0e3fed41738fdd370977f58394 : SUCCESS in 16m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9627aad4beeb440f94a152d8c268e291 : SUCCESS in 1h 20m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/292808b10bbd4402815de1a605b5327c : SUCCESS in 21m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/97978df1964140afba0adb9bb3c81699 : SUCCESS in 16m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e6cad7e59717485abe2c1dcaa539d338 : SUCCESS in 1h 09m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b148d8af140d4ed3862a580f2d00c6ce : FAILURE in 1h 38m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8a5bbf48fb4df1812fda145228e696 : SUCCESS in 16m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eae8109ec004834b1e869511a04a3c0 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09191d2ba83249bb922688a0fcaded91 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4691e011172420892fbedc482558839 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc47141f5e5d47239dc92ef6e6e3bd44 : SUCCESS in 6m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c568e57a836a439f9ee4c0f27872a0ac : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8f0b6be16b0e403b8d6221307c9d5251 : SUCCESS in 24m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/11ccac571a984acdaaef018b5dce21f1 : SUCCESS in 1h 18m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6bd02560965d4b439a6b11513db50d20 : SUCCESS in 1h 23m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e8064b578956497abfb1bdf9ed66bfc8 : SUCCESS in 1h 58m 10s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/3dc942e9b8d7488a818461e01aa6b98a : FAILURE in 2h 01m 32s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/97ec69769a174df7987aa8c467d15059 : FAILURE in 1h 15m 04s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ec39cdbdf6db469db7acaa9688cd14eb : FAILURE in 49m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/323853046db54c938a45c77debb0e29d : SUCCESS in 26m 37s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/602edbf5d10642e4b8d2ba738b6ed74d : SUCCESS in 57m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e312818d0b4b4e59a762c269b38793fe : SUCCESS in 1h 24m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/be919d6ff94242cfa51a87fb421fbab5 : SUCCESS in 1h 57m 19s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f4ecdf9af1a04530bd56eb0a8de635e4 : FAILURE in 45m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2aadf15fdfc64e43b794bbe765ede783 : SUCCESS in 16m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cbcc05fc06b44beeafd1dbe87248d84f : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/27873667a46f43a2aca53f2f93538a18 : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/844bc444cc574f0c8d0cabf413760e85 : SUCCESS in 1h 36m 34s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d255d645e699438ea20223ad20d7af1f : SUCCESS in 1h 52m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5357ab421f9e45fc9a4f78c1a6cee66e : TIMED_OUT in 3h 03m 16s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/19e32a6125fd4b459a939cd3a4422d67 : FAILURE in 29m 35s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9a8e2aadf3904b49a7975c0ae3380e0b : SUCCESS in 25m 19s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9139bd68fda2417ca0eb40b61d6bd4d4 : SUCCESS in 1h 19m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1c800f1972864288b1f9b3513bf64681 : FAILURE in 1h 24m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7edb5e63d626435193cf0a19725aace7 : SUCCESS in 58m 16s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/096286d949d74b3bba7b8e914e285113 : SUCCESS in 34m 12s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/157a4c3daa1f4ce39c980307d4576776 : SUCCESS in 1h 19m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eed2e80ff39e51b9870e239da5d35c483272c43b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:53:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629dd7e10a824e0594ddfcccd3e460f4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5e883198bebd4208a6a0b910c8fae5c9 : FAILURE in 30m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3041c33ec7f54875beadae20878a9d4c : SUCCESS in 55m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be243816f0994fb89da1d44f8b7f2fc9 : SUCCESS in 2h 13m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/489f8543800b41f2a7f34ae54f08f82a : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ad00cb58ecb4c68910d44ebc12a2a11 : SUCCESS in 5m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5435cc67c1b94283ba7d8432b7824368 : FAILURE in 52m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cb936ab5012d479cbaf44d53280fc735 : SUCCESS in 1h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"81b0578e6e6126e18702f2512dd91db0f04670c7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 10:59:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a4521af1d52a52703ae4c23405570984c7e0491e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-29 11:00:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"37f7571d0e38d06b1fc6d9f6b7e506d190c239ad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-29 11:14:04.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"66e6bfc5932eefd13485a451785eb01e83e2528e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 13:14:46.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bcd97ea0ea9475f9cd401391b10474f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a0e10abc8e6d4a34bae087ee9e3f4e19 : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f17f488f79e043529c1e82081016405d : SUCCESS in 56m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c52fc729e82a43128167a628fe4dee55 : SUCCESS in 20m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/56a5aa0e0d6948eba74960844993ecd5 : SUCCESS in 30m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bff316f9140047e0a41b35b63b560aef : SUCCESS in 49m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/26d4abefaa9643c18fb9a979de3fcd29 : SUCCESS in 1h 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36aed804730b42a1984f8ae1d41c5951 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/741703f2d2ef47fcbce2e5bc9cbf3d7a : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2a6ae97927a47b9a6618b97ccd595df : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61f2b299eb0c4a929176ff317c7b2581 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69df7ea5beed4885a30e741ad7f3211b : SUCCESS in 6m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f495d2eddf424776b2b06013b82be39b : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e2af21876e547bf8d2a31693aa3b3bf : SUCCESS in 20m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c4bca8280eab49ebafc17b7313f188b0 : SUCCESS in 1h 13m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3ab73a97653841f0be317dc3c3b37605 : SUCCESS in 1h 06m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d77644d00da948d1bafdce6b86281284 : SUCCESS in 1h 42m 27s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2d55a5c2b3704abab716e351fd03df26 : SUCCESS in 2h 05m 38s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b2ccab80e4c64d878d53f315187c409e : FAILURE in 1h 29m 30s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/bc13e63db3cf43258c07b81256743d5e : FAILURE in 39m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/22858aaded6e40cd82ce3ef1500c9b6e : SUCCESS in 20m 33s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/63f88c42e7304059aa751eefa54428e1 : SUCCESS in 42m 38s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0e48567f70ec43af96c4c52ec6ef9346 : SUCCESS in 1h 13m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/dafce89a83e34734b93b3c9471f0fc7f : SUCCESS in 1h 37m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b88c942f421344149008a9e610a82d12 : FAILURE in 28m 46s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b6e6e840959d41d2b1e88313c7945726 : SUCCESS in 22m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/431efe5d3e96492a8b523e690cbd0180 : SUCCESS in 25m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cf1e53da27984b57a9312270202c486b : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d8b79ed06da843feb560b8425cd2ae76 : SUCCESS in 1h 23m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f152d3d283fe422e8824850e3d0b9801 : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ff9d3557997248308a6eeaee3c2c85b7 : SUCCESS in 1h 24m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/236910276cfa43a4b54ad6dc02a84169 : FAILURE in 29m 58s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/49946100231541b88f81f322b7907b48 : FAILURE in 19m 50s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4db0c569e0f54d10bbe7e1b32d1e9a1f : SUCCESS in 51m 15s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f26182fa338a4dfd886ccc3627d0dd06 : FAILURE in 49m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2355259fb52f4ff9b78b1f05b41aa98c : SUCCESS in 1h 22m 15s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/38520acef2a745569c8e304ba29b0b9e : SUCCESS in 31m 56s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/ffcd221c4221408da446ee59047edbb1 : SUCCESS in 1h 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a584c979d52a206ce542b6053595163e7535e454","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 17:50:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870950bc22b64d3d87990d4e06ccc4dc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f58001beaba547839f3f638a94ccd481 : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3360129fee214cb99d4921de82be5b4f : SUCCESS in 46m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77b01146b9a4400ead059676738eb778 : FAILURE in 1h 57m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a207b185d38e49ab88292238d69d8b39 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c0152c70a784785b1d63aee9361c065 : SUCCESS in 5m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/674e7ce0cddd4ef0b80b8cc96eb8c86f : SUCCESS in 36m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b66ea5d8b6a4a1abddc095eaeaf123a : SUCCESS in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"692f476b25ad7d3567e3673506dc18b51dc3d7ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-03 13:25:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f7973679bebba6e53a0c77b8dff01889ded8f57f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:43:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51ccbbdf25b49d98859784d7612745d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/86174a19f3f043d28f1f722f5fa9c073 : SUCCESS in 23m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9e1ef610f19f4d808fe4bdda70fa276a : SUCCESS in 58m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dc6024756fe448439c1496f2da1f5b93 : SUCCESS in 18m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fa1826c43c6d459b8c8b9013365bab32 : SUCCESS in 24m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9969ae0cbdfd4cd4a0169ad4984a8cfa : SUCCESS in 53m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3aa51ef4ad4d4c8aa7bca949fbedfcb2 : SUCCESS in 1h 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4133993754124c81aa126828f162e580 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a1e37684d274fc381e5195af4659730 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc9a86673119471280d5ff66255c9f13 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6777c9303bdb46a780da5d5f53942f75 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a985428503493e9cd691bbcd13a90b : SUCCESS in 6m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3f48f4634e44407db6a444405937ce10 : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b676696affc04a06a9a4fdf7606d3b89 : SUCCESS in 20m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3ee833dcb0b34f26895b2848b147b717 : SUCCESS in 1h 11m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a68326975b304a7daf87a8581224ec93 : SUCCESS in 1h 05m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/eef709258ec34ef9b1dbe876f1ffec6e : SUCCESS in 1h 23m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8b953b614afa4f68904e62bb8ebbf332 : SUCCESS in 2h 04m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d3c11a6357ca4502a188fb78fa855ccc : FAILURE in 1h 32m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e81f4b23800c4168b578aed2b79a46dc : FAILURE in 1h 00m 49s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/525f744be0124ebe82e5c0e6f2e41eec : SUCCESS in 25m 10s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/64e711dafeb64f7a9a1ab67bc9418c3e : SUCCESS in 29m 43s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/be25ffbc2ba74cca92ad661f35cf769e : SUCCESS in 1h 04m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/635556cc529d4b4699d5d7608eaf2523 : SUCCESS in 1h 52m 46s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/556fa3de8206464eaa2f66be72527c01 : SUCCESS in 33m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dffb4568b2b34d838f354fa520382fd6 : SUCCESS in 24m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e72af332864f4d03a5edc4251f800491 : SUCCESS in 25m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d1109b27bde34cf9b676b45464d084f7 : SUCCESS in 54m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/64d6a13a75f04c20b501ffcd44ce04af : SUCCESS in 1h 10m 07s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ab2a650291cd4f38a5a1e3cdc154b9a7 : SUCCESS in 1h 41m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/50f9f6f6792a4e97b64e02f4931aec58 : SUCCESS in 2h 16m 23s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5bf1194771a94c5aaf82c004f1091a90 : FAILURE in 1h 02m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f5ad5e0c21974ae5b850097536910960 : SUCCESS in 26m 43s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b829598daccb4b1bbe28a6c6359230cc : SUCCESS in 58m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/95731a1d8320432da94f3c58548f3903 : SUCCESS in 1h 12m 28s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6a5639af10244e12953c894cf8c754dc : SUCCESS in 1h 24m 22s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/72eb8455ce864f8887f83972fae6c0cd : SUCCESS in 27m 05s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/89ce9524f3d244a89127ff63426abb7b : SUCCESS in 1h 54m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cc6e44795b90ecd0ab344e56da89c716be7f3651","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-03 16:14:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d845767e93dc74c7a18751301038e79e0d2afddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 18:29:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afc7bb63f8c9442aaa0595f071fb3b4f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4da1527ffd6c491a8d3e623c812a9af4 : SUCCESS in 22m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8e7de534d6864f5a9d8be63840eb0fba : SUCCESS in 1h 07m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fcc79aaa18664c06b5e76bfe8b8d3bda : SUCCESS in 14m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/debe8842333a4a1f9ee7cd0f4f580716 : SUCCESS in 26m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/622e71ad2a294747a57b4def1f2c16da : SUCCESS in 50m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/51cac9ac52b3488c825d22139fdc3751 : SUCCESS in 1h 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c85cfea4e14ca0aff30099ed5f0465 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d7bbd3e209347668d4a58ba3e078bf9 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3254e6a6b7264bcfb83da6453ab2063e : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16bd6d9f98114745bdb9216f7d789e44 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b0699efebd4bbdabd7d4782a68ea04 : SUCCESS in 7m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2c0bda0503ac47b6ab002651904f787f : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3724cda7cb6f44b4ba1ed72127f044af : SUCCESS in 24m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7b601303badf45ad8962f9229013dc69 : SUCCESS in 1h 20m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/89632c8b31794da98071c28541f6f211 : SUCCESS in 1h 02m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dc111c04bfaa4d4bb71503f1c534f05d : SUCCESS in 1h 39m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f96fae5414eb4e05a9d9876b9e508f90 : SUCCESS in 2h 10m 26s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2e9d77cc849247e9a1ca868d1a2a4a5e : FAILURE in 1h 35m 12s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ab1c496aa00c412d95268edfaa9ea560 : FAILURE in 58m 36s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bc8194865f5a49d4a49a3e01b9d36779 : SUCCESS in 25m 16s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/80e4498cdd7c4fa186033c41b07e75f8 : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f95f1f5d1cee450baab0337aad192a8f : SUCCESS in 1h 05m 04s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/671bbe1d9a434282a4301ea0cebff454 : SUCCESS in 1h 40m 05s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea41a802bbb44a9d9c00857dd86eb246 : FAILURE in 29m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8f440ffebe594966add4c7debe403ab7 : FAILURE in 22m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b07e9a5fd9ee4507aef9d3e09a08ccc4 : SUCCESS in 24m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f0f5dbcf9bbd4d7eae2bafb7819918fe : SUCCESS in 48m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b346387db1a24ae19492f0d0f9222dd2 : SUCCESS in 1h 11m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a782c693870049eeaea93c57288c087e : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/550b9462f2bb495fbe36bf89e2e65c16 : SUCCESS in 2h 03m 12s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ee662a3543d64de8a25ac564c208abff : FAILURE in 1h 02m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/da25c6103df64fc9b9d9c3aa21943660 : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a1b4b08af808438e8e0e5ec779d0bdb7 : SUCCESS in 56m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2d9c210af22b4d5b8e46011f65f768f8 : SUCCESS in 55m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bc6ae435260248b3a54f58a410f8a422 : FAILURE in 1h 04m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f447697c493740d6b5a444d10d7de5c3 : SUCCESS in 25m 20s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/44c9c0d259924a1c8fa4a0bfd86715f0 : SUCCESS in 2h 08m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b65554c30f6b94300a6737cb2d62ea986f3e81b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-03 18:57:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"007327f9fa431a0664501d9f858ec21fda62f26b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 19:08:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a62a19377a344631a61f192ce503d7d8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a877139f9acb4ee0b6ab200f96a4af3b : FAILURE in 56m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/068bb7a0d80b4fc2b386c9fc7be20d0b : SUCCESS in 39m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/10897aa19ffd42278aa0266bb2517410 : SUCCESS in 1h 38m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5bfa5658bec4ff3b188b0304110abc7 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d301255b45f54662a9383c941c4d96a5 : SUCCESS in 5m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cee4646a582542c48834f893e20d604b : SUCCESS in 53m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9db08eb0fb01414f837d5f1641803a8a : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9c253f21efd3232acfcb19402c2688e0477edeed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 21:07:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd9c26fdc9694f98a23c2ecae830ffa5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aef43b11d3944ef39e9d3a9539dc56d3 : SUCCESS in 27m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1452633058e34567904af59ff1f0a599 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/812c00f9ceb946fcb96249052b107b67 : SUCCESS in 17m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1163fd031f49454bba6666df76838116 : SUCCESS in 14m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3035fe0218414fc3b5513463f4500d10 : SUCCESS in 46m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bcb9e4a38cfb45e8ad1f1683ebe3b09b : SUCCESS in 1h 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1120ffd25a4505a77d11183ae3358c : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db73180db47444f684328ccbd99b5d7d : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cd5da338cf94f449c22be1d4f9a7060 : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d4c2132369a487abdf4464e931ee145 : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1a35965ca74676a1717c0c2b489a17 : SUCCESS in 6m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/320a97e74837480995dc2e1e20ae0fc6 : SUCCESS in 1m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8c2f10c3e17540c3a1cf49f3ed1c9ec6 : SUCCESS in 21m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/06c4d277eba6441ba75e4de9ff3accc0 : SUCCESS in 1h 21m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2df4395c9c74ff68e34fe36003dcb4e : SUCCESS in 1h 09m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ddd46d84ee56498e824cb9285b1804ec : SUCCESS in 1h 31m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f9fa717dfea842a4b55f0fe4bca48e75 : SUCCESS in 2h 04m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e9acbaae9b504d9db62db2ab397fbc23 : FAILURE in 1h 30m 43s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2eeba1feee3247eeab2430d1dde558df : FAILURE in 32m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2fb9855e30f64717bf45a84c76c94243 : SUCCESS in 26m 57s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/69eead94ff844cdca5dce33c21df3f5e : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/415063328c39464d902b77c220255330 : SUCCESS in 1h 18m 26s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3e3c599795f645ff94a701bd04cae51c : SUCCESS in 1h 24m 50s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/55dc7e53b16e49faafd4f81b25d78488 : FAILURE in 27m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3c4f9f3e2f4b408baea4b47e641d1853 : FAILURE in 16m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/64e18c4482d74debb6f3f10892c786a0 : SUCCESS in 26m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e82969f30e4145f1a40eb439046c2b59 : SUCCESS in 48m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1e937b846b3047bfa8a860e71a4a452f : SUCCESS in 1h 12m 04s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/321fa8f39d374693a98a61be801cb7b5 : SUCCESS in 1h 33m 11s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/89d4fa9770944aacbc3af83ab646bfda : SUCCESS in 2h 03m 11s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/da5674df73a648b68f963859035cf0dd : FAILURE in 15m 50s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/98ba7291c16e4affa518b3f73200475b : SUCCESS in 22m 56s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5bd525cf36ee4a9d8f5b02c236cd7865 : FAILURE in 15m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b166e11232f445e381a409b970d94e39 : SUCCESS in 51m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a7077360d1a344e9a0a1f161ed9bb3ab : FAILURE in 1h 00m 17s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2fcce08605ad460da9d5c383747e1de8 : FAILURE in 31m 46s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/a9bb7d3fd56a42869aa7e6f713dd9895 : FAILURE in 1h 41m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"548c38c2c6fa662b42718a2d2829010bb9aa3eb6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-04 06:57:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"39a8dbbc987604867594398c3b761b1cfcf27cbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:34:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6565e5e50a054a9c991b6cee1bf64127\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9083d4ae668d429b835686eecb780ead : SUCCESS in 48m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/27ab0ce055a64de58ab72c3f102be217 : SUCCESS in 46m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a3794655157a4b5da67e616bee964223 : SUCCESS in 1h 34m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/239f39612a1c4da4a4199dc6aa29a092 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ce1657b01f64ce781168861e8a03479 : SUCCESS in 5m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/42002c1b0ea04c229d9d50157d5591e1 : SUCCESS in 45m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2e558f4e94864d2ca72b18faa1760fd9 : SUCCESS in 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cd3c81dccda3a6195db221121583624872c248ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:59:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a0f32a8972a469b81af81f21d6ccf39\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2ae010ba6b144ed1af4c6baa15fdf1b2 : SUCCESS in 24m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ea09566006de45b1a71dd282bdc2ca32 : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b13fcdfca4334b3ba0383476dd0847a1 : SUCCESS in 18m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/64eb4288a2ec438785b682781b45d806 : SUCCESS in 28m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/612fa45745c14bc780dc014aed967505 : SUCCESS in 48m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/be2069f7b38f491aa86e370aeedcc1dd : SUCCESS in 1h 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3fdee8c60cc4e3383fe642a1974c738 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6991616a28f48b0a576410c8163d03a : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e73f1357c4a046c88069bb8a3c0e3be6 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d568f385c93b4714bbc2e3352d2f7d9f : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e986b0f220814c5d81bd89631d52b648 : SUCCESS in 4m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/667b078eb22845e1a370cb71a442abd5 : SUCCESS in 1m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ac440d0ce51e404e804bab02bb11ec94 : SUCCESS in 17m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a23f8085684d470b82a7128f4491a542 : SUCCESS in 57m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/59006f92609840f3b797966cbe3d7b66 : SUCCESS in 52m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a1674e3bf6db484b87c9db5bf6f9420a : SUCCESS in 1h 41m 26s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bb06cca67b4747d8b52e6c305e526306 : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5a168c4f3cbe446d95530e14c80a1535 : FAILURE in 1h 41m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f9620653f3df47dcabc450a9faa86a1d : FAILURE in 41m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3cd56d591df54e62a3fcef6efd4cb137 : SUCCESS in 26m 16s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/262f6a2f62954dca9966f9feb67613a3 : SUCCESS in 57m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8d9cf9631f8849f8aaaca010ce8f42a6 : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ef1e159da80f47b2b84b8fc87cffbe97 : SUCCESS in 1h 39m 20s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fe32445786804689917eba25d9ce0b70 : SUCCESS in 25m 16s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/19538758da4e43669e76c640ca5d86e2 : SUCCESS in 21m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f60cfb5ca91a46e58d82faecbbd1755c : SUCCESS in 31m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/10ad00fd51964df58c9763bb1f0e942e : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/22f730610a514260a184db42af46e861 : SUCCESS in 1h 13m 32s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/27f72452bc8144e98a5a8d9e3465f5e0 : SUCCESS in 1h 31m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/34f8bcf5903b4495a707f94d8c41f755 : SUCCESS in 2h 00m 47s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/da031a702c7d4d3cb513b7c0a606efc2 : FAILURE in 1h 01m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/abaa7cc28c1646308bcb92716225727b : SUCCESS in 27m 31s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/803b8eaa81304f3cbbb48fdfec6e190e : SUCCESS in 58m 40s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/59181a7aa04548fc9b6fec80b32c42d8 : SUCCESS in 50m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6af8557e651a4ee1a9f6f48c65b758db : FAILURE in 1h 00m 11s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9114b49e8f954ab8b27165f232586302 : SUCCESS in 23m 43s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/e4d5affb66d04e85ae746fa1f9f24ad7 : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"73382a38bf9931fe6a998d4792e1b741f2f2bf5a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-09-04 09:00:03.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"7eb896e578f32d4485f337229715c4e1264088a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:00:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"822e4e4fe953dc188c3f2ae62bfe4e1f2b62ab09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 13:00:42.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45f9f48c9f8c45acaecdf629d0e53b72\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e6106d2466f142049e5871a250befc2e : SUCCESS in 15m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5afd1c9373ac45bcb9a99157753d0c22 : SUCCESS in 18m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d73de6688bd54686a4a7cd5b5b13b9a7 : SUCCESS in 54m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/91503f2bc4fb4df1afbf9f9884bffddc : SUCCESS in 1h 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/034f6f256cd043989faa2afed63f9cec : SUCCESS in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/167de9483c4a477f96e81431dce8b3a8 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69190b054a264d0696850ab78a5ae566 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c480e7de1934f64aa84726b0a2b97ec : SUCCESS in 3m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/337f38bcc30e4109987c290d77ce7c01 : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8be139423cef484da51a293862ec2b29 : SUCCESS in 19m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ce07020375e3488b91267cfee7225043 : SUCCESS in 42m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/168557fb9a5a4397b161a022e0f023a7 : SUCCESS in 1h 10m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a473cc9b61bd4c4dbc7f2cb8657c5bb8 : SUCCESS in 1h 24m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d6053a709b24be19b9994354e09e5ff : SUCCESS in 11m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/37dbf026731840dcb5486a1e6862c53f : SUCCESS in 36m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9ca127736c9f4a05a99105842f099839 : SUCCESS in 47m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f4c038375c14439aac98b8b39c9fc32e : SUCCESS in 1h 19m 32s","accounts_in_message":[],"_revision_number":7},{"id":"265d5d46a1750f770de125272cb8a1ceee4d0cde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 13:00:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"af5ffceb8f0e3b5e8d12f52bd3766d4f4811ec6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 13:02:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63427be0f0064d7f8f0d2111f564bfe5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d04e95cc84247fc87dbf27dc9b90580 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9db2666c3974c3a8261a218f57ff7a8f09e2c891","revisions":{"d3e06cf5974c62ea1384bf1f2d088c7e062948c9":{"kind":"REWORK","_number":1,"created":"2025-08-05 08:16:08.000000000","uploader":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"ref":"refs/changes/38/956538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/1"}}},"commit":{"parents":[{"commit":"9398dc43bab48a85f86bbec7d93c28d5bf622a95","subject":"Merge \"change erlang arm64 repo to rhel9 based\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9398dc43bab48a85f86bbec7d93c28d5bf622a95"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:13:40.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e06cf5974c62ea1384bf1f2d088c7e062948c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e06cf5974c62ea1384bf1f2d088c7e062948c9"}]},"branch":"refs/heads/stable/2025.1"},"f9aff97728c53c63c87976293f3f4d445f03c3d6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-13 08:34:36.000000000","uploader":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/956538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/2"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-13 08:34:36.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9aff97728c53c63c87976293f3f4d445f03c3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9aff97728c53c63c87976293f3f4d445f03c3d6"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"036930e7fe50a55a5ef89c36e1308e180d2c052b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-14 08:43:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/956538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/3"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 08:43:54.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/957319\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/036930e7fe50a55a5ef89c36e1308e180d2c052b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/036930e7fe50a55a5ef89c36e1308e180d2c052b"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"abe57c65fa848271ce8610c88337d2d90c71d9e6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-08-14 13:46:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/956538/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/4"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 13:46:03.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/957320\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abe57c65fa848271ce8610c88337d2d90c71d9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abe57c65fa848271ce8610c88337d2d90c71d9e6"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"6bda5048a1da0e71ee6e61d7a3b68a38bfdfe236":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-15 07:00:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/956538/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/5"}}},"commit":{"parents":[{"commit":"08248b170c5cb2771319cc546387020e7ff7dd4c","subject":"[stable-only] CI: Fix pep8 ansible-compat version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08248b170c5cb2771319cc546387020e7ff7dd4c"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-15 07:00:51.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/957320\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bda5048a1da0e71ee6e61d7a3b68a38bfdfe236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bda5048a1da0e71ee6e61d7a3b68a38bfdfe236"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"2a48d50de8506bcc588c2eff10c9313259c17abc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-08-27 15:21:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/956538/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/6"}}},"commit":{"parents":[{"commit":"6c08b17423de465e0a6daf578576053c9eb9f384","subject":"Merge \"docs: Update contributor guide\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c08b17423de465e0a6daf578576053c9eb9f384"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-27 15:21:52.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/957320\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a48d50de8506bcc588c2eff10c9313259c17abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a48d50de8506bcc588c2eff10c9313259c17abc"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"9db2666c3974c3a8261a218f57ff7a8f09e2c891":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-08-29 06:34:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/956538/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/956538/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/956538/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/956538/7"}}},"commit":{"parents":[{"commit":"6c08b17423de465e0a6daf578576053c9eb9f384","subject":"Merge \"docs: Update contributor guide\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c08b17423de465e0a6daf578576053c9eb9f384"}]}],"author":{"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","date":"2025-08-05 08:09:49.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-29 06:34:06.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: This change was reverted on all the other branches.\n\nChange-Id: I49015b8fc72c83449767bfce0737b0e44ed75808\nSigned-off-by: Mariusz Karpiarz \u003cmariusz.karpiarz@codethink.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9db2666c3974c3a8261a218f57ff7a8f09e2c891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9db2666c3974c3a8261a218f57ff7a8f09e2c891"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
