)]}'
{"id":"openstack%2Fkolla~956641","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I1ba004df309667460aa152928f789aaf6be6051f","project":"openstack/kolla","branch":"stable/2025.1","topic":"mariadb-fix-stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-07 11:50:18.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-07 10:07:43.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I1ba004df309667460aa152928f789aaf6be6051f","subject":"debian: Use Epoxy extrepo","status":"MERGED","created":"2025-08-06 08:58:05.000000000","updated":"2025-08-07 11:51:21.000000000","submitted":"2025-08-07 11:50:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956641-mariadb-fix-stable/2025.1","cherry_pick_of_change":949752,"cherry_pick_of_patch_set":8,"meta_rev_id":"74b8653b8bca6e82c468cd00d78168dbf03477a7","_number":956641,"virtual_id_number":956641,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 11:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-07 10:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-06 09:34:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-07 10:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-07 10:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 09:34:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-06 10:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-07 07:16:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8ec6d71df7c88e373f6ca6e150664e39953e3b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 08:58:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b5f45c171cfb02d99e7d92089a97eddfb775dac7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 09:34:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"677d1e539399cd049174b732ab10038a6c88fc1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 10:49:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d4c66159b014f7eb6f3876b83532c46\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2e6429b30d8e4f989f625c0163f5fe0f : SUCCESS in 23m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/12e7c40640c84a8e883287c1448c7424 : FAILURE in 12m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dc2768a97acb48379b256cdfd1835d2a : SUCCESS in 19m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f847ef0e7c3e4f3694da10c44e38f85e : SUCCESS in 14m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/46c2efe053184b34b48572543b2046d1 : SUCCESS in 1h 03m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fcb201f1f9a547e0a5381947851f39df : SUCCESS in 1h 48m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d04ab35aa0548e78e32e72a3fc4f3d6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa27361c20014f88ae84c106d76b9f1f : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5b15e02bc994701af4945bd2dc022fd : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdfcc8ad6b0348c997542951176b7a18 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7decb97653b6422684679f5743b87978 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7215f4d6e3db418a8cf12af4ffeebac3 : SUCCESS in 3m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/66e6643cfe5c4d4298668de435da8423 : SUCCESS in 1m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7a525817d1f1469f846a9e67c2245994 : SUCCESS in 20m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c910997075d44f21a10dc25596becd35 : SUCCESS in 1h 11m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/db3e017c2f544407b1e32a5b0588d6d6 : FAILURE in 8m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7590c40acb82455f8d0eee801993fc22 : FAILURE in 12m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/05c123b57756492b927cc409ce958ce2 : FAILURE in 8m 51s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/43f5e7263572478da336baf5fd311095 : FAILURE in 13m 47s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/bd5ec99053ff4c9798567b691aea5e57 : FAILURE in 16m 08s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/fc607bf127184406abb873c0bed020d9 : FAILURE in 7m 43s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0f39d76e97b7463e9df685d610d5198c : FAILURE in 8m 47s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6751fbfb1ac446d18426fde9602e3c92 : FAILURE in 12m 45s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ef99ef7baf724ee1a72dc7eba8e6e51b : FAILURE in 18m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ca37f9eac46e472fa62f8ffb33c5f38d : FAILURE in 12m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba3fbd912f2d4e20a8a2a4170253476c : SUCCESS in 20m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/771d333b8db541d5a7367c77d0649e8c : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59197746d45b4e15aae276b117d6b580 : FAILURE in 19m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e482abf804044f4998247990ab8b214f : FAILURE in 18m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9520b0e2409340d9bced427a21f51579 : FAILURE in 16m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/381e368af40a448dafac176304b46443 : FAILURE in 18m 17s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d3520bde5d3a46cea2d9d5535d1f018f : FAILURE in 18m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a59968c118a0408fb5b2337997e10e37 : FAILURE in 12m 13s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/009c0e3c5c274d3d92d276006561ae4f : FAILURE in 13m 56s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4663b553855345ce972ff3d90acfbdb8 : FAILURE in 18m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/02b310a2a1fd48b2b5ae1f1844408b8c : FAILURE in 17m 40s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/be6689e4ef9a44b0a6e45f7a34bdb844 : FAILURE in 10m 47s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/f1cca49d66a546bc882169f732485a82 : FAILURE in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccee0b7cf702975e979ce2798d6d79722865fa0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 12:41:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df476f7a66e1494d803ff18eec3e50a8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/56823ee0c83740e9aa0185b5d405c860 : FAILURE in 1h 10m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fc45887e4ecf4911b9c871ddc1ed1cb1 : SUCCESS in 1h 35m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/48aa570454c948f48b42f0424baf852a : SUCCESS in 1h 50m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f63cbb63f0dd4dcfb8c3ef48ea3a2180 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94a296f32fcf4238a029d818fd02ea1a : SUCCESS in 5m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d2b3ffc698ec43999e0eb071d0766ede : FAILURE in 1h 10m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6c5ed23b312946faa7279b2e50915c22 : SUCCESS in 1h 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"212ae9c07115be4c8037ee2543b8c8bd702c09a8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 07:12:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b461f7d0b14f5d80a8f7ae96c7fd834edeff2288","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:16:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d56b88fe3844c1895b4c75fc481bc89\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/44874af289d34f0d8fca43eb60ae9fbb : FAILURE in 1h 02m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/209f7638e3fa489eb49be971a516156f : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0fb7ae5727e84d3f91114b0fd925e29b : SUCCESS in 2h 01m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c25395b048b84657a07389789c153227 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a71eee9b8651494cb00e7081edd07275 : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5470c327d57d4566a72a2a5ed66141c3 : FAILURE in 1h 03m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8602e5d832d84053bb2d477e6e9d6c5d : SUCCESS in 1h 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8543ade27757457445d0af6b6022b8c52ede1a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:38:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f941e4500c64cc7acaddf742d40bb4c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4afbdd3f6d5c435bb67b9ccaeead0294 : SUCCESS in 24m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/85b2103df97145a1937d6bc23775fef9 : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f984d4e882c9472791a23c9c2cea5e21 : SUCCESS in 9m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e89e9f727d034561ae0bd1adc0d88ccc : SUCCESS in 14m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff1bf9a23bbf456bad1c591cf0eef8bd : SUCCESS in 30m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/865da92953c64fb8a052543c75895407 : SUCCESS in 1h 31m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6cd6fa0ea1743548535455bbed74e52 : SUCCESS in 1m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/543d76c2719a444d9e24776cc60935ca : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac0a9774916544d79bdbbe7cbe0804cc : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f8b3cc63ddf49e2ad7af3a7fb58894b : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4184b77b6d4e9584e7a9dc909d1b35 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7aa361079944398770d209fe9f1713 : SUCCESS in 3m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1821d4992b774c8aa1dd2766ffeebceb : SUCCESS in 3m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1043a814c68746589e5999a17ed55dc7 : SUCCESS in 18m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3ba2788d1dfe4848b5005a8773e080a0 : SUCCESS in 1h 14m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a3ab16ebefb648e2922b186394758777 : SUCCESS in 1h 07m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c37066f323de4a84ac392b7963761a5d : SUCCESS in 2h 10m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/23ba7a559f024098afe45afdde25fdaf : FAILURE in 1h 44m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/43dc145b1e5c40b5aef2e854e3886f78 : FAILURE in 54m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6eee80ec2fc24a7fae2afb29515a3873 : FAILURE in 1h 14m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bd43eabc62774a25a63370ba22acd5bc : SUCCESS in 25m 52s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/794e800450ff4dfba8fa9c5c79a3e8fb : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2b0f07283baa4daeac9133e1f1a0a6df : SUCCESS in 1h 48m 14s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1062f1ee519344579f65ef5823e7ea32 : SUCCESS in 53m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0df258f2b62e4236bffe662e9a9b3da0 : SUCCESS in 30m 57s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/136972d15cc54123978aaf6cc4c864c0 : SUCCESS in 26m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1ffab52ed0fa4bafa35b5ef3dad45b23 : SUCCESS in 31m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ea670703c1a748119cf1c58c2a8714bc : SUCCESS in 1h 01m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4a620a12af9746608cf2813a28468c39 : SUCCESS in 1h 47m 43s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6f40ac8d92bf4072b55c600eb85ab3d9 : FAILURE in 1h 16m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a62ce1f6c57549ab80b8232831b10b5b : FAILURE in 1h 31m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a0d4851f885542a89c2783d659cf25fd : FAILURE in 54m 47s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3b7f1a200b5643e8aa5c5ccc48ffcb32 : SUCCESS in 24m 52s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e4b3a20d5fea482dabd67d36094d49a8 : SUCCESS in 57m 42s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4eac4f10d0d74003a630c78a7d21a4e9 : FAILURE in 1h 36m 06s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5a6a1f66743b4c5a817aeb025c5643c2 : SUCCESS in 1h 36m 19s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c25b09dd973e4a3893c03b799e7d8f78 : FAILURE in 23m 18s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/a85106242d194699b0405a6400f1515e : SUCCESS in 2h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eba2e47aa9c538467359fbb4e08ae4f16181ea65","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 10:07:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb4f4dc36d706750ece7ce82c81e3719f472edd1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-07 10:07:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20922cc3ec4e4f1a98ef3df9d0924e368cd4680a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 10:07:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4d24edef4a298002d380166f5eaef9f08d259f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:50:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894022f5adf74a3da0af0403b1e2b8f4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/43085eefbf89424a811775168693a954 : SUCCESS in 12m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/93a797b5d771480b91ebec2abf2e36c8 : SUCCESS in 16m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bf2ee763a93844ff82829bdd66cf492e : SUCCESS in 1h 03m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e0e5c6ab61a444d68b2ea9a3af485435 : SUCCESS in 1h 36m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15197a71de23438ba5a65884fd6ee84a : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03a05a14c8614986804a598e3ef4ac9d : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a715e6aa60904892ab19fcdaf6fc1e67 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb729d2d270442c86e916043579eec4 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fea6dd195554b4ebdb1a427ad34e3f0 : SUCCESS in 3m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6dade8c0423346828aef0742b2a773f2 : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/831af0bfaf00485b8e087c534b3ab889 : SUCCESS in 19m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/227877e6b59f4c86a1e4fc7301544ae6 : SUCCESS in 1h 31m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/365c4d6a412d48bcb06337a645969f06 : SUCCESS in 1h 07m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a182e81dabb9426cbf767edd31d30562 : SUCCESS in 1h 37m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/77dd3eaa8bfd4f90a88485c553804d78 : SUCCESS in 24m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ac110c2b424b4e019ae0ddcd9107dd16 : SUCCESS in 34m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c14f526c968f404586a9770f23ed817f : SUCCESS in 1h 18m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3b5eccfd6f00420bb6d130a0f10d70a2 : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":1},{"id":"43dbf2270b0f4efab2ae9e9a219c201893951dac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:50:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"74b8653b8bca6e82c468cd00d78168dbf03477a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:51:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/701b15a3681c4b7aa92f2c805696c631\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f515cc1ec2946738736fc63a8e6ce61 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ed7e8a1bead45f78d89e7f527216cbc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02be21d47b98a045a45d8d7daf0a7bf7104e462f","revisions":{"02be21d47b98a045a45d8d7daf0a7bf7104e462f":{"kind":"REWORK","_number":1,"created":"2025-08-06 08:58:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/956641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/956641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/956641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/956641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/956641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/956641/1"}}},"commit":{"parents":[{"commit":"9398dc43bab48a85f86bbec7d93c28d5bf622a95","subject":"Merge \"change erlang arm64 repo to rhel9 based\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9398dc43bab48a85f86bbec7d93c28d5bf622a95"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:00:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 08:58:05.000000000","tz":0},"subject":"debian: Use Epoxy extrepo","message":"debian: Use Epoxy extrepo\n\nChange-Id: I1ba004df309667460aa152928f789aaf6be6051f\n(cherry picked from commit ede71c02c4d38fe0b2513829d6d6ebf889119f1e)\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02be21d47b98a045a45d8d7daf0a7bf7104e462f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02be21d47b98a045a45d8d7daf0a7bf7104e462f"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
