)]}'
{"id":"openstack%2Fkolla~956648","triplet_id":"openstack%2Fkolla~master~I95500612168871454ce269b3562a8daccdf4de4f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-07 09:31:13.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-06 19:54:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-07 07:26:04.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I95500612168871454ce269b3562a8daccdf4de4f","subject":"ironic: Stop copying rootwrap since Ironic dropped it","status":"MERGED","created":"2025-08-06 09:57:31.000000000","updated":"2025-08-07 09:32:06.000000000","submitted":"2025-08-07 09:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956648","meta_rev_id":"a3f1662da72f96d481a164a0bab258e21d29d117","_number":956648,"virtual_id_number":956648,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 09:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-07 09:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-06 19:54:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-07 07:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-08-07 07:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-08-07 07:50:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 10:02:50.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-08-06 10:02:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-08-06 11:18:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-06 12:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-07 09:11:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"674a0b8baaa7a64982f99cf57ace88d52456d9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 09:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34f0920e5dc9c4805a8ea5246b8cc9533ecf874e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 10:02:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"82e0ba58860e233a6bc2624f5f5dc374bf9f9df3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 10:02:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd8ee497b648010ef6c477c94a32737e137942e8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 10:04:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b5ad83ec074ae26af134effd1446f9ce15ea2b00","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 11:18:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"98f80b2ee33ab7d51f16e7c5ae280c376ba181e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 12:14:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee957f2d35064fa8bf6d55d5b2cfc9f3\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/0d8ea79688674c73bc39db008b01bc6b : FAILURE in 15m 11s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/9bbcdcce70bb4ce3b56d1935900b65eb : SUCCESS in 43m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c58dcf1a45224e2d96b1ec49ad7ed190 : FAILURE in 12m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c7b1f11f0bcd42e0a37939e5a1c1e9c7 : FAILURE in 15m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7b1227172f3d4943824c044a68554eae : SUCCESS in 36m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f1aeac2c53e7463cab677dc93e72ee91 : SUCCESS in 55m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11256480377d489eab73f0bbd0963df4 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e4823756ba34379994827c68aba20cb : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f5e24e520ad46c8ba7dd4e342fa4f17 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92ce28093f674e91980d9a7b2ac244ce : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4492ede790244283b93aa11bcbed1800 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/021913a05cbb491691bbc9044ad4c2e3 : SUCCESS in 6m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/96e4a412c93e487f97219f88f50a8d8a : SUCCESS in 1m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/be234cdce546437ba3b03d820f6a96bf : FAILURE in 18m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4eb465e7284a4adbb0b5d35157ff65bc : FAILURE in 26m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/99fb7848238f4e239dd225db5de61099 : SUCCESS in 30m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/03672f688ed740e9b5be70ecbef80622 : SUCCESS in 1h 31m 05s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8e7fe1f3599240759a274535e004009d : FAILURE in 21m 44s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/57f6bf81a5e649bab17e6a3121d2e6d1 : SUCCESS in 2h 09m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60c53d8675ca467c545219f195b2f1a606c00350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 13:55:46.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf5fc9dedd794c87ffad5903810568794530a046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:01:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b62d2363e22f48649af44ec004560a85\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/aa8b8ad1f9814382a594a01e6aefd600 : SUCCESS in 14m 25s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/41316e6c11934f85893f7ab7c10c8adb : SUCCESS in 1h 16m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86bff25acedb4cc4898ac0b4d3c0becf : SUCCESS in 9m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b0c290732eab44aeb5c666f77fe65e94 : SUCCESS in 16m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a230dfe68b83442c829f2ce4cfabc743 : SUCCESS in 1h 02m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a959fb386cf74936b98876df956b4979 : SUCCESS in 1h 40m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e501ae745c43cb8eaa325ecd619d91 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4451c25bddd4d0a938449a3699a653f : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2db302f932764cbc9e8db4c853bf127d : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd9990f517804ae69977f8d82669bddd : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46c0db18a2e243ffbc148d057c841839 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df6c2166ba24c2ca8eade109fb0f4ad : SUCCESS in 4m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c3d24fad5c264d6dab364f7d37eba1db : SUCCESS in 1m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a99a24b6dc534856a597726f4e9c20aa : SUCCESS in 26m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c1736d1bf2bd43a687ef3f780e44c2d3 : SUCCESS in 14m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5afb95867f9e4c4fadc90ea30ae6da2c : SUCCESS in 1h 08m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0e94c701e0cc46088b7d699220d224f1 : SUCCESS in 49m 12s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/555e5e55527045fd8c579725cec04a43 : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7cc6c1c95c2d4d08823e51f9cf54d3cc : SUCCESS in 2h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a3a2b4f703d320a5521d9d67bf796a6bd00bcca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:46:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8ac7c0ddc6e4e328d649b14afe1a824\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/11210cd2197546d3a64ffd800c39c0c9 : SUCCESS in 1h 22m 46s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/70a8d128335e4b33b2a6c0a2cedc8301 : FAILURE in 2h 07m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce00914abc5345ec89653eb47a0a55e2 : SUCCESS in 1h 19m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90318e9c9f80437ab46ac14171a04879 : SUCCESS in 2h 31m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb623ab0c1514beb87e731bcdb2ef10a : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ece6e0be82641e39442a6a27fcb8657 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91827b1d4c0540ea953c92e70f82db49 : SUCCESS in 10m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d174f3befd43480bad5df8db4d757c67 : SUCCESS in 1h 26m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"586815b8ff29062bfd61070c0f541c9081b7ae72","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 19:54:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4d56ee9b5ca19966a55f592609e9678b0823eeb9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-07 07:26:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"64dc207c515a43c606b90a2f2a4710ccfdb9ee9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 07:50:00.000000000","message":"Patch Set 3: Backport-Candidate-2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff25b033db53f2f2dc2155189a1ff7786dd0c3dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 07:50:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c1d60e18455b8d1eef1c5791ab1c753a46dd7cb0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-07 09:11:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3e36a720cccda3e3dfdb2c20909c7ef201d49489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:31:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e577283e99814f2f82ebb3177c1e8351\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/02c429216368482d9e089479d02dff50 : SUCCESS in 21m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/360c18895129451eb8132eae2fa432bd : SUCCESS in 25m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5ce11c2d1ea3410e81de582d879a5f90 : SUCCESS in 1h 03m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4018012efa4db182ae0a3150235f0e : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c29eb3e23443399181bb1bca0f5ab1 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f2d0ae9e5d846faab39b9084a7ba067 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a006300f6c4bab8bdfb398264598fa : SUCCESS in 6m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a71a41e1a1a4c3aa5617d6192b5ece9 : SUCCESS in 3m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a372eff631ee4ff78298168bbe93b7d2 : SUCCESS in 12m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0af968874be649549bf7aae0f58285b4 : SUCCESS in 31m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5f41a7f984bc43ea8b8a2aaf5e72b50d : SUCCESS in 1h 11m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ec43019b5579458bb8fa65b67f942ccc : SUCCESS in 1h 33m 21s","accounts_in_message":[],"_revision_number":3},{"id":"5f359316a8903df3f8086861bc0125e0068ef8bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:31:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a3f1662da72f96d481a164a0bab258e21d29d117","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:32:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb224114616940e699535c11dcf53cfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63501eabcac4ba2b934ceb93ff53cd4 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63c4202d207b6b4686156bca81c84ca756513f3c","revisions":{"3f1b18259b4e23b43d04118fe582d673b76f2c4f":{"kind":"REWORK","_number":1,"created":"2025-08-06 09:57:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/956648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/956648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/956648/1"}}},"commit":{"parents":[{"commit":"f25cf12455e3180dd422581c96e2a73df5d1d666","subject":"update debian release in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f25cf12455e3180dd422581c96e2a73df5d1d666"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 09:56:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 09:57:28.000000000","tz":120},"subject":"ironic: Stop copying rootwrap since Ironic dropped it","message":"ironic: Stop copying rootwrap since Ironic dropped it\n\nSee I0a8e26e8990eae8108537541159f7810d35b70f1\n\nChange-Id: I95500612168871454ce269b3562a8daccdf4de4f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f1b18259b4e23b43d04118fe582d673b76f2c4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f1b18259b4e23b43d04118fe582d673b76f2c4f"}]},"branch":"refs/heads/master"},"7fdfd80c67683b6f22f6a01a46216a91ed20f5f4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-06 10:02:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/956648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/956648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/956648/2"}}},"commit":{"parents":[{"commit":"ddac7ca1ed393609b7bda8753db3598f1ae02648","subject":"send empty X-Registry-Auth for anonymous pushes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddac7ca1ed393609b7bda8753db3598f1ae02648"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 09:56:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 10:02:37.000000000","tz":0},"subject":"ironic: Stop copying rootwrap since Ironic dropped it","message":"ironic: Stop copying rootwrap since Ironic dropped it\n\nSee I0a8e26e8990eae8108537541159f7810d35b70f1\n\nChange-Id: I95500612168871454ce269b3562a8daccdf4de4f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fdfd80c67683b6f22f6a01a46216a91ed20f5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fdfd80c67683b6f22f6a01a46216a91ed20f5f4"}]},"branch":"refs/heads/master","description":"Rebase"},"63c4202d207b6b4686156bca81c84ca756513f3c":{"kind":"REWORK","_number":3,"created":"2025-08-06 13:55:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/956648/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/956648/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/956648/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/956648/3"}}},"commit":{"parents":[{"commit":"ddac7ca1ed393609b7bda8753db3598f1ae02648","subject":"send empty X-Registry-Auth for anonymous pushes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddac7ca1ed393609b7bda8753db3598f1ae02648"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 09:56:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:55:46.000000000","tz":0},"subject":"ironic: Stop copying rootwrap since Ironic dropped it","message":"ironic: Stop copying rootwrap since Ironic dropped it\n\nSee I0a8e26e8990eae8108537541159f7810d35b70f1\n\nChange-Id: I95500612168871454ce269b3562a8daccdf4de4f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
