)]}'
{"id":"openstack%2Fkolla~956670","triplet_id":"openstack%2Fkolla~master~Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-07 11:30:48.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-07 09:21:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-07 07:54:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-07 11:30:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475","subject":"keystone: Add local wsgi file","status":"MERGED","created":"2025-08-06 13:58:53.000000000","updated":"2025-08-07 13:57:00.000000000","submitted":"2025-08-07 11:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956670","meta_rev_id":"5c44e12ea1c073dff40cf36a9fb1e1ec6271abe0","_number":956670,"virtual_id_number":956670,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 11:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-07 07:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-07 07:54:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-07 07:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-08-07 07:50:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 14:45:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-06 16:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-07 07:28:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-08-07 09:13:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"4f13190e92ccfd760121c9d5e7cab29b12443c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 13:58:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd373fb19f299403be9aa9476fc2622d19996f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:00:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3654db1fe4d402bf718893aff90bf01681f05874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:03:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a45c4637bac40f0d3c623d12fcb7139ae0a79935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:21:28.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"60211d421647846e02e6bda6a805711eb3e58564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:21:59.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f232d79f02482373f63b984f927bad1905b01179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:24:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f7aa9c8e8477b020cb8a5c860f10050870fd2d27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 14:45:01.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7731890c2755815abc30b3586da4a171457e7490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:17:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6288ec875011498f8fa0ecc8661d3736\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7f97c9a3c7f64fcdb2448dce49115935 : SUCCESS in 16m 04s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/d1ef5f93bd984843867ada8baafe16e0 : SUCCESS in 1h 14m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/901a3dbbf5274b5181a3f9c389235134 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d25f5af0e09b403385cbe92c06c4de94 : SUCCESS in 31m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f881514aa3734c79805f0ad1b2ea8c65 : SUCCESS in 34m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/43598618a4ca477396e3e125fa3e03fb : FAILURE in 1h 50m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f41ac9482ae4c14a1eb427553ffb575 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6de9e269dd24ff4ae7c92b2bcfb3853 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca5006c1e0d445daab1205fc513b4d1f : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6773778fa554d609cd796f5750297cd : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/111654eff7774d43b427e3d62e9379b7 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5412ddf221a1443b81919294b9761c3d : SUCCESS in 7m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/239c63b6d1074abdac6cc7798d49eabb : SUCCESS in 3m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8958aca363a14ec8bf1d89f7f7eedc60 : SUCCESS in 21m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/42bfa7061e9642c699bc04a3b98e1175 : SUCCESS in 25m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ed6fe0c590884c2ca6274f8e1104c062 : SUCCESS in 1h 03m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/004355db8d1841c0b8435ada860fedf0 : SUCCESS in 1h 39m 03s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/0bce38a80268404685d15d875ba7daf5 : FAILURE in 1h 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1c850268c55fdc9c47d90a0fcbf6a15c727db873","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 22:54:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ce8dce99654ae6bb4fac1228fcfb96\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/569c1d629fc14a5481494e24a37f0552 : SUCCESS in 1h 05m 31s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e020b5ba607d44eb9af7454a32bb73c2 : SUCCESS in 2h 06m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce7d3a392cf34335ba8f7f3ec464ad24 : SUCCESS in 54m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/895a91929c6a4f979a7d5a16ae208332 : SUCCESS in 2h 13m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70bae64e798b47a7bc8f501eb7642462 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4922f7c427ad44eb9a2189acd385ccb6 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e63c2d03d9d0436ea5612b12e06b05df : SUCCESS in 9m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f0fcdd4bf1674b1aa9d0da964750ab0f : SUCCESS in 1h 05m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1713fea35ce9edf1ec94326b93f730f7b1a9d0e4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-07 07:28:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3c5b3d7c6ce06917603a2b22c91b0b77d9de7b84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 07:50:14.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"448f10e8c029d8f0c31df819086db868775880da","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 07:50:36.000000000","message":"Patch Set 7: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"82bb59e9d579ac892fab9923886d27425f6df55b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-07 07:54:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"4ddd236fb16a8d4eec1baeb06fc1dcf72c9ed970","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-07 09:13:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"974626d35e8c740769119c4bcd4f573d1c46c120","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 09:21:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5b7c9dc49094403d155c8352c5fe1edfc51569f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:46:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/924a07c3ad8d4824983d13d2100220ae\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/249f6dd8168f428c883e610bc9daab95 : SUCCESS in 14m 44s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/10611b9b87d34d13a90841fcb10701a9 : SUCCESS in 35m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/00e3c9507233408e93d256ff3370525e : SUCCESS in 18m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/825e0373cd6f455f9ece47c9b4b2d2e7 : SUCCESS in 16m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8a8a71548c54f159d3e7e81a7909274 : SUCCESS in 32m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dabcbe2ad7ab46efa45011f8f68b0098 : SUCCESS in 48m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d425c356d419474987dc646e1c03239c : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b484c5dbba4457b9a72004c9752ded3 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4dfd514b63ac442f882ce5d6da8676cb : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91fbab312d2c4aeba0889ea671c5e4b5 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb263b389791490c97e6cad47022c77e : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a20db0260ed428f840a387860dff93c : SUCCESS in 5m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dd8adfe409064d6ebef08c4dcf20cc9d : SUCCESS in 2m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f08fa67b92cf44a4ae58c41ae0040faa : SUCCESS in 12m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8f64df2f21a04592a6668797b7d741cd : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6ef1c4a3d8824508a3815d6c70a20a53 : SUCCESS in 32m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/058b661f3d4546abbfed5bf80757011c : SUCCESS in 1h 26m 55s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/28ae66dc57a146b78633fece81cf06fe : SUCCESS in 1h 51m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1c3f34c1f1e8baa87d39b2292fa6faa22aef44a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 09:46:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fb4b04a21741fb401c2569f95112ebefc1bff798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:30:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3f2c8d2010f43e69ad6c2ef0f58533c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/986931bab9f34229829ca02176249887 : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/90c63d18a70d49a187d687865a182164 : SUCCESS in 31m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3f271a74d514436fa73740c5a3aafb3b : SUCCESS in 1h 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee0620f109d4c7cb230fe0d6eb982f1 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e31da28e9b694fd18375651a266e5792 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3840897e5c7b487aa63551b7659a1437 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d99f35bef84ab998cc1cb90693297b : SUCCESS in 3m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/047ee651db8c441382bc045113e98bfa : SUCCESS in 2m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/adb5b5371b104ab1ad17f554142c1fb1 : SUCCESS in 11m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/912a305f7e604d36ad14da8e20680b3a : SUCCESS in 31m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4589319e79f64b2cb75d658ab31a0590 : SUCCESS in 1h 01m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2ad6d05ac7934bf3ab47e5a8726fe8d1 : SUCCESS in 1h 40m 02s","accounts_in_message":[],"_revision_number":7},{"id":"f5dcb6d503f752ab1a0090fbcad5aee7a627bd99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:30:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b8566ceb6a0f1f2a4e51d1341bfbd27c12779b59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:31:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2c7563ed920448995c3f7420700f6c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6460fca37e417bb767c669f3470dda : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7},{"id":"5c44e12ea1c073dff40cf36a9fb1e1ec6271abe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 13:57:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5414dca3a59f4f7287766814f9c8ef13\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/adf6df2af8a14d16a848b791e53f6d1d : SUCCESS in 1h 01m 07s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5345cd3d5817484c90d23ba605de59a1 : TIMED_OUT in 2h 32m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/edbfdb7e2d744b1e96bc47251514ac7c : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f763b21b01c4ceba0d34931e43b5d98 : SUCCESS in 2h 12m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7269451701884c20aa09067204c6a460 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8fdf814b2bc4a3a89b39cd35367d495 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad1db39c8fb34dcfad48411217683feb : SUCCESS in 7m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b47ef92a7a4e4a9ebfa22a94f2f9a807 : SUCCESS in 1h 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bbc842720c0b56f79cf28e7156dd13ff8deb537f","revisions":{"b0b7cc82a925427c1e9b5f22f4138b0416c0f324":{"kind":"REWORK","_number":1,"created":"2025-08-06 13:58:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/1"}}},"commit":{"parents":[{"commit":"f25cf12455e3180dd422581c96e2a73df5d1d666","subject":"update debian release in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f25cf12455e3180dd422581c96e2a73df5d1d666"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:58:51.000000000","tz":120},"subject":"keystone: Drop copying of wsgi files","message":"keystone: Drop copying of wsgi files\n\nThey have been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0b7cc82a925427c1e9b5f22f4138b0416c0f324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0b7cc82a925427c1e9b5f22f4138b0416c0f324"}]},"branch":"refs/heads/master"},"cf79807e19bf2c5eadb9cddea2439b14b4d29f1f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-06 14:00:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/2"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:00:17.000000000","tz":0},"subject":"keystone: Drop copying of wsgi files","message":"keystone: Drop copying of wsgi files\n\nThey have been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf79807e19bf2c5eadb9cddea2439b14b4d29f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf79807e19bf2c5eadb9cddea2439b14b4d29f1f"}]},"branch":"refs/heads/master","description":"Rebase"},"16c43fc2e183a37d4f8b18e1c452e4d350332b1e":{"kind":"REWORK","_number":3,"created":"2025-08-06 14:03:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/3"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:03:04.000000000","tz":0},"subject":"keystone: Drop copying of wsgi files","message":"keystone: Drop copying of wsgi files\n\nThey have been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16c43fc2e183a37d4f8b18e1c452e4d350332b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16c43fc2e183a37d4f8b18e1c452e4d350332b1e"}]},"branch":"refs/heads/master"},"53566804bc2e12349299377723cc927a3c5648c2":{"kind":"REWORK","_number":4,"created":"2025-08-06 14:21:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/4"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:21:28.000000000","tz":0},"subject":"keystone: Add local wsgi file","message":"keystone: Add local wsgi file\n\nIt has been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53566804bc2e12349299377723cc927a3c5648c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53566804bc2e12349299377723cc927a3c5648c2"}]},"branch":"refs/heads/master"},"f3c0a5da8686f0c9b8ac86a2c84dad12c71cdadb":{"kind":"REWORK","_number":5,"created":"2025-08-06 14:21:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/5"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:21:59.000000000","tz":0},"subject":"keystone: Add local wsgi file","message":"keystone: Add local wsgi file\n\nIt has been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3c0a5da8686f0c9b8ac86a2c84dad12c71cdadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3c0a5da8686f0c9b8ac86a2c84dad12c71cdadb"}]},"branch":"refs/heads/master"},"37a05bcd20504fb8f22f4429a4a04266301ab05b":{"kind":"REWORK","_number":6,"created":"2025-08-06 14:24:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/6"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:24:41.000000000","tz":0},"subject":"keystone: Add local wsgi file","message":"keystone: Add local wsgi file\n\nIt has been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37a05bcd20504fb8f22f4429a4a04266301ab05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37a05bcd20504fb8f22f4429a4a04266301ab05b"}]},"branch":"refs/heads/master"},"bbc842720c0b56f79cf28e7156dd13ff8deb537f":{"kind":"REWORK","_number":7,"created":"2025-08-07 07:50:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/956670/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/956670/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/956670/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/956670/7"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:57:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 07:50:14.000000000","tz":0},"subject":"keystone: Add local wsgi file","message":"keystone: Add local wsgi file\n\nIt has been removed with I9fa7cd60bf27768db06f69681675fce2e55712af\n\nChange-Id: Ib42be3d56bdb64247ac6f17b8ac1f1baa873f475\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbc842720c0b56f79cf28e7156dd13ff8deb537f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbc842720c0b56f79cf28e7156dd13ff8deb537f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
