)]}'
{"id":"openstack%2Fkolla~956691","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I7a2f3fce223afd74741b40bf62836b325fca5b19","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-06 19:22:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a2f3fce223afd74741b40bf62836b325fca5b19","subject":"send empty X-Registry-Auth for anonymous pushes","status":"MERGED","created":"2025-08-06 15:14:32.000000000","updated":"2025-08-07 00:17:27.000000000","submitted":"2025-08-06 19:22:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956691","cherry_pick_of_change":956456,"cherry_pick_of_patch_set":17,"meta_rev_id":"e8e74fdc63af98cd54b358fa98c59b15525c6df1","_number":956691,"virtual_id_number":956691,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 19:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-06 15:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-06 15:17:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-06 16:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-06 16:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 15:14:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-08-06 15:17:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-06 16:57:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-08-06 17:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c917a2cd1d66f974db5cf3143a8df75224302946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 15:14:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e421e88a5e6c39603e479811e024457eebab27a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 15:15:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3445063e32300de8db266c4ddbd96f2512e55918","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 15:17:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"26fa7184e4ed09b9462bba937a0f5cf0ed3eb4a0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 16:57:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b14e07b2bf5604397d29f6556992f67ac65fc463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d37bfc674b148ff9386afa1a56c31db\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/60f789f7bb9f4ef99845df3d0dc7f81a : SUCCESS in 35m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5b5d44e11f9144898233f777337d0b5c : SUCCESS in 1h 15m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bdb908f333944434a8a6dfee2e3a46fe : SUCCESS in 10m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e4e3abee6c304eb1a8a412a6f261fb39 : SUCCESS in 30m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bcbaca49f61647d791d84cb90dca6eea : SUCCESS in 30m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/16b18d2e3c0e40b8aeab981270ba179b : SUCCESS in 1h 36m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5662a849dd6b4d7f94212ec97c361d00 : SUCCESS in 1m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0721782c785c460b8793041fd4f4249b : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80990a68f717495f90933e5106401d1a : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea57fd413d0845ce938b91d975bffc8d : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2930b40addf1499898a01e72504c0ffb : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b499dcf4b7d1476ba6a777f6bd943dd2 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/79489c0d3be0460695dc9cad48d2eb9d : SUCCESS in 24m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b1ca202a5a4c41c0bbf0ca76ba58ba69 : SUCCESS in 1h 10m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a7b91317d3374253a7dcba02517f71c0 : SUCCESS in 1h 10m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e2ae310305a7402a8a6626b7e5aff00b : SUCCESS in 1h 41m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a357aadb7bfc42aa9f5591914cd69c4f : SUCCESS in 10m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5aecfc760adf4f8b8d62d2ec66fd82ba : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e5a530767fa94411820f261f3bab6a19 : SUCCESS in 1h 06m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ef0d7b642c194c098eb848847f0fe17e : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"98e2a2a6df58952693fae427810513231bb96fa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"689c002e7b3b073c70f697ca2d459591968453e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:22:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a72d74fee164ed5975824147a6f4f37\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5827cb8c76834457a719215fdab7f11d : SUCCESS in 11m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c7b93e3be32f4decbf217c80e60b7304 : SUCCESS in 16m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3eae0df7fbe04f05b49b6ac224dd087e : SUCCESS in 30m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/65e8686aa11a4510a100e96e210920dc : SUCCESS in 1h 33m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c070e17946b49a282cb838c0a0620b7 : SUCCESS in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15dadcc18679408fa321849f20c26bd4 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a513c321ee714adbac11d0b94a35982a : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af8c54c5ef24afa80130f6becbf53c1 : SUCCESS in 3m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ee6b3d21d8bc4a9e882cb93dd7f77e51 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/70e3c5e45c144ace924eb1710f7f36d2 : SUCCESS in 28m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d4b8f312eae3498c97811f49f66528ac : SUCCESS in 1h 09m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/339ef50e7ae846338322273222d93ab0 : SUCCESS in 1h 08m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f8869cbd26454f4e9d2e69d543cd4f93 : SUCCESS in 2h 08m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8bbe24e4d0cd4ad2809009d84a261fec : SUCCESS in 11m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1d73a7cbf5d84832b14d84a67113bbe2 : SUCCESS in 25m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8cc2adf599ae4403927855293a5eba75 : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/931ec7b69ae94bbbb68445b3aa181a8d : SUCCESS in 1h 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"154db1b5f7ad61f3f68af8a80f9e1eca41a9fe33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:22:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7b0380294ab0bf36248fbde9b72dd70d679e1874","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:23:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1286f9c1586d4409824f6db48e354afa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a1a36900fc4a2da03e1ef8e1dc1ef2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1},{"id":"e8e74fdc63af98cd54b358fa98c59b15525c6df1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:17:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb35c6954f5943a9a953e652424f1274\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/18719a630f134234893ed84e0d2975b6 : FAILURE in 55m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9a49215f2a7e4229b6115ec18e76542b : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cf10facf6ac740e698e058687e127869 : SUCCESS in 2h 15m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/711495a69a894c968d927bb6f73fc8a2 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c416ddde9104459a4d51f3e01ed2ee4 : SUCCESS in 5m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/18be41562ff4489d92ef78d78502094e : FAILURE in 59m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/55f6487e227f46e080ae92ef16dfb966 : SUCCESS in 1h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b14d34b4b863706a2236e12b3d4f2471a5aefc6a","revisions":{"b14d34b4b863706a2236e12b3d4f2471a5aefc6a":{"kind":"REWORK","_number":1,"created":"2025-08-06 15:14:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/956691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/956691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/956691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/956691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/956691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/956691/1"}}},"commit":{"parents":[{"commit":"9398dc43bab48a85f86bbec7d93c28d5bf622a95","subject":"Merge \"change erlang arm64 repo to rhel9 based\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9398dc43bab48a85f86bbec7d93c28d5bf622a95"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 15:14:32.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nSince Docker 28.3.3 the daemon rejects a push that carries no\n`X-Registry-Auth` header [1]. The SDK already sets this header when it\nfinds credentials, so the breakage happens only on anonymous pushes.\n\nDuring `PushTask.push_image()` we now check whether the SDK can resolve\ncredentials for the target registry; if it cannot, we inject\n`auth_config\u003d{}`, causing the SDK to send the minimal \"{}\" header that\nsatisfies the daemon while leaving authenticated pushes unchanged.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nCloses-Bug: #2119619\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit ddac7ca1ed393609b7bda8753db3598f1ae02648)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b14d34b4b863706a2236e12b3d4f2471a5aefc6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b14d34b4b863706a2236e12b3d4f2471a5aefc6a"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
