)]}'
{"id":"openstack%2Fkolla~956692","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I7a2f3fce223afd74741b40bf62836b325fca5b19","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-06 18:30:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a2f3fce223afd74741b40bf62836b325fca5b19","subject":"send empty X-Registry-Auth for anonymous pushes","status":"MERGED","created":"2025-08-06 15:14:43.000000000","updated":"2025-08-07 00:55:41.000000000","submitted":"2025-08-06 18:30:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956692","cherry_pick_of_change":956456,"cherry_pick_of_patch_set":17,"meta_rev_id":"57174581d2b75c0fdf0a91a6d4d96aabf1cec22b","_number":956692,"virtual_id_number":956692,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 18:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-06 15:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-06 15:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-06 16:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-06 16:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 15:14:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-08-06 15:18:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-06 16:49:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 16:57:44.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"a04f00e3adace89a415c4ba6308d430f02c14c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 15:14:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3a3e7c44daaa5c39e5c16103a212c459fb86db59","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 15:15:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f01607171f8e65890a0c226bb072c003189aa0ef","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-06 15:18:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7a3ede4d0e9edababf88066d4e28d1385f6d59e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:49:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0378fb2c0f8a40b4ac83cae0692f0c12\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cd38ae23d99c4fffa80d4cf88cbc999e : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/02cd8a84858846f5903f85adf90235d5 : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f6c450d6eba04101b65156aa30bcf8f0 : SUCCESS in 23m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4fe9acbcba91468a930290777360161b : SUCCESS in 31m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4511eaca044d43d3ba8833016eaa8445 : SUCCESS in 58m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cc90b3146ca64f668cf9e4cdf17dc01e : SUCCESS in 1h 25m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb880ae377141708f03fffe7ed48486 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e12485a08fc4e919a4640ddf705e830 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/327df92dc0584ef391987b5ab938f5a5 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d76d94c05dbf4192a075cd0bf67d7dd2 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2c97173c0ef4010a377addc2a28667a : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9974b6406e4265b2e2d717f21abe57 : SUCCESS in 6m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/569497481574442394f721082a99af4f : SUCCESS in 5m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/08e7646d8f77499a99073b175710654f : SUCCESS in 20m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/04691c7b0adb48ffaa6af9c02c28f340 : SUCCESS in 33m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b3f7900eeb224d77bf00ddb2d70a8079 : SUCCESS in 1h 06m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cc3288f3011c40ad99c11725bca8a429 : SUCCESS in 1h 29m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/234dde6ec3f44f97854fa72de70f1423 : SUCCESS in 21m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8cec2e235f1248edaab3013e52b995f9 : SUCCESS in 30m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b1ea0c8831654e69b0cdc18e3b2ecd51 : SUCCESS in 1h 03m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/401dca4cd2ba4658804737a3163deb68 : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":1},{"id":"854005d2f9d7aab394c5e95d0bc98040ee0e7cde","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-06 16:57:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d0998c6ce9bb9888f709fefe332f8157320d846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:57:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45d1234fe9e771ad01aed1c40b5e1fbdf6cd0366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:30:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d02682e2fbe64be79c5252cfeba7cf95\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d2631e1f3c342b9aa5aabe7a8f701bf : SUCCESS in 9m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/42124a51ef9643de84856e3fcbe5e690 : SUCCESS in 13m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/54a6e73b7aa14b14a4a06b7136f98cf0 : SUCCESS in 33m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5b7177c0d88943a4a72af59ced93e00e : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20597e6bd0cd45f5a0b8d49f34ed08c3 : SUCCESS in 1m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f685c97a71c4c8e8085a38f3e52ed3d : SUCCESS in 1m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45ecc78e15d34c7ba3d135309e61c158 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1e60a82fbd4ff9916e9f4e34283bcb : SUCCESS in 2m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/884f87b60e0b4cd09793f79e3eac4842 : SUCCESS in 1m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/21879f2517a9418a838d0b592e837a8f : SUCCESS in 11m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/33f12bce821a4822a15977ae21aa32a3 : SUCCESS in 35m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/99be924948784f989d88dfa5110f9d5a : SUCCESS in 31m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ec387d8902804c308ab02ff803862e93 : SUCCESS in 1h 27m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5ddd9238cd30446da31e49c1b5af3aa2 : SUCCESS in 10m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a0a9b72b59cc4bb48b1138af6062974c : SUCCESS in 13m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa10ac1502ea4f58b99c918a6314c681 : SUCCESS in 1h 01m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4f6ba4279297403f81d09e9fbaae315c : SUCCESS in 1h 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c67717775091fa14157608fd378bd04d39786559","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:30:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b761456ac06710350b87dae058a39888d7b7df3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:31:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee1d9ceb93d41d99070dbfa46c188cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9c7c1f119f4eb2bb20ac5950d323ab : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"57174581d2b75c0fdf0a91a6d4d96aabf1cec22b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 00:55:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8a57b880eb4eaba0359fc6766ef2c5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/be28d4dd15e945b4832eaba5fa673e16 : FAILURE in 52m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/98e74e1256894e8082ab381db27a0317 : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/faf339c432c942da961974f5d0233ff4 : SUCCESS in 1h 53m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09e2aa4a3fce4450bb008025cd216ec3 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8fcddb44676d4653915b16a400c61fb5 : SUCCESS in 5m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1439da53a3184888a0463fd2e3db8842 : FAILURE in 51m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d06e4389187a418a958b098959a5dd5f : SUCCESS in 1h 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa554d1e8e132d7203eb0642b72e575dae0a16d0","revisions":{"aa554d1e8e132d7203eb0642b72e575dae0a16d0":{"kind":"REWORK","_number":1,"created":"2025-08-06 15:14:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/92/956692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/956692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/956692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/956692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/956692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/956692/1"}}},"commit":{"parents":[{"commit":"b83d409a23e4f73e50b198dd4e4e34dda6faedd5","subject":"Merge \"Add rsync package to the nova-ssh container\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b83d409a23e4f73e50b198dd4e4e34dda6faedd5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-04 09:31:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 15:14:43.000000000","tz":0},"subject":"send empty X-Registry-Auth for anonymous pushes","message":"send empty X-Registry-Auth for anonymous pushes\n\nSince Docker 28.3.3 the daemon rejects a push that carries no\n`X-Registry-Auth` header [1]. The SDK already sets this header when it\nfinds credentials, so the breakage happens only on anonymous pushes.\n\nDuring `PushTask.push_image()` we now check whether the SDK can resolve\ncredentials for the target registry; if it cannot, we inject\n`auth_config\u003d{}`, causing the SDK to send the minimal \"{}\" header that\nsatisfies the daemon while leaving authenticated pushes unchanged.\n\nDrop this addition when [2] is fixed.\n\n[1] https://github.com/moby/moby/pull/50371\n[2] https://github.com/docker/docker-py/issues/3348\n\nCloses-Bug: #2119619\n\nChange-Id: I7a2f3fce223afd74741b40bf62836b325fca5b19\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit ddac7ca1ed393609b7bda8753db3598f1ae02648)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa554d1e8e132d7203eb0642b72e575dae0a16d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa554d1e8e132d7203eb0642b72e575dae0a16d0"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
