)]}'
{"id":"openstack%2Fkolla~956759","triplet_id":"openstack%2Fkolla~master~I136bb702b20e941611dd90830e81c9d358a2e355","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-08 11:06:42.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I136bb702b20e941611dd90830e81c9d358a2e355","subject":"CI: use opendevmirror for debian/ubuntu builds","status":"MERGED","created":"2025-08-07 09:18:32.000000000","updated":"2025-08-08 12:48:51.000000000","submitted":"2025-08-08 12:48:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956759","meta_rev_id":"ba56165fc386849c0fb638f7797ce6970a8f7383","_number":956759,"virtual_id_number":956759,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-08 12:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-08 06:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-07 10:09:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-08 11:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-08 11:06:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-08 06:22:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-07 10:09:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-07 11:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-08 06:22:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-08-08 06:22:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"711d3b8414a9539d30c91a35e10f332699be12c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-07 09:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"918d91a5af8631835f9971159844755efd377c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-07 09:20:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2e4054944fa46641883844d88a3c68828962e99f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-07 10:09:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2c019c277ee7cacb80cdb790047256540eac0873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:00:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58554401e80342289aaf466e71fd7147\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d88d1efb22de43fa9d726a3c89fd244d : SUCCESS in 16m 01s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/d51c60903706498ba12fb4bf92a0beb2 : SUCCESS in 33m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4edcea589c574f68ba6024b2e7370896 : FAILURE in 4m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/454ab2ee63584a5db4507421b5bcdedb : FAILURE in 3m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f26f7339a80847429b6ade643336b270 : SUCCESS in 1h 06m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3a4b349f9d5a4727af82545d56a55627 : SUCCESS in 48m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df62f27ee6684ea78b0f569e1f75abc2 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d1378f1fa094a6c8142e451cf630620 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ea484a590c64c3aaebde4e65d33d1e9 : SUCCESS in 1m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f81f9030a5124e96a3f6f8ea9607303a : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40deb9c28129450cad5e4ec2cfa598ba : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848d79971bd34037884cc94e4d9d6dd3 : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ee3539a69d494d68860bf6c95a0db2fe : SUCCESS in 1m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a7c69b9141104fd18f52be44842dc9c7 : SUCCESS in 22m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/59a0abb278aa476884a96297b9b43736 : SUCCESS in 31m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e3bf0bfc9ca740f887e26f28b5028582 : SUCCESS in 1h 05m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/02ba126359244767abc0a63a89542574 : SUCCESS in 1h 29m 57s","accounts_in_message":[],"_revision_number":2},{"id":"915d05f6e099627281d78719b50b5a44e11079be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 19:20:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2c28088b5ac47b3bbedc81929b09132\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3423df13c6094f38ab22b93c78444ba0 : SUCCESS in 1h 35m 43s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/82571b9379ab43a7abf5f980169b9307 : TIMED_OUT in 2h 33m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d48ef10a6e804d519680ee0670b265c9 : FAILURE in 6m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/10f0296703f74cd7b0b413a685b4f358 : SUCCESS in 2h 34m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf7b3aa5948d4d6aa50902c1d93e09ef : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b209c2dcda064730ba1efa1d70d22113 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6622b4d80c9481f952c97885971a731 : SUCCESS in 10m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/96d1a1f1aca04f4b9cee20092d61a73b : SUCCESS in 1h 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8957546b9d9acaf6682c750da7259989602efbc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-08 06:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e361a1ee8fc606934e4c6ac6240c9a0d403ce711","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-08 06:22:31.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a91a3a01d035882aabee84fda1e88a960469f85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 08:15:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ab7c8affb0e48b0b5721c9eb440bbac\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6564b8e0b0524f99bafe797dd4f32d39 : SUCCESS in 15m 54s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/498f9a00f4214dff945bade75f7e994b : SUCCESS in 38m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/94f212f283024107a369eb8376b735fc : SUCCESS in 17m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bd14b91779094a69aca8be14befba928 : SUCCESS in 26m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1e52b9d0f3dd4013a9202081a781fafb : FAILURE in 6m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2ad895c9122648ab9deff4ac8638362e : FAILURE in 1h 50m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99435b2fec84ed8b9bf2b37533cff09 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34cb2b72f346419a805b7ac21dc2baf9 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0fbd62f7430414b867c216e11e1ff4e : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e6a2284833c45bd9b9e19b201532683 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c27c0b9898a4ee2ad2236455855e691 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4e8e23640845aebeb8276318e3d964 : SUCCESS in 5m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/489a6c79ba894a01add2f1eccc146f59 : SUCCESS in 4m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7ad6e908fcd54fa39f7bcd79a0eecbe8 : SUCCESS in 22m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0c53efd5cb8a40c7ab9bb2780e8e2c4a : SUCCESS in 25m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a81e404c7c17417bb81eaad475a318a3 : FAILURE in 6m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7cfbec9a2ed347918a912230dc4a5f68 : SUCCESS in 1h 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"840429f45ff94d9b1bd2c111a9f4da5c5d7759df","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-08 08:28:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eaaf032445939afca096fcbb1aa3867406ffcbe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 09:18:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/988197a5608544d49e4004794a6b303d\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/e349729c5e00443794c95d27936b66da : SUCCESS in 14m 32s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/388a79d274b3474c9afff8e8a2cd9b11 : SUCCESS in 33m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fd51690270bf438282a48c4a6b7d5cea : SUCCESS in 9m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/56df907a31864ae591593ed0b66e0f11 : SUCCESS in 14m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7df30b1d4cdc426e83e4f1b2a82085a8 : SUCCESS in 30m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/87e5cbf36e1a4a0da0e597647687bbe9 : SUCCESS in 46m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db0173ccc5f4e3cb559bfd0e7636b4d : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20c304434a5e4b4681c55cf01d308678 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a052b3cb9b344a7adc275435cc03bc4 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f6ce26f74674e298d31ae745b9017ef : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84fb9a09c500489ca69f09d12f745157 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b1cd5f3ab54bcba0cd6eb14cd91172 : SUCCESS in 5m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/33de50ed79c04ddfad77f9612d3e5489 : SUCCESS in 1m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9babf93b6e1649238ab78a99a30f1e73 : SUCCESS in 20m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f7fa1ed420043a7aec07e39cc330aee : SUCCESS in 13m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ee2f0f6ea98f401d9c33503f79ef8725 : SUCCESS in 31m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7e749897d5294caf8f07d3a77063ec9f : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":2},{"id":"96074687dc97b570d173b4aff93ddd53fffd78a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 10:34:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db95eee6d2f4aacac66ade986a6191f\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7331ff13740e404e8f2403ec45cd5117 : SUCCESS in 52m 54s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/0c210fc0b0244102998a250a4984154b : SUCCESS in 2h 20m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/908adb6b85ee44968fe928e341840263 : SUCCESS in 48m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be0873ffa75040cb8197c8044ff65123 : FAILURE in 2h 21m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/327603db066342918b24d1d80697acf3 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/701a45e50bef4faa8e27bebc2c2c6767 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe2a708ad0f64f049d4767097a7b93a5 : SUCCESS in 8m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e1b706a97052453690060005779d20d3 : SUCCESS in 1h 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0bf6f7ca71ba7d1c4958f10ed691604ae7eff82","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-08 11:06:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae1c29df9f5b312f2b02911cff3c54451519a0e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 11:06:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cde336eb69b52d2c10bc5d47aa3bc617aef97581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 12:48:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99379e051ea248639c9e798edc65faec\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9799386f275c4c5ebfeeccb3fa8e976b : SUCCESS in 21m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fd83cb1429e54bd4987029fcbbd5f1b8 : SUCCESS in 21m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/374a1e5393aa403581d6ca266ab2e89d : SUCCESS in 1h 01m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfb0d71af6d4f609ce40ea0ea831d1c : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e12c3c83f2b4c6ea6d315ed310179f3 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aaef0d7dc9645aba8835488c8bdf255 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52a094f9cf544c687960e26224ef167 : SUCCESS in 7m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/41ef707a0a794578a0f49c11cd8e25f2 : SUCCESS in 3m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b890c832c0ef4662a3a4bc30b0fa368e : SUCCESS in 21m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/16c314e594a24570a34d4d0abbdfabe5 : SUCCESS in 33m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/563f49d7b69b4df4b748db4ecd6c5b83 : SUCCESS in 1h 12m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/befd842d5ee840a1b5a83f341e39eff5 : SUCCESS in 1h 35m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e375b639c00d9e1a44c724221ac4c825300b6ad0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 12:48:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ba56165fc386849c0fb638f7797ce6970a8f7383","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 12:48:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e138d8e0eb475f9b95746653942d13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbd217e70984f81b0be2cdd708e887f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ec52073b62aee431772293afb85f248b2709b09","revisions":{"82ded04a9ab9484a5bc6a859729a24d5b195ebf0":{"kind":"REWORK","_number":1,"created":"2025-08-07 09:18:32.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/59/956759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/956759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/956759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/956759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/956759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/956759/1"}}},"commit":{"parents":[{"commit":"f48e1b40f267f58838134b466d550413c510fc50","subject":"Merge \"collectd: Deprecate support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f48e1b40f267f58838134b466d550413c510fc50"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-07 09:17:29.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-07 09:17:29.000000000","tz":120},"subject":"CI: use opendevmirror for debian/ubuntu builds","message":"CI: use opendevmirror for debian/ubuntu builds\n\nChange-Id: I136bb702b20e941611dd90830e81c9d358a2e355\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ded04a9ab9484a5bc6a859729a24d5b195ebf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ded04a9ab9484a5bc6a859729a24d5b195ebf0"}]},"branch":"refs/heads/master"},"6ec52073b62aee431772293afb85f248b2709b09":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-07 09:20:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/59/956759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/956759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/956759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/956759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/956759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/956759/2"}}},"commit":{"parents":[{"commit":"f48e1b40f267f58838134b466d550413c510fc50","subject":"Merge \"collectd: Deprecate support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f48e1b40f267f58838134b466d550413c510fc50"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-07 09:17:29.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-07 09:20:12.000000000","tz":120},"subject":"CI: use opendevmirror for debian/ubuntu builds","message":"CI: use opendevmirror for debian/ubuntu builds\n\nDepends-On: https://review.opendev.org/c/opendev/system-config/+/956760\n\nChange-Id: I136bb702b20e941611dd90830e81c9d358a2e355\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ec52073b62aee431772293afb85f248b2709b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ec52073b62aee431772293afb85f248b2709b09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
