)]}'
{"id":"openstack%2Fkolla~956812","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I1e334b47c6ab16eb032458141faae55cf39e3eb4","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-08 12:56:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e334b47c6ab16eb032458141faae55cf39e3eb4","subject":"mariadb: pin to 10.11.13","status":"MERGED","created":"2025-08-07 16:42:56.000000000","updated":"2025-08-12 19:59:00.000000000","submitted":"2025-08-08 12:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956812","meta_rev_id":"130551ebdff28b43999f650c0d5c500d095731d1","_number":956812,"virtual_id_number":956812,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-08 12:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-08-08 08:17:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-08 11:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-08-08 11:43:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-07 18:42:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-08 08:17:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-08 11:43:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9b334dfbf07f4d7d65e56db5f4547b0e62938f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-07 16:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdf1c640c62349abc40f76cbe123d1f1bd3797d9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-07 16:43:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5da2603b016b7e2b6b83cb97a13b16721588897e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 18:42:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9366390514a64b4caf25dbcb48ba976a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a2ba00518b09416aa78e55ca2e259526 : SUCCESS in 20m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/331a09af193a47eb83c2d7f116200e47 : SUCCESS in 1h 05m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/74188d2e58dd4e6ebf5ff32ff8d9a9d3 : SUCCESS in 21m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5fc722e5c4124d0a9699cd195c4dd465 : SUCCESS in 25m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a7925bcb29145c0b458a16bad8da7d4 : SUCCESS in 1h 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d9b9c9857a418e8851ccad8b705fc3 : SUCCESS in 1m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70316e343e8449a87e8ac301ca2bcfa : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ca3b6ea279a4a7b9b986ad6ff761f1d : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a2dd4549db64828892b86d20d3a7997 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79a47fab09c8487f84302fd77e79dcbf : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f1e10952ab4b138706d4e7ae0c1a71 : SUCCESS in 4m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/09ed07ce7f304f38bf19c479254a9d11 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cfd511588be143dba4ecc14ee5cbd118 : SUCCESS in 19m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/01f41904211a432387819fe8bf3b9764 : SUCCESS in 27m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/27908332091e4db29721828b1a2d5a54 : SUCCESS in 1h 06m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/93c4d2e5a02f4813b774fa75ab30d8d9 : SUCCESS in 1h 52m 25s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6776b121a7904901a6cd602b0b2b6d25 : SUCCESS in 1h 21m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/21e42a2eca50495093f58f7f0d294639 : SUCCESS in 1h 30m 48s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8c5482f4edde49a0b69ebfb450eec97d : SUCCESS in 26m 04s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/37493aa5568f44209b36046d89144d8c : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ba2ec44a5a304e55a07f6b43359c42e1 : SUCCESS in 1h 11m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/77678bbb3d5a4ba58399ef9eb44755ac : SUCCESS in 1h 27m 09s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2128c48d2bd0445aac5c4291891fa063 : SUCCESS in 35m 59s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc4b25b156643c4bc71c0f808697fb9 : SUCCESS in 9m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94b6d348ccfa4e358f8ca22d31d772cd : SUCCESS in 21m 42s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/dd02b575b9e74b959eb6fad874e1139a : SUCCESS in 53m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b5949081b7114836808f13f3af8ab3f4 : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b6b85f1bab4ca4b1746114205ed4e11be10cdad6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-07 20:22:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"bb2648206a39eba01d080885dee15357d1052e8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 03:04:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94a8344654874686a3b182e420dcc472\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c257a35e02d6489ea212714c065fc9a1 : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/47e6c6fe9814442cbb648b74f92cb0e1 : SUCCESS in 2h 11m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d326ddc398c437ba851879758f0c560 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61a95e628a1f4ce2a2f063eee85106d3 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cabf20f18aaf4b2fb9fa7e2e91a94d35 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5cd024a3d1cde34359b22e9d27688c760b0dd4d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-08 08:17:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db1386778ff0d2bdac89d78f2362bf6d6403b644","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-08 11:43:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6e80825a99b5409d873d35311b0a1b169d7b60c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 11:43:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02e9bb7336aba4a8cd28ca8c35cd9423a01da4e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 12:56:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a993900466348daae390fcbf455d87f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c682a891128d48f98bd370d9b2682e65 : SUCCESS in 21m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/04c2f06fb620499b98f07a70e74ba423 : SUCCESS in 29m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/89f057a4a3f74041b33e3cecf1ca69cb : SUCCESS in 1h 07m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406d306cf600499cb4ab81e1b363fd24 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e7a427cf0024701a77406128d202139 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59a5e625affb4dc68127847990347f3c : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55c84b254864c2393066eef90cc8f3f : SUCCESS in 4m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/440276d9b1624009aa81eca768f325dc : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/27003237a6d74804b20002821ee47cae : SUCCESS in 19m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/95aa05a5d964420aaae1eef29aa69b51 : SUCCESS in 28m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd4dca321da64cca86b139a504b7b80f : SUCCESS in 1h 06m 54s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9439fc60834a349d356e5acc5aab9b : SUCCESS in 24m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fdc1848023c4d9ca78a57bbc493c313 : SUCCESS in 25m 30s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/d67d467a58cf46a7a758ebb932939f8a : SUCCESS in 59m 12s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa9d8415d2c54ffcb0b0d34ae9743fbc : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ce2a83f9c9ef6097733fc181e47f0f849c695108","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 12:56:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84ed2b2bb7a6d30d78f613f5b6c7f6aba1cb031e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 12:57:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99bd4a29f6741508ae22c0e9ddbd903\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c0722d547b4e0da009468aa4bf5287 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1},{"id":"130551ebdff28b43999f650c0d5c500d095731d1","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 19:59:00.000000000","message":"Created a revert of this change as I0b7d73cd536ef2de4d0034596558827fba108e37","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b1a12cdb47c6156b732f5b83fdf94679ba548e6","revisions":{"9b1a12cdb47c6156b732f5b83fdf94679ba548e6":{"kind":"REWORK","_number":1,"created":"2025-08-07 16:42:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/12/956812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/956812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/956812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/956812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/956812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/956812/1"}}},"commit":{"parents":[{"commit":"9f7cc8ff320c940edf59839977c5bd8e866714c9","subject":"Merge \"[2024.1 only] Add conditionals for Ubuntu Jammy\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f7cc8ff320c940edf59839977c5bd8e866714c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 14:22:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-08-07 16:42:37.000000000","tz":120},"subject":"mariadb: pin to 10.11.13","message":"mariadb: pin to 10.11.13\n\nCloses-Bug: #2119954\n\nChange-Id: I1e334b47c6ab16eb032458141faae55cf39e3eb4\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e478adcbe63d81faa44bc09f72461fd6a7fcdf94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b1a12cdb47c6156b732f5b83fdf94679ba548e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b1a12cdb47c6156b732f5b83fdf94679ba548e6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
