)]}'
{"id":"openstack%2Fkolla~957013","triplet_id":"openstack%2Fkolla~master~I852f2ea723adce89b998402bc2464efbe9167af1","project":"openstack/kolla","branch":"master","topic":"kolla_ansible_ci_rework","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-12 06:49:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I852f2ea723adce89b998402bc2464efbe9167af1","subject":"DNM: Test 903771","status":"ABANDONED","created":"2025-08-11 11:31:11.000000000","updated":"2025-10-29 09:46:13.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6bb00d96e11e53be9eb8668c1ddb806336ceaeef","_number":957013,"virtual_id_number":957013,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-12 09:29:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-11 11:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-11 11:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"110e0ea57c5d2276eed003cc38986e85afe3e9e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-11 11:31:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d45f88b6945a12848e60fe9e8d963920cbc4a6f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 11:49:36.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903771 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"41dd3605c58705ea8debeef5ac3f2072910122d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 11:49:42.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903771 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"1fb30a12a77fb5208046a55324a75403d15be348","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-11 11:49:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c289d7a1ab4ef9cd2756aa3ba4624a4b06056e5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-11 11:50:00.000000000","message":"Topic set to kolla_ansible_ci_rework","accounts_in_message":[],"_revision_number":1},{"id":"a849c98db3f0845cf9d26284251b5efa8a5b472a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 12:58:52.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903771 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"5daf89238345f1aaf5faafc2382edea26f51fede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 12:58:55.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903771 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"b275af5262f5ac62e13bbcbd58a9bb19bf315903","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-11 12:59:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"caa13c337b4ddfc5219197a14535feca2ad9072d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 13:40:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/626742568280467f82624aa32f73ed04\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/199c669038a440acbbef4b7311b06c45 : SUCCESS in 15m 14s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6cecbbcf430e4761994108607951645d : SUCCESS in 37m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e09cfca9fb2c4ad499cb2046d3455597 : SUCCESS in 11m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0abb910852294be2bbed5cf994e81f12 : SUCCESS in 16m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fdf6ddc55e804d2f8a33930dcf39a1f4 : FAILURE in 3m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/82f5a167cf294f8badd507baf160622e : FAILURE in 5m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923ee47587b74f1d89173239b9aeb1f6 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a50ba754314c03bb7d1547e5695063 : SUCCESS in 1m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d47340e597c34aa08d6d945766e3f6a7 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba00fc9d2ed94a06b816949903cd0ddd : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aac0ab4960bf4356b87acb741a630445 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dcff263797b49829c227f45d8a7dbf5 : SUCCESS in 6m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c5e0ffebe5564bd8af697b8d3eddb65a : SUCCESS in 3m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0824deeca66a4b1591cc27d0a21dc0bb : SUCCESS in 20m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3f575efbcd0e473ba0066bc94f78e5e9 : SUCCESS in 30m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d2b05fc64a14a4c99a274244141e4be : FAILURE in 7m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/785066afc9d94a459157dca1481577d4 : FAILURE in 9m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5f6c9d8e8e42475e807f51c34632d510 : FAILURE in 5m 24s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8aa9cb348c3d4b14abc4e59cf2d79dd4 : FAILURE in 8m 10s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/cf913b0a892040d0a279dfb363e5cf70 : FAILURE in 10m 21s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/fd6a1a0f17ef4366839586aabf37b755 : FAILURE in 7m 52s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/12bc16a39f904578966f52cb50d742f8 : FAILURE in 12m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/71c1698646814ec0826a83988c3be6ec : FAILURE in 7m 22s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5f10719527894b0cbfdde3fe78e69abd : FAILURE in 5m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b45b10ba1f9644a881b7a35e9d2ad393 : FAILURE in 7m 24s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/cb8bbd801ca649acbac012164422d6c8 : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b3757ab83c62cb0f04a61fbd9d50b8de4d8a8b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 15:04:54.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903771 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"a8d5d7efeae2dfd48c0488c954753306e1f1497d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-11 15:05:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc3c5b34746d22e523ca3070dba9f25437c9fe34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 18:06:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fd5a6be32a845988568a7fedda7a792\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/964aa286137446f9b2d6564cadee842f : SUCCESS in 14m 42s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/aa4b5b59b5274ccdb32269a50f8c7b80 : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/28949eec47b449cb99a8ca3eb2da2652 : SUCCESS in 19m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/77caf8a51666406dbf801d42113fd7ba : SUCCESS in 27m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2a4d26a42b2a416797ad398d4d9b129c : SUCCESS in 1h 06m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/076c51c3fb5549ac80f0836752b69c67 : SUCCESS in 1h 46m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d60308c8f1434aa8eca5678c4491f2 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3df31b7f2e5b4eadb006c49a7c901c6b : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c63a64ee24b4eae901af14290b9bbae : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adad946122ca43db96df2c2605885095 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c42cefa3d0d64903b375f22c4391ec14 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d174d211f0844b978087c9a2dd9e94e4 : SUCCESS in 8m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4f88a88bf07d4d96989a301bfb663e95 : SUCCESS in 3m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b4492c4840e7484aae46f2aa3431816c : SUCCESS in 23m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/63adc82c7be048a4a92f40b2beb3cfa5 : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d874b09579da4aed98603f395ad85fcb : SUCCESS in 1h 07m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/57ad6eb2e68147c1bfdc18c57c078e62 : SUCCESS in 1h 47m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b92efde69a514be3be994ada485bc178 : SUCCESS in 2h 06m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5409f23084c147db9fc03cc8e263b3ef : SUCCESS in 2h 53m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c6ab8114cc8f42aa96ebb62f26bc7c96 : FAILURE in 23m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7eca715b80674b1cb0681ec6cd494a2b : FAILURE in 15m 30s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/141a835ffa1b401499f1ca2ca5de1db0 : SUCCESS in 1h 06m 00s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c6321a8b606e4145933281cbaf519477 : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/edb1b9039302427dbf4c5d302ffadac9 : SUCCESS in 1h 34m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e8903f30943a4d04a8bfc38c5469ad1d : FAILURE in 21m 36s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4b058c26463746b4b9dd6ebfb451a314 : FAILURE in 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba79506babe0d395c00ec308cf0672d86c228422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 23:08:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65144e1498244b6d8cf585c826677d00\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/2b240caa622c4dba9a8b5484d76f1836 : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1bf3642e8ce24963bffee9ab9e02d2b5 : SUCCESS in 2h 31m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/34a59aa7cdb2470f9a09c823a19971a7 : SUCCESS in 1h 06m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/03128cfba2024fb189cec5c7e34d7a71 : SUCCESS in 2h 31m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d2d703f2d984fd087b528ca8e6add77 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a41cdc6405f64e4a999cc99845d6b7a6 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91f02ce9d9894e5a9acce08ba1f6ade0 : SUCCESS in 11m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d05d514f4f864d52b83edb965a29b0ba : SUCCESS in 2h 03m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57910e224d2c1914d9f30ffdbea9927f22410bc5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 06:49:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"249074b7764641e5f1af92a5d25ac3baeb149ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 09:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db89fc807edb4909a6e6d7ba15cdcf83\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/61a0866a8aa3441bb7cb22f41a96c7e3 : SUCCESS in 14m 32s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/68ae7fd683814383a05eac2bd310dc15 : SUCCESS in 36m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d38bd05ab461469eb78d2e685c7def2f : SUCCESS in 19m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/14c0728fb2f54f3bb284206646d88952 : SUCCESS in 26m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a03d352428248b2ab227fb86ee96616 : SUCCESS in 1h 08m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0a887b7e683e4a628493a52ef48ace46 : FAILURE in 1h 35m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47a38251eee4d4583c0a6c3aa7cddc0 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fd363e55e1540078802cec0f53b84b0 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2d1121c730b43cf8a48fae46e4ff646 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3457928f59240729015836652fcd397 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0f889d2bb064b368a63b91904f97d43 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb53791d669470fb4f1c1139b437979 : SUCCESS in 5m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7bc0e64fff794fc282cbf2483078d9fe : SUCCESS in 2m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a99fe6cb035c460e99e00429649d3c37 : SUCCESS in 14m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/591e51c51aba4accbbeb9b29413dbe2f : SUCCESS in 28m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a611d97eaeb14b6b9cdf54c241f68786 : SUCCESS in 1h 03m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b23db508d5904ababb1992fb2a4350f9 : SUCCESS in 1h 30m 03s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0cbe9f19826841acbc34be2cf2f8b45b : SUCCESS in 1h 48m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/92c6a15cbed24b278efefd1827652390 : SUCCESS in 2h 34m 42s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/81781e24980f41638ec503501587686e : SUCCESS in 1h 20m 16s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/09e0d889a87a42cb8602e1d663b12391 : SUCCESS in 27m 53s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5a15a69794d647b194038c080eb586dc : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fc0fe4fa1d894ea9985f7af5e6487916 : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e3e62e1080034f328f880a1a7184920d : SUCCESS in 1h 36m 29s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5a1ad2da25d3465bb409f5c8ce34f5b5 : SUCCESS in 25m 46s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/23989336a3b64c39b5223a7722e76bb2 : SUCCESS in 1h 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8947fdac7a0407bcc5abb701f2d141f27d5d6286","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 11:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018a35f6e301492b9425498d2727e781\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/ff28c839b49341d79c56520bc0fcfa26 : SUCCESS in 58m 08s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/eba7396fcb9042578006d4cf7c6b1896 : TIMED_OUT in 2h 33m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6658a761c5e5488a8097ae595436663d : SUCCESS in 1h 06m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b45636e6098f4565a9d5c07581b491f7 : SUCCESS in 2h 17m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32840e562fc943ce8e097932952c0e33 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75d7254aa1b742db8bc7915089888be7 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7857f21fa678404eab514b0fe8197643 : SUCCESS in 10m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fca78a2c58fa4ee0a42d9f1284a36fbe : SUCCESS in 1h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bb00d96e11e53be9eb8668c1ddb806336ceaeef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-29 09:46:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8784413f0df1d5283f459e7cc94125d99fb5fdf2","revisions":{"8784413f0df1d5283f459e7cc94125d99fb5fdf2":{"kind":"REWORK","_number":1,"created":"2025-08-11 11:31:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957013/1"}}},"commit":{"parents":[{"commit":"82b52bee0ac30bda9d92f06036d26054731ca408","subject":"Merge \"CI: use opendevmirror for debian/ubuntu builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82b52bee0ac30bda9d92f06036d26054731ca408"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-11 11:30:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-11 11:31:09.000000000","tz":120},"subject":"DNM: Test 903771","message":"DNM: Test 903771\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/903771\n\nChange-Id: I852f2ea723adce89b998402bc2464efbe9167af1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8784413f0df1d5283f459e7cc94125d99fb5fdf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8784413f0df1d5283f459e7cc94125d99fb5fdf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
