)]}'
{"id":"openstack%2Fkolla~957091","triplet_id":"openstack%2Fkolla~master~Ib694d3be0d2a199469690a01ba17a3349f3802b1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-12 13:33:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-12 09:59:17.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-12 15:09:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib694d3be0d2a199469690a01ba17a3349f3802b1","subject":"Revert \"mariadb: pin to 10.11.13\"","status":"MERGED","created":"2025-08-12 07:17:30.000000000","updated":"2025-08-12 18:34:11.000000000","submitted":"2025-08-12 15:09:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"revert_of":956792,"submission_id":"957091","meta_rev_id":"62a1f9a76b9bac11fce31419bec6b3b6078e5f1d","_number":957091,"virtual_id_number":957091,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-12 15:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-12 13:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-12 09:59:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-12 11:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-12 13:33:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-08-12 13:33:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 07:17:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-12 07:17:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-08-12 07:17:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 11:49:28.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b5c1a5c00c839c27b274d8d047d8551b121c79fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 07:17:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd2cee697ba4ecb8443ef366ef5827d9d2022664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 07:18:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8b46a5b0710e7b8735eda62cccfe3419c0bfc743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 08:32:21.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957087 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"1f7e57025548da8896f3501d4afeb7062dd15e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 08:32:24.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957087 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"91ff066fc4f6b5bbc0f8c1d1540bfb96c278cc41","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 08:33:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b504f69d0d23b31358454336f6994cdaa4fd232b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 08:43:02.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957087 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ef42afd34617fe05c45230169771617b4a5a2817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 08:43:29.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957087 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"687481d5dea233394f2834247f78a3b05bb4f0cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 08:44:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6511eb14fb1ad96109b60cefb099657e7396e623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 09:18:52.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957087 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"d1bc73242e9e9f92c037c9401f356709af37f793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 09:18:55.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957087 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"0af39f83808f7b323708d8ab8a1ee6c88d79b199","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 09:19:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01bce19b1ec4a214a124d2928ed0754a566d0b6b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-12 09:59:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0d9e5af19eceeac811758ce44496fdb6b6f48ad7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-12 11:49:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"462466863070fc7857951504ee055ccf99ef5a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 12:32:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f985599d97344769c1eb7eff57ec027\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/4371426b69a6444989c01de3e2341262 : SUCCESS in 16m 24s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/eeddecce0e234feeb1c3c6d6528b6d0c : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/33bec7ce9dec44ff8118ba498dff1a22 : SUCCESS in 20m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/763df5e059b74b69825cf5ef9818629b : SUCCESS in 25m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5d172098ce3c4aacbcd44fa8596ab30a : SUCCESS in 1h 07m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/68c9431dbf45490ea0751ff63dc4f090 : SUCCESS in 1h 25m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bc4a9569e54e7ca438f55a03e984df : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b361c6c8cde4d4c8bde30f8f3bad6ca : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a355c6d791624ba18819006b4b411743 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61c46e29ce684db1ba57b796b3c50377 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a13edd96d5f041a49a254c8bb6c58d18 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d492743ac0b64fc18af8a7c67d971a97 : SUCCESS in 6m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/36f865ce12c147bea0f2c2c6c2ce279c : SUCCESS in 2m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/47aef506346e4861b9bf5fd29b172bc9 : SUCCESS in 18m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/afb7a1be7717443d9cb22c27114f4cb1 : SUCCESS in 31m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/13bbb6326a6b4037a3eb53f0975bdaab : SUCCESS in 1h 13m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ce9c0d69db84d31a47e9d71fddeb830 : SUCCESS in 1h 47m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c25aee8d6a384c848ce68f5f7e47c8b0 : SUCCESS in 1h 56m 55s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f57b2e9c1bd74b27962037b1259b2272 : SUCCESS in 2h 53m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/88e311ab7b844c75afcafd8194450d1e : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6beab0ab59224de2ac78033570a88f42 : SUCCESS in 29m 01s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f91cf40288dd4e0690019d9cbc715b89 : SUCCESS in 51m 43s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7d5c81fcfdd9403188487f25e51f8b26 : FAILURE in 1h 27m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/87e41b62f81a43018f6a6e41f852725c : SUCCESS in 1h 39m 35s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/90d3b05c9809465181a9fffdf4b52dd7 : FAILURE in 24m 34s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/08066fc070704296ba169637766495a9 : SUCCESS in 1h 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b62cffa8070a0da10d8edd81e27d1335458e889","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 13:33:53.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37337e725cd7f40ff3698a2290dc1682cb0e9b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 13:34:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e67341499c533596ee945ae87ff653b92e3468b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:09:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6baa720805242828cc75c544c6bc361\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ebc7ad69e35249b899b04aa395887e4b : SUCCESS in 18m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9c9d1d6ee6fc4aee8595d0b53375d421 : SUCCESS in 28m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15f45fae3d374dd28a25b009dc5a57ed : SUCCESS in 1h 00m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93263b0ea87942e0b6e9f90b7bdf07e4 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0d406ff8224447791a56cf73fffb47e : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14979774e48f4a91a957045fc2d2f23e : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066390b4770f41928bdf079060e2b5df : SUCCESS in 6m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/72301622fe1f4e7aa385dd7182504d06 : SUCCESS in 2m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eb06a3e640694e05ad51d69f95e4eb31 : SUCCESS in 20m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fcd5c15d7554420f921871097a37be1e : SUCCESS in 49m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/84521031ec474961ba317c3a45d37e41 : SUCCESS in 1h 08m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3a23a8d9a43744b3b36d6905639e6f38 : SUCCESS in 1h 33m 51s","accounts_in_message":[],"_revision_number":2},{"id":"749d2269b8f4d0bc29acbae0cc014f1f307786a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:09:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"234f55df64ea9a7bfa8da1d676a82d22007af0ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:10:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b95b05bb2d459ca990e540508acffc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a36ccdb9f92417faa660f16a985c0b5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2},{"id":"62a1f9a76b9bac11fce31419bec6b3b6078e5f1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 18:34:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ea0aa7f798047689065a127c3f13ae9\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6daf46a71e7840488710ea3e94b38223 : SUCCESS in 51m 43s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4acbd3b0bff044df9fa14d2ded83e6df : FAILURE in 12m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/237f5f1684e645a1a60c15905b52be6e : SUCCESS in 58m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d404a209ce94777924634ea32593814 : FAILURE in 1h 53m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30bbe16c63074c558f2cb9119811dbbf : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6349c92ce2164c9d945482e48c812c2d : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb1ae48cb6274d54a1093dd4daf424dc : SUCCESS in 9m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aa936868ef0246c5903b53b8cc7c266c : SUCCESS in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e77d2ba7942c833c7705ef9a879ce85a62879e93","revisions":{"dc8aff186c266451e2f750777df87757887fbb21":{"kind":"REWORK","_number":1,"created":"2025-08-12 07:17:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/957091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/957091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/957091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/957091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/957091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/957091/1"}}},"commit":{"parents":[{"commit":"e478adcbe63d81faa44bc09f72461fd6a7fcdf94","subject":"mariadb: pin to 10.11.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e478adcbe63d81faa44bc09f72461fd6a7fcdf94"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 07:17:30.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-08-12 07:17:30.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.13\"","message":"Revert \"mariadb: pin to 10.11.13\"\n\nThis reverts commit e478adcbe63d81faa44bc09f72461fd6a7fcdf94.\n\nReason for revert: mariabackup fixed\n\nChange-Id: Ib694d3be0d2a199469690a01ba17a3349f3802b1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc8aff186c266451e2f750777df87757887fbb21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc8aff186c266451e2f750777df87757887fbb21"}]},"branch":"refs/heads/master"},"e77d2ba7942c833c7705ef9a879ce85a62879e93":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-12 07:18:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/957091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/957091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/957091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/957091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/957091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/957091/2"}}},"commit":{"parents":[{"commit":"e478adcbe63d81faa44bc09f72461fd6a7fcdf94","subject":"mariadb: pin to 10.11.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e478adcbe63d81faa44bc09f72461fd6a7fcdf94"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 07:17:30.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 07:18:23.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.13\"","message":"Revert \"mariadb: pin to 10.11.13\"\n\nThis reverts commit e478adcbe63d81faa44bc09f72461fd6a7fcdf94.\n\nReason for revert: mariabackup fixed\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/957087\n\nChange-Id: Ib694d3be0d2a199469690a01ba17a3349f3802b1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e77d2ba7942c833c7705ef9a879ce85a62879e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e77d2ba7942c833c7705ef9a879ce85a62879e93"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
