)]}'
{"id":"openstack%2Fkolla~957150","triplet_id":"openstack%2Fkolla~master~Ic381d782594d80f8acaa32ae09be3f9f484a9d9f","project":"openstack/kolla","branch":"master","topic":"kolla_mariadb_11","hashtags":[],"change_id":"Ic381d782594d80f8acaa32ae09be3f9f484a9d9f","subject":"mariadb: Bump to 11.8","status":"ABANDONED","created":"2025-08-12 15:24:10.000000000","updated":"2025-09-18 15:10:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"e0997e71a11ec0b634374e3fcaaa6ccd7ce48b22","_number":957150,"virtual_id_number":957150,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-26 15:28:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-08-12 18:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"698c70079f7a2c599cfe80d45359fe235c977aee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-12 15:24:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ab14aa7b02568c3204a312876cdc6b93f09e8b1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-12 15:24:20.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"4f92410132633c190db91f4f921d0b1a3b340ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 18:39:20.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3623950a06054a43bd2c4b1f065142a5\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/63ac045ea01a4209a4648706153c2b07 : SUCCESS in 16m 24s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ae96133c48824795828e0b0c42f27a3b : SUCCESS in 58m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7f9d550797ce4b53a9c34131ef60360e : SUCCESS in 16m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2009cc794b094fe9b85f6e80efb85e94 : SUCCESS in 26m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/005935a108f84315a9c24909fed71dd2 : SUCCESS in 56m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/414b83cc067d45fda685ed075bea7d73 : SUCCESS in 1h 34m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0824022bd84c95a6eae517db6c1d4b : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/505b10ce2ca049e9ad5d2980e38f1811 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f97d6107f5194498bd54c1228a6ef551 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f36aa4fac344b6f860494f41f9cfc26 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/604ffd5b69554475962234bc452cf684 : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cff6cceb7e5448c8bf9dae79fb20efe : SUCCESS in 6m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dfbdc1f6ee8245988f97c17284d7c688 : SUCCESS in 4m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f19c4e2a18a46b2b4f33db8440b38bc : SUCCESS in 23m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/12c2340a2ba947f0976befb954315cc8 : SUCCESS in 25m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6f9e8489d05040ecb2ec65e5dd91e53e : SUCCESS in 31m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3f6deca346c043aaa6fa2f8a718650de : SUCCESS in 1h 34m 51s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e37dbcd7454741ed8f86a9489742212a : SUCCESS in 1h 46m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/65921820cf5f489e82ee8839c85175e6 : SUCCESS in 2h 47m 54s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/343eb3ce5fc443d98641a548c5068143 : SUCCESS in 36m 36s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7ef884a1a808475d864df4bf61ee2ff8 : FAILURE in 20m 08s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/985fb456d6b54bdfb26f7cff671a528d : SUCCESS in 28m 54s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e7b40b098da348c494e042c3b981917f : FAILURE in 49m 27s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7e2b426d77be4b5894f1ab375fcf7574 : SUCCESS in 1h 20m 50s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a24d4fbd56c44920a321026f20faf149 : FAILURE in 24m 48s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4a440f4f8b5444109060657ef77f4730 : SUCCESS in 1h 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e10307d70b3fba8e97b1c5667ca1efb0d5563a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 03:09:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d27aa21073e419e993055328a4afd14\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e52c28f8592f4f188ae773570572d8af : SUCCESS in 1h 15m 07s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a63d88c8f16d47038c07c9c0a7faff9f : SUCCESS in 2h 17m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cf0166f41d8b4106ae5878a528d005ad : SUCCESS in 1h 15m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5045a768eaee4fadad7ef7f77fd5b03f : FAILURE in 59m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83de73a60c7e45a991332f261aa1dd3e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a72b76a8c80c42ee983d746cddace1d5 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45a9694acff04af9b62f1c6c305f7bd4 : SUCCESS in 9m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5e5f12c8df494a5aae07b3e7aa621da1 : SUCCESS in 1h 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"989f4ec43ee34a4e8d28953f262ee5a2d9eb73b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-13 12:28:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5995961c334ff99d5573516f8cc1c366133876d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 15:02:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ec63e9dca40426ea8f29699d0a96efa\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/28350ad145694c19be21d4045bce4ca7 : SUCCESS in 15m 05s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/deaa1cf5e3254626be00a598219a0707 : SUCCESS in 1h 10m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/550ab5fd6dc344e88eb31d7b3a262c74 : SUCCESS in 20m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/71772a75808144529b317ce9fbc85221 : SUCCESS in 28m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5e0b5f22587543629e6fd4969882cbfd : SUCCESS in 1h 00m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f7269e324cd14aaa9b04a658973d34db : SUCCESS in 1h 23m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e5adb537fb48aa9a5701d4ae1937a2 : SUCCESS in 1m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/689ebad2560644aba3574e6bfa02a119 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc6cdeb5c61f4945b0b4600230394051 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d133bb799b94baeab05d651adafe75a : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e214ba1c8be745b397dfb233d1000c00 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cc5caa4315b4e41a859c3b1ea31e5a8 : SUCCESS in 6m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fc1c987e5ec042a0a97996407713d7da : SUCCESS in 2m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/916892d18e8e40149910b6b62d7ffb08 : SUCCESS in 20m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/633926eea00e412490289cb5f18e6091 : SUCCESS in 33m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/51cb33c1900d4aa7af0aae479db2f5e0 : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9c6412e04fc34788812128ce5565627a : SUCCESS in 1h 43m 17s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/df4b005893414abcb85cae4c49b32d6e : FAILURE in 59m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4c78f13f35b643ad92e03cc3b8a5f929 : FAILURE in 1h 37m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f2046c3324834f459e5254fa15a620a5 : SUCCESS in 1h 12m 25s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/12a9b3c2119c495ea2603d4db3b54d12 : FAILURE in 26m 48s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0cd74df85140419ba3d24cf39d60f833 : SUCCESS in 55m 10s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/573935d422ef444db799922884b8f0ca : FAILURE in 58m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/dcf8031e3a684f29ba0b030008053895 : SUCCESS in 1h 29m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/094b0fa2bdf340caaea113bbaef04a37 : FAILURE in 24m 51s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/61a64f7f78754f66a1e096acaca59914 : SUCCESS in 2h 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fc3e333ff77a9e14933dadc4bba499916302374","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 23:18:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a61643161f3d4a7ba897ec34246af70e\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/522f8053c5664bec90d60d53da910607 : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4132d871d81d4291a8f0b4af7ce490b6 : FAILURE in 1h 30m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df2b2a3da6e148d097f6bdf2dab05b68 : SUCCESS in 1h 43m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b6412fbc34e54791a248666cecb08f82 : SUCCESS in 2h 07m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d33af104fe6f443aa42c5283942e0675 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e38f99a346a34757bf10151be9741aa9 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0a72be94622436095c48971b3ab4439 : SUCCESS in 10m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/996438462a6546429088da066d593035 : SUCCESS in 1h 52m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f4ba79157e8423d61721925ba538517bcc05e34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-21 09:30:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"655dc36dac36a864773b77622e016d3ba870cc9c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-21 10:51:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96ed7a04d0bc0235894949fb81e911d519cebe1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-21 10:53:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2d41d73922d2b5ac775c7438354a749cc21c322c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:27:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e932b4959de4e708e1d8116621aa514\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/808df236c5ef496db352ad7c873e4069 : SUCCESS in 1h 33m 28s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/4a62dbc940b0487a8ab094f844bc6c68 : TIMED_OUT in 2h 32m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dda7d78132cd4e4eaa3e69460761e7ee : SUCCESS in 1h 32m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fcf32ac76f4447f5844947fbd5718db2 : SUCCESS in 2h 28m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f8bbeb6394a4241bb600948ba300eef : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a4d380c79ea4adcb77c696b8b5ba108 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab8e652a03d843f984f715f62410ab0d : SUCCESS in 10m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c070646b0d8b49adafa7241d34275bbb : SUCCESS in 1h 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"795ec4ec30592e9f21f121d3b304d3ec80e30135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:35:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84bee844717e4949ac31011b653ac6fa\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7cfdfbc795e0460bbd468543b407a05e : SUCCESS in 15m 55s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/c7662bbe24c749cabd12f55cee5c1900 : SUCCESS in 51m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9fbd466b65604f00ad886e57c115c6c1 : SUCCESS in 20m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/35abf7fbe8b44630add657a847e29da9 : SUCCESS in 29m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3861ae5aa1414b548fdc621824b59b33 : SUCCESS in 44m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c069b0e263a0467e81ea8ad3885921f6 : SUCCESS in 1h 08m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8d1179d15b341928292dacb8040de93 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73ae44744b3b43a583625e0072e901fc : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2af3d27c9ddf4f9dbbac6e0e8c1a85f5 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fb4785492f2417bb06674b8ce4cec63 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4007f3a42a248509baae9e6a1d71655 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8cd99098e642dfb110137f8dcf2707 : SUCCESS in 6m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a6466b8e55f743ca9b02b631521e25e1 : SUCCESS in 3m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/db094e7ddb144d47baec18ed7029c185 : SUCCESS in 11m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f493344d7184d028c68236607b1078b : SUCCESS in 13m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/36e8b091f8e146628a04def47d3493fd : SUCCESS in 51m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3bb30e5856f14215b03d7fda79aa89f4 : FAILURE in 1h 18m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9745a33dd4154b049726d4dc1d408e5f : FAILURE in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dfa6629d899a4d8087a2592a459135bc : FAILURE in 2h 37m 45s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1a3a51f930b349f3bf62df633877ca69 : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6ed562550e7242ca87c1bd1f591aa79c : FAILURE in 24m 08s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/94510d6abc2c4c53a7f56f59e3541c07 : FAILURE in 25m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b6bfba9c9d594cdeb9071fd02120dcc5 : FAILURE in 49m 48s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cb6468c95e994aaf900924dfc41cbecc : FAILURE in 1h 16m 04s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/736719555ccf4045a051fce31f9951e2 : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/bc30e20c57ab4ed89ec8f175834248d3 : SUCCESS in 1h 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a29da032afffceccf2be6ba0bead9da19d655982","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-22 13:46:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bae0c777ed3eb3354cae765b9476c30339afaa58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 16:19:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/052cbb2081e04d0da129a0ebf25e6311\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/b4a459db59f7487f9f53d59730b74ba0 : FAILURE in 59m 39s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/9879169865b14ba28db5636807d77711 : FAILURE in 1h 06m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5e49d5cef86641b59cb9cb383486fded : SUCCESS in 23m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fddea143b7c54b03861589249b96eb9c : SUCCESS in 29m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e39ccf89bc864d4bbbc7b933178633b4 : SUCCESS in 42m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3d220524397940b9b1bfc790a32769d2 : SUCCESS in 1h 06m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835997394f79497ebd22d7362f83c8a1 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf6ccc8f388c4ec19a56fca15140e6a3 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63e21f4b192647b69ee402a612940feb : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3a30e440fda49219c3327f37e5cbeda : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c73cd047769c4c2cba3790a812c3a352 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386cb7e3c0a7402a87d153c93332b2be : SUCCESS in 6m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/96961835176c4cde8537c57942f4fe44 : SUCCESS in 3m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d19cdbebe48846558798a2bd22e8dc1a : SUCCESS in 21m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d355935740dd44bba1b3f406b0b70c2a : SUCCESS in 24m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4fa5566af6964c2991fd31df7d3bd1e0 : SUCCESS in 1h 14m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bea5de81442640e084fbe789c9703074 : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/916444a7aea64ca2b46dc6e22f7b6869 : FAILURE in 45m 16s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9dddd032516944a38079235dd030e454 : FAILURE in 2h 14m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2ff2ed93d95e408098680e5e03d1fae4 : SUCCESS in 36m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/18d469a2b36e48e280cb9ade40eedddb : FAILURE in 25m 57s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4c2a9450a2c7451f9c8b05a0eb5bd94d : FAILURE in 43m 04s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/875eeff686694e409999fd2e4666e905 : FAILURE in 21m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e96a4e9f5e614e81a6368a2658ca17af : FAILURE in 1h 16m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b3c4fd02b8104ceaa269769173a81eb4 : SUCCESS in 14m 04s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/438ca5e201b04029886199d647ebf4ed : SUCCESS in 2h 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bf756f2d17646205f05fadec69cc2c8d747c9426","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:16:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b00bdac1864ebfa7cdecafa4be9f48\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/0a7f5fd5ef1042788d31f5e9c498b96b : SUCCESS in 1h 45m 22s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e59ceb23625f44bf9dd18bae62e21af5 : TIMED_OUT in 2h 33m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb254e0eef724156b672331d00b07640 : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8aec283c9a0549878419ba0d175f7ccd : SUCCESS in 2h 14m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c71a3b8c9224102b12a44b27791bd5a : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da061c06df274798a15809efdfcc719b : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/feddf6d4199f488da654d14f31497c0f : SUCCESS in 9m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/56f2a6e7c7064fe8badf368b50b1291e : SUCCESS in 59m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d9bc1b224ad08999609c499c908ad394e104fca8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-26 13:09:18.000000000","message":"Uploaded patch set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a5285e1ecb134c63e387c37222ce4ceb2a790a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 15:28:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c2ce78940444c0b0342618699de2e9\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5ed2a4f862e242778795aaccf64d217e : SUCCESS in 20m 27s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ee779af8742a42f9a440cf86d7814d7d : SUCCESS in 1h 28m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b3dbd7dccdc145dcb82d8a65fa3f54e2 : SUCCESS in 21m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/42a8da079ea94957a4c9885924218ed2 : SUCCESS in 32m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3cd745e2516d4fada6f3fc21d2069899 : SUCCESS in 43m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/64966a3ba5da476a92b2c8a977510d53 : SUCCESS in 1h 04m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4361bdfc9664c138a641240207e8a57 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a89370c8ee3a4853871366956d286ffc : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db31607c623d48babd82b20d0bfdd248 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45792b6f096b45a5a0d151c29cc58f70 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d59443b6bc04fd38303928082d2597c : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c536ad71d54e4cb38c51680cb6ad99 : SUCCESS in 7m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/546da9fca1014f07ac11d131ca6508bf : SUCCESS in 3m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f70c04ceec4407086a445ad5f1aec79 : SUCCESS in 21m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d0daecd05b1f4e5d99b79f7ff28986d4 : SUCCESS in 19m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7c1242b31e064366a6f2dd47584720ed : SUCCESS in 48m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e56184693d7b4e17b283a4ab779f42c2 : SUCCESS in 1h 11m 14s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b15e6204de6c4aceab4db905efab126c : SUCCESS in 1h 34m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f74ab1385da241f48f05fa05e717c763 : SUCCESS in 2h 17m 28s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/695212e62a8e4f63ac2671c62af43470 : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c05ef46f813d44daaddec70882b6b272 : SUCCESS in 26m 27s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e016ae1e2bde482f8f047ee3a25830bf : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7ab32e78ee654afc955a12b421de12c8 : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7a098668256847709f478184b9677f82 : FAILURE in 1h 05m 19s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a91e8b6f394840a3b7d0f8591f73d657 : SUCCESS in 27m 08s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/ce2b55506bdd47ff9703e04cca9b49cc : SUCCESS in 1h 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dc2f3cce31957f3bbc34269009d5bdc6975b5833","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:37:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f68594eceee4f81be66418a01b86188\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6b2a629f6eb84e158e5b38a37a93abff : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5ab408e49df146608d2f8f3d96fb724d : FAILURE in 2h 03m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3cf2d47ed5f34842a876b394166fc578 : SUCCESS in 1h 06m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69e3a708383445a49a002f75ba50f2c0 : FAILURE in 1h 47m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e78ae7ae41f74b3b888dee72d58049a8 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a49da479ffa8417b896fa7006e58a082 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39862177d7ff49cd888a47a68bc996a7 : SUCCESS in 7m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1b60b8de76ff418186a8308b5404203e : SUCCESS in 1h 07m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e0997e71a11ec0b634374e3fcaaa6ccd7ce48b22","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-18 15:10:49.000000000","message":"Abandoned\n\nGoing with MariaDB 11.4 instead","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"89d874971919aff0ea40c10980d2f62b4474aa4b","revisions":{"378f907d0429efa6bbddb9986c3ce061eed4572d":{"kind":"REWORK","_number":1,"created":"2025-08-12 15:24:10.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/1"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:24:02.000000000","tz":60},"subject":"WIP: Bump mariadb to 11.8","message":"WIP: Bump mariadb to 11.8\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/378f907d0429efa6bbddb9986c3ce061eed4572d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/378f907d0429efa6bbddb9986c3ce061eed4572d"}]},"branch":"refs/heads/master"},"83c3dad4cbeb32897703effb6b1c11c021d82d27":{"kind":"REWORK","_number":2,"created":"2025-08-13 12:28:21.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/2"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-13 12:28:14.000000000","tz":60},"subject":"WIP: Bump mariadb to 11.8","message":"WIP: Bump mariadb to 11.8\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83c3dad4cbeb32897703effb6b1c11c021d82d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83c3dad4cbeb32897703effb6b1c11c021d82d27"}]},"branch":"refs/heads/master"},"7bf8d3a06e41be19907bc3d06e348a2da9ab5ff0":{"kind":"REWORK","_number":3,"created":"2025-08-21 09:30:08.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/3"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-21 09:29:57.000000000","tz":60},"subject":"WIP: Bump mariadb to 11.8","message":"WIP: Bump mariadb to 11.8\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bf8d3a06e41be19907bc3d06e348a2da9ab5ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bf8d3a06e41be19907bc3d06e348a2da9ab5ff0"}]},"branch":"refs/heads/master"},"f68b417378866061a99948508b30fe0e6d765cda":{"kind":"REWORK","_number":4,"created":"2025-08-21 10:51:50.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/4"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-21 10:51:43.000000000","tz":60},"subject":"WIP: Bump mariadb to 11.8","message":"WIP: Bump mariadb to 11.8\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f68b417378866061a99948508b30fe0e6d765cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f68b417378866061a99948508b30fe0e6d765cda"}]},"branch":"refs/heads/master"},"c726a5e7359261eebe0db78f8c0eb64ca2f5513b":{"kind":"REWORK","_number":5,"created":"2025-08-21 10:53:00.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/5"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-21 10:52:52.000000000","tz":60},"subject":"WIP: Bump mariadb to 11.8","message":"WIP: Bump mariadb to 11.8\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c726a5e7359261eebe0db78f8c0eb64ca2f5513b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c726a5e7359261eebe0db78f8c0eb64ca2f5513b"}]},"branch":"refs/heads/master"},"596ebd44c2a72834ec60b50c42e1ac529b4a89c6":{"kind":"REWORK","_number":6,"created":"2025-08-22 13:46:27.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/6"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-22 13:46:15.000000000","tz":60},"subject":"WIP: Bump mariadb to 11.8","message":"WIP: Bump mariadb to 11.8\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/596ebd44c2a72834ec60b50c42e1ac529b4a89c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/596ebd44c2a72834ec60b50c42e1ac529b4a89c6"}]},"branch":"refs/heads/master"},"89d874971919aff0ea40c10980d2f62b4474aa4b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-08-26 13:09:18.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/50/957150/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/957150/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/957150/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/957150/7"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-12 15:22:01.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-08-26 13:01:16.000000000","tz":60},"subject":"mariadb: Bump to 11.8","message":"mariadb: Bump to 11.8\n\nThe clustercheck.sh was downloaded from old location and mysql command\nwas replaced with mariadb command as mysql commands are removed.\n\nThe default value of MariaDB client option ``ssl-verify-server-cert``\nhas changed from ``FALSE`` TO ``TRUE``.\nThis led to database backup failure.\nAdded ``--ssl-verify-server-cert\u003d\"${VERIFY_DB_SERVER_CERT}\"`` to\nretry_mysql_query() in backup_replica.sh where\n``VERIFY_DB_SERVER_CERT`` is ``FALSE`` by default.\n\nChange-Id: Ic381d782594d80f8acaa32ae09be3f9f484a9d9f\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89d874971919aff0ea40c10980d2f62b4474aa4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89d874971919aff0ea40c10980d2f62b4474aa4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
