)]}'
{"id":"openstack%2Fkolla~957161","triplet_id":"openstack%2Fkolla~master~I8c0ad9a096351694c7ae1e705e924282da2b6fd0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-25 09:35:19.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I8c0ad9a096351694c7ae1e705e924282da2b6fd0","subject":"openstack-base: Enable EPEL for uWSGI installation","status":"ABANDONED","created":"2025-08-12 17:17:03.000000000","updated":"2025-09-09 16:55:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"7620cdedf21e612ff315772e1d5582e51480d516","_number":957161,"virtual_id_number":957161,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-12 19:51:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-08-12 18:39:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 18:39:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-12 19:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a03d77407a486e03427e6c7821440f14e34b72fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 17:17:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8bdbca87042bd999592f3efafa8701eb591aa35","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-12 18:39:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8a735579f89f0889fcef85a1fb30724ba3560443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 19:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ccc60e6d3c4d2e879f71c49926b0c0\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/963e57ef9d164ab6ad40ea182a73ddc6 : SUCCESS in 14m 39s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/9dfddcc20dfc4cd68a91c040451d6152 : SUCCESS in 59m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0701b131359f46838a33c8eedb4287e8 : SUCCESS in 20m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aeadb77aa96a402f907b2b440e7fd4c7 : SUCCESS in 19m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7d0707c5bc784eeab98249f530adb7da : SUCCESS in 1h 01m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/53725b0fff994a4babc1143841d5273a : SUCCESS in 1h 22m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9712053974db4208b8bff54c5760c119 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ca21bd0e4224e9e90c932933a0b95aa : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b8e3b4054ec4c7486f052c3ad29792d : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a1ab6a289084c9cb03b072c28038853 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1c56bc1e25149a8baa3c280110c7d05 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80a66f4489141c0b088ec0b2cc3b29b : SUCCESS in 6m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/58eeba5961a747258794554a87664a51 : SUCCESS in 2m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8cb837f4ad3a4757a26899b699a46982 : SUCCESS in 17m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2faad519e32744f0bfe3fd4c6a7207e9 : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7da17d3d99f34801a197c3a74d49e952 : SUCCESS in 1h 07m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b07dcffde622430b807c586473b985e1 : SUCCESS in 2h 03m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a96dcc9f24084752b4f709e971ffe10c : FAILURE in 1h 05m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01198bd971654c2bb98aa95cd44252dd : SUCCESS in 2h 30m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5c8bc9dce592467697ff668163780841 : FAILURE in 39m 18s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2f0f2beb66654972aedd5883c6395b09 : FAILURE in 34m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/db61072048b04c73812ed16a500d160b : FAILURE in 1h 50m 08s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8b8269d2c5464115bf454b5211affc4d : SUCCESS in 1h 32m 58s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/9482d70943f744c0bb44967ffa32b897 : FAILURE in 1h 07m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a20bb6c73673cc0e0aa613cf26d3c9c065bf841","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 06:10:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc518808e111453e9a3d82cbc432dca9\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e7f964bb92bf4233b792e4c5ab2342bb : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1d00cf20538b4ab2981ba2f709711b91 : TIMED_OUT in 2h 32m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bf4b6d0a6fef485cb97cdd437d8738c1 : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/73dc05f3e8d8414483926a41220923cc : SUCCESS in 2h 15m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a2d0aaf4d794108a5b28eef6f509f9a : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c835a0c673a4505876848d7066dff27 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45c604d050d343708acd1c9bb591f0f3 : SUCCESS in 10m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/11764510e6834ea2b1eafa2406062619 : SUCCESS in 1h 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19adac2dd3541b8d7da183b7f06da9840cf45539","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-25 09:35:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"7620cdedf21e612ff315772e1d5582e51480d516","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:55:50.000000000","message":"Abandoned\n\nincluded in RDO drop","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00a4c92f9f37ab8b13a48bcad34131135ee8a42e","revisions":{"00a4c92f9f37ab8b13a48bcad34131135ee8a42e":{"kind":"REWORK","_number":1,"created":"2025-08-12 17:17:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/957161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/957161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/957161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/957161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/957161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/957161/1"}}},"commit":{"parents":[{"commit":"c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c40cc8c2efe6e8df3c0271d0ab8b5827b84394ad"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 17:16:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 17:17:01.000000000","tz":120},"subject":"openstack-base: Enable EPEL for uWSGI installation","message":"openstack-base: Enable EPEL for uWSGI installation\n\nChange-Id: I8c0ad9a096351694c7ae1e705e924282da2b6fd0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00a4c92f9f37ab8b13a48bcad34131135ee8a42e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00a4c92f9f37ab8b13a48bcad34131135ee8a42e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
