)]}'
{"id":"openstack%2Fkolla~957182","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I0b7d73cd536ef2de4d0034596558827fba108e37","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-13 11:56:18.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-12 20:38:42.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-13 08:33:49.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-08-13 13:08:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b7d73cd536ef2de4d0034596558827fba108e37","subject":"Revert \"mariadb: pin to 10.11.13\"","status":"MERGED","created":"2025-08-12 19:59:00.000000000","updated":"2025-08-13 13:12:56.000000000","submitted":"2025-08-13 13:08:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":956812,"submission_id":"957182","meta_rev_id":"0e5d53c171bf72d95ef4c08c3e205c112f912535","_number":957182,"virtual_id_number":957182,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-13 13:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-13 11:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-12 20:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-08-13 08:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-13 11:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 19:59:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-12 19:59:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-08-12 19:59:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-08-12 19:59:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2018995629fd5ee5f2afaf68be0f7ec7044c23ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 19:59:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39ce6aad5722a8772f49e89c858999358d9b6ad4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-12 20:38:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa2d70370ed3b5d86c8ee20b9240a8ded159da5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 21:46:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17406dcf7f4f4618bd0446172a4c2795\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d4a0eddd90dc456d847bc20e2eafec53 : FAILURE in 13m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fa5e4afc2bd74dc4b6cad7a6528b194e : SUCCESS in 1h 13m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad6d4df6850d40aeb7c534376d108a48 : FAILURE in 9m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/040dcdecbdf34cd09bd8176f62bc064f : SUCCESS in 28m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3b0b23437dd245408965e6de76efd33e : SUCCESS in 1h 08m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5de3a5f91645bebeecc09b4779860b : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4948b653f9c41a9a459792d2be1f69a : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecc2c63eb3594c0e855875659951ddaa : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03196edc6c894b739da69672565ac1ca : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b14404e279d74239aa7394069b66eb85 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f846638073e4b51a67973cddc70c7f6 : SUCCESS in 4m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/01a0c740a85f43d6b6ea13aa6d40bed1 : SUCCESS in 2m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/05bd24ea9dd448d78eb2c8a2c8bf9f27 : SUCCESS in 20m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f2bf04f74d1c40509ae3d765a12b3f84 : SUCCESS in 23m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/44528e78fe144defbdcb56ecc038a13e : SUCCESS in 1h 02m 31s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/3ac55a2301144b25a2f2d9941cd1bb5a : SUCCESS in 1h 41m 31s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/975cd13df7d64afbaf8c6abb2a1ada9d : SUCCESS in 1h 18m 55s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/113f4caab395426fbf3867e16557e30f : SUCCESS in 1h 11m 53s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3562ba1ac0004744815aa74014d2b6d3 : SUCCESS in 33m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/fd69859700044096b1aa1b40eccf0446 : SUCCESS in 29m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/13d3f06f39a647f48d1ce5f2d03e6098 : SUCCESS in 33m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0d2027021c644f5c82290f510cdd0c1c : FAILURE in 6m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/75d54de638064ec28199647d2751d222 : SUCCESS in 32m 39s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/467763d0c83d4992bde79018c445b9fb : SUCCESS in 20m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efe61f957d4c4e3c8e71972210538c84 : SUCCESS in 21m 26s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/3e7d938d71a44718a8a8b95028fdaba0 : SUCCESS in 31m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d83a3d09ae754c6f88b39c3fe9f2ada2 : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4b3462b1de4e9ed2c983753a398f3419cb925430","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-13 06:26:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"244b5ec14147fac3162175ff7fca29bf46ad1d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 08:11:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4723882670f34b7bbc15adb4357a4326\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/47da4aac85de46d8a8ad08bac5b6d8ed : SUCCESS in 23m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c8d716b4d85c4f95a46ebe58c05dc039 : SUCCESS in 33m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d8d74973913e44d1a1881413ad974757 : SUCCESS in 16m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/44cb1d37a6e44cf8ad49fd0b2ea8b52f : SUCCESS in 23m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b322d8b9e56144e18af2b02f3d6712ac : SUCCESS in 1h 09m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d7f0cd6e7174f92a6a71629a2a24120 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b386c817ccb74b64952c288f9c261ef7 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd1cc3947b86427782a6e2114c53dc43 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e64be06888ee49a391c7ee91e9a01b45 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b22f552b4b16453987ea8b1d7c8fe701 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b95d9329e814203b8c0a95f295985b4 : SUCCESS in 5m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/46af6cf85e794349aac335f4486a9d01 : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3a97aea7c0684bbab51abbaf0021c68f : SUCCESS in 19m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/37e0e4cd2714490eb65662b3890b2582 : SUCCESS in 25m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/af0f1944915742a0b5c6c1fd828c7a7f : SUCCESS in 57m 01s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1ddbb7bebb83475cb209c51f6fd120eb : SUCCESS in 1h 36m 17s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/45fe602d38da452690e68a2b21a16966 : SUCCESS in 1h 30m 35s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ea95e2d4b0bb4c5988aeb8054d1c0a04 : SUCCESS in 1h 21m 58s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/863ba9f3daa840d38a4eda9f20fbec15 : SUCCESS in 26m 06s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f0f18b33680c499c8ac7254e4eeccd22 : SUCCESS in 1h 02m 37s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/dda544f45d8744a8b3ac7b8b66a0d85e : SUCCESS in 1h 12m 58s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9fc796bd526845c7b02dbf7c987b9544 : SUCCESS in 1h 24m 13s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c92e3a4b50e84f4ba42c3e8a98c84bc7 : SUCCESS in 32m 15s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcaa17ecbe8749ccbc7ac587a016edd3 : SUCCESS in 18m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a018077848f41ebb9196a1e9c73db57 : SUCCESS in 20m 08s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/72b0946c6641470283a72bfa764605c7 : SUCCESS in 33m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba44e86d8ed942ff865a2108861bdedc : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d3720a5acf80f923006878cda3f2ce83bac92679","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-13 08:33:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc69140745bde046d96b59c0b7d0e849a53acc94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 08:39:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3a136a22d314b43a69d676c6164585e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/033cf2b064344f10aac01da3d729f9a5 : SUCCESS in 1h 07m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/537c79384daf46478b2feb01cd1f1c04 : SUCCESS in 2h 11m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23851fc6132d411f90922558cecb776d : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/911df5cbfaea4a3bbcd619ddbe4a142e : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc201f39a13640eea0a329a422334119 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2ca5bdabd1a30bbe767d322cc5721eec5bc005c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-13 11:56:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a23aafe576c70335ba21db790724186dc347d021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 11:56:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc7a607bb4b7f3152b6816ff343b9c2e5c2d2e31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 13:08:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c1a20640b64114b4658a02d02b1933\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/720caa44d50d4dc494e48d374fd75033 : SUCCESS in 9m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6102fd0d9f1d4b9383ece30800cc9783 : SUCCESS in 25m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/afca7c6091bf4caaa91425696eed2cd1 : SUCCESS in 1h 09m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c2f2b450f54ba3936a1953dab30b5d : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810d8014f2e14dbe9f51cff42897ba8f : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22481b1bedb94552b1818b635c651aa3 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defdf6d010a64f0fa29bb4b76c05d276 : SUCCESS in 4m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/370b30263b9949c786e9f94671de6f61 : SUCCESS in 2m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e8d094295c2a4ff893aa2ba07b0179b9 : SUCCESS in 20m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/465b69f6d41c496da5fccf54499d3396 : SUCCESS in 26m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8604b33e5a1b42b0a084b70d65cdcaed : SUCCESS in 55m 01s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4f8c24750d244c3ba6673142e09e7c7 : SUCCESS in 16m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63c33704c6174053a91d09b76f47ec74 : SUCCESS in 21m 05s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/4df91087203946a584a36137d9dd48ca : SUCCESS in 48m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b48b708030184a7da54c387e48139180 : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"68d1b312463b68b33c53f337efe99ed99a1bd442","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 13:08:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e5d53c171bf72d95ef4c08c3e205c112f912535","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 13:12:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c61a807a66bd42aa94c886bc18eb9f92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8c91f10cfa4994b42ff8dcc105b498 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"519c7e860fee494ab54d2e8601b43047eef8e53f","revisions":{"519c7e860fee494ab54d2e8601b43047eef8e53f":{"kind":"REWORK","_number":1,"created":"2025-08-12 19:59:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/82/957182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/957182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/957182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/957182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/957182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/957182/1"}}},"commit":{"parents":[{"commit":"9b1a12cdb47c6156b732f5b83fdf94679ba548e6","subject":"mariadb: pin to 10.11.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b1a12cdb47c6156b732f5b83fdf94679ba548e6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 19:59:00.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-08-12 19:59:00.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.13\"","message":"Revert \"mariadb: pin to 10.11.13\"\n\nThis reverts commit 9b1a12cdb47c6156b732f5b83fdf94679ba548e6.\n\nReason for revert: mariabackup fixed\n\nChange-Id: I0b7d73cd536ef2de4d0034596558827fba108e37\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/519c7e860fee494ab54d2e8601b43047eef8e53f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/519c7e860fee494ab54d2e8601b43047eef8e53f"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
