)]}'
{"id":"openstack%2Fkolla~957313","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I3484163f3c3198e8500e4e464499ce5776e71a5f","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-14 09:34:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-14 13:52:33.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-14 10:33:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I3484163f3c3198e8500e4e464499ce5776e71a5f","subject":"[stable-only] CI: Fix pep8 ansible-compat version","status":"MERGED","created":"2025-08-14 07:18:07.000000000","updated":"2025-08-14 23:52:46.000000000","submitted":"2025-08-14 23:51:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957313","meta_rev_id":"762629f672fd15053cd214a8cd51f14d5d4c7a68","_number":957313,"virtual_id_number":957313,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-14 23:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-14 09:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-08-14 08:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-14 10:33:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-08-14 13:52:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-14 09:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 07:19:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-14 07:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 08:53:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ea2340c9d34253bd8b0fcf685620e697d77863a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 07:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cd9569f5ec1750a9390f11570cb749a4b76e06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 07:28:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/477ad7801ee640c482a20384503b3700\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/327f47fc40464d91b2da5534566e0521 : SUCCESS in 1m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a22fc4cd5c84ecebfe93649c90d4b6f : FAILURE in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1c0a75dc0c54fc49ecdb6c18f9f4898 : FAILURE in 1m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b80d813a1e84adf945e6e318b528957 : FAILURE in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1399e20dae0148559b83c822c0ce7559 : FAILURE in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b050aa1c5b7e40b5897c6c75aab3c841 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65335545027f49f2b7c02aaf88db94e2 : SUCCESS in 3m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dee2c1a6383d42f99b39aa3fe543a555 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7ac245ae8075069f5e30e9db0a391dd690774499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:14:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"338fa3b00e4f0ad811c4f373680722663e118be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 08:23:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be89686aa5c5442b84e31cce653aba50\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d4d2b84045d4c489398aeae7fbc6404 : SUCCESS in 1m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86001f4c9a96457085dac13f3e649c05 : FAILURE in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0e02b745c1b46079e05d2888720a8fe : FAILURE in 1m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7346af8cd734cc386a61bcdc637b390 : FAILURE in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f28c1ce1e6547d291428e4bc711bb8d : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffbfbd60cb6a46b9b97740a2f78bb11a : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3afbb2c159e54777bc74f69a704ea9ae : SUCCESS in 3m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d9dfdeebd1794d8290066a185faf2578 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"05dcfc1a4d528ea10c7a29a6adbe4b1654c1df0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:23:22.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1c59810e33a285aad6c72fb85a09cec3137d7c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 08:39:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738e4b4078264b6e9484ec2da8ecf723\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f9c705453ea04fceadac3cef70726fab : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c9480a6e6841b8b7433ebe497ef4a3 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5210c8021f446fba874f9a985b1073e : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa61b3e725214eb6b581cc6d18452b71 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c1660fc300149d2a61e7729ab1c1e3b : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25fb679d64a2421c8b8684194fefcb3a : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/637bf57e3a15474886b670bcb74c7ccc : SUCCESS in 4m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/870b036b434044089eb6ad7890824bb8 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":3},{"id":"721421916701b366ed9f6297985fe55684c485e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 08:49:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b92e7c4ed73642daa35226279171d44a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c49c6b7ffd74fb58c73d3fedd683fab : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e48e25a2f84044cbb2c5439f712998a4 : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5fbb5fa66f6f54b5419b7772812e4b4263f6df11","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-14 08:53:26.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"853ad1b92b10f7886bb93ad0bc8c532d28f70363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:56:09.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c607ade71d4ef2b0296032a4fb189c22bc3aec45","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:56:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a9303bf74685b19629ceefa407d09c71cf2af29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:56:46.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d04e9a563110fd7ec39cace6645807aacaadf69d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:57:03.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"038de5e6a9b2f9217b485882c401c3d8f1c41cd2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 08:57:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bd278ee0433730567fa86ba7b132e0ae9d8b4b0c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-14 09:34:50.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":6},{"id":"c011127e19a38cb4d1f55248ed0067d68d78ba93","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-14 10:33:03.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f2fcc77fca16de6890c25b665bb4b2569637edda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 12:04:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf48dc3869341c1ab2dd32675add36a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b73320a52cc428489ca2d256eddd872 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/770c026012c64aaa8c01529408f88f16 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce3a1e153d6d0e43cca64d7e404f3eea58641e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 13:50:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42090e6b02524b5eabb2347ced514108\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67766f2b79494a00801d75397f1cdd7d : SUCCESS in 1m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9ada10b7384f069f623f807242778a : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35737949ddfc49d4a48c13f2286a771c : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebffec482b4b43afb27ba2e8c70ada64 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff09978e1193402fa76f72d16935a92a : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a23559fd26d4b96bf112010e0a3fed5 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/693211401a4345a28b13de1ad121cbb9 : SUCCESS in 2m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/907f654d781943dfbd39d90a850fdf56 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":6},{"id":"25bbd3e9d2bbc97ef76e589fd90251de48ed73e2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 13:52:33.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9eb23059e92fab103f844648af8ed94902aeaa59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 13:52:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"27f79d98ac8a6c9e4cefb7b19b246eace88dd292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 23:51:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fd128bbffd748d1a3006a3312c5496f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9bebd7ef77c42f48c3174bfeb10dffe : SUCCESS in 1m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c338df9157b45c0990addc5a56d3623 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9b7eada7a014852bba9e490b6c5ac49 : SUCCESS in 1m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65767b9104014d5a855295c0d9af49bf : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4856ac7cb1f646a5bc4db0c3e04eee59 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/725c05cc6c0444258a6018553da209b6 : SUCCESS in 4m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/92b6b062f3cb4061ad9f2f6c3370832b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":6},{"id":"9c389d80251b4d72b8ba341363081f589fe3dd95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 23:51:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"762629f672fd15053cd214a8cd51f14d5d4c7a68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 23:52:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da4a5995b737449c8b087037327a74b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e562244f4a475bbe6625942820eba3 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51911d9cdfd54f63b761a5a71e6872f7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"08248b170c5cb2771319cc546387020e7ff7dd4c","revisions":{"cb494b1e794d577090b138db7aa6707550eabd41":{"kind":"REWORK","_number":1,"created":"2025-08-14 07:18:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957313/1"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:15:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:17:59.000000000","tz":120},"subject":"[stable-only] CI: Fix pep8 ansible-compat version","message":"[stable-only] CI: Fix pep8 ansible-compat version\n\nFor some reason pep8 job on 2025.1 (probably upper-constraints) is\nnot installing the latest ansible-lint version - but is installing\nansible-compat that is required in the latest ansible-lint version.\n\nThis unfortunately results in the following error and failure:\nan AnsibleCollectionFinder has not been installed in this process\n\nChange-Id: I3484163f3c3198e8500e4e464499ce5776e71a5f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb494b1e794d577090b138db7aa6707550eabd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb494b1e794d577090b138db7aa6707550eabd41"}]},"branch":"refs/heads/stable/2025.1"},"5cfa9b0106042034c8847e8a0e80a0f4de1a357c":{"kind":"REWORK","_number":2,"created":"2025-08-14 08:14:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957313/2"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:15:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 08:14:30.000000000","tz":0},"subject":"[stable-only] CI: Fix pep8 ansible-compat version","message":"[stable-only] CI: Fix pep8 ansible-compat version\n\nFor some reason pep8 job on 2025.1 (probably upper-constraints) is\nnot installing the latest ansible-lint version - but is installing\nansible-compat that is required in the latest ansible-lint version.\n\nThis unfortunately results in the following error and failure:\nan AnsibleCollectionFinder has not been installed in this process\n\nChange-Id: I3484163f3c3198e8500e4e464499ce5776e71a5f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cfa9b0106042034c8847e8a0e80a0f4de1a357c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cfa9b0106042034c8847e8a0e80a0f4de1a357c"}]},"branch":"refs/heads/stable/2025.1"},"1d58e66c31f63f5252185423a18155f51329ff50":{"kind":"REWORK","_number":3,"created":"2025-08-14 08:23:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957313/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957313/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957313/3"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:15:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 08:23:22.000000000","tz":0},"subject":"[stable-only] CI: Fix pep8 ansible-compat version","message":"[stable-only] CI: Fix pep8 ansible-compat version\n\nFor some reason pep8 job on 2025.1 (probably upper-constraints) is\nnot installing the latest ansible-lint version - but is installing\nansible-compat that is required in the latest ansible-lint version.\n\nThis unfortunately results in the following error and failure:\nan AnsibleCollectionFinder has not been installed in this process\n\nChange-Id: I3484163f3c3198e8500e4e464499ce5776e71a5f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d58e66c31f63f5252185423a18155f51329ff50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d58e66c31f63f5252185423a18155f51329ff50"}]},"branch":"refs/heads/stable/2025.1"},"3ab6bba66da1deace7447f9d766db635e43ab56d":{"kind":"REWORK","_number":4,"created":"2025-08-14 08:56:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957313/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957313/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957313/4"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:15:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 08:56:09.000000000","tz":0},"subject":"[stable-only] CI: Fix pep8 ansible-compat version","message":"[stable-only] CI: Fix pep8 ansible-compat version\n\nFor some reason pep8 job on 2025.1 (probably upper-constraints) is\nnot installing the latest ansible-lint version - but is installing\nansible-compat that is required in the latest ansible-lint version.\n\nThis unfortunately results in the following error and failure:\nan AnsibleCollectionFinder has not been installed in this process\n\nChange-Id: I3484163f3c3198e8500e4e464499ce5776e71a5f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ab6bba66da1deace7447f9d766db635e43ab56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ab6bba66da1deace7447f9d766db635e43ab56d"}]},"branch":"refs/heads/stable/2025.1"},"97bd093fa7a86e47616d2bce0957fb5f341754a1":{"kind":"REWORK","_number":5,"created":"2025-08-14 08:56:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957313/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957313/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957313/5"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:15:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 08:56:46.000000000","tz":0},"subject":"[stable-only] CI: Fix pep8 ansible-compat version","message":"[stable-only] CI: Fix pep8 ansible-compat version\n\nFor some reason pep8 job on 2025.1 (probably upper-constraints) is\nnot installing the latest ansible-lint version - but is installing\nansible-compat that is required in the latest ansible-lint version.\n\nThis unfortunately results in the following error and failure:\nan AnsibleCollectionFinder has not been installed in this process\n\nChange-Id: I3484163f3c3198e8500e4e464499ce5776e71a5f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97bd093fa7a86e47616d2bce0957fb5f341754a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97bd093fa7a86e47616d2bce0957fb5f341754a1"}]},"branch":"refs/heads/stable/2025.1"},"08248b170c5cb2771319cc546387020e7ff7dd4c":{"kind":"REWORK","_number":6,"created":"2025-08-14 08:57:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/957313/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/957313/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/957313/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/957313/6"}}},"commit":{"parents":[{"commit":"00ebf586e3b34cfb5c0fba6446081542d4b1661a","subject":"CI: use opendevmirror for debian/ubuntu builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00ebf586e3b34cfb5c0fba6446081542d4b1661a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 07:15:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-14 08:57:03.000000000","tz":0},"subject":"[stable-only] CI: Fix pep8 ansible-compat version","message":"[stable-only] CI: Fix pep8 ansible-compat version\n\nFor some reason pep8 job on 2025.1 (probably upper-constraints) is\nnot installing the latest ansible-lint version - but is installing\nansible-compat that is required in the latest ansible-lint version.\n\nThis unfortunately results in the following error and failure:\nan AnsibleCollectionFinder has not been installed in this process\n\nChange-Id: I3484163f3c3198e8500e4e464499ce5776e71a5f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08248b170c5cb2771319cc546387020e7ff7dd4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08248b170c5cb2771319cc546387020e7ff7dd4c"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
