)]}'
{"id":"openstack%2Fkolla~957528","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ib738325603173157b4c482f9de6d5d9cea0b758a","project":"openstack/kolla","branch":"stable/2024.1","topic":"bump-libvirt-exporter-stable/2024.1","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-08-15 22:11:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib738325603173157b4c482f9de6d5d9cea0b758a","subject":"Bump libvirt exporter","status":"MERGED","created":"2025-08-15 12:14:10.000000000","updated":"2025-08-15 22:12:31.000000000","submitted":"2025-08-15 22:11:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957528-bump-libvirt-exporter-stable/2024.1","cherry_pick_of_change":957432,"cherry_pick_of_patch_set":1,"meta_rev_id":"58cb8304ede2585f3d9c44d5009a3cb81d810eaa","_number":957528,"virtual_id_number":957528,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-15 22:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2025-08-15 13:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2025-08-15 20:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-15 13:59:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-15 12:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-08-15 20:46:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-15 12:56:51.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-08-15 13:11:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-08-15 13:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-15 13:59:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-15 20:46:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bf6ef4f0d62f4f4eabe729eca10959e3715c2740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-08-15 12:14:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f8558f1d23b46e616146d52ea5483e7e1293063d","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-08-15 12:56:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b0f6ec4dd53326685f8db5d8fa04005cc7c0680","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-08-15 13:11:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe84bc6cd4c2f4add6a9c49e7260290db59a2189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 13:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62bd4be204c44c7b917facc9ced79992\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aad385a2832f4e3f84b941196b4b314d : SUCCESS in 23m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7ed8c63335444bb8a10270bd7999e441 : SUCCESS in 30m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/547f4d02222e4cb1a80b7912a54bedbf : SUCCESS in 9m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c6c3b66c8d8e4272880002bb22226c78 : SUCCESS in 14m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3df510c415f843d89ce326bec9449d28 : SUCCESS in 59m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cac39551e794ef993f4469a9fbb433a : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb6e4f4592604b9caa96141aebe8bf82 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e194fc479cf84e7cbf972c4731e9bedb : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1d9e1b6c2e44e0f86b4ee4adbdf9e32 : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2610bf624efe46c29feeb46a1c586ba5 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def626c24f3845b5b21d2fbb4df03570 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1798e35d90d3434eac837b0f184eb1fb : SUCCESS in 2m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ef3b2af82014da3afd49b954f45a6cd : SUCCESS in 2m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2e68ffcbd8634e50b0cc08f7ad0823dd : SUCCESS in 16m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8471c0a341f34fd6821d6e6d2f79a9b3 : SUCCESS in 12m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a8ff8d73493945389f170b77cac3f3e0 : SUCCESS in 51m 21s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d1fb03cd805437ab603a85a3e2f999e : SUCCESS in 16m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b7d0e37479340be90609081ff0324b5 : SUCCESS in 10m 14s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/12d9f0aec9b54659967165eba0d15872 : SUCCESS in 36m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c4e9a793855486c88b37c4ca548c97d : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bb6129a0ee424aa1359f4dc98f7f4d7d1f983078","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-15 13:59:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c21cb0f9be729cf81f84ed900836d9db1766744","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 19:06:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e09cd6e087047439c5348f907c2ded6\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa7041440fcb4f398b4e764dbbfabac2 : SUCCESS in 49m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/33f40a0bcc60499c907c385c8a947cda : SUCCESS in 2h 12m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/948b881a2c95408ca98619e8f3dfbadc : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/270bfcf5421d416ba83f8832e86bb9ea : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6c9dfa2e38b424e8f229a4bf98978d2 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3e826333b8cab639b09283644f02651a15106704","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-15 20:46:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bc54a9075ad1fee318eec009a26a01b8f50f9f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 20:47:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"202efee40e19658f3b22aa8204c0fcaf9827e32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 22:11:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7b4de46cb846a990863d50da0e1d01\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/494e2979a9ea4ecea4768eb4373f6ce8 : SUCCESS in 18m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/694ebc84e08744bcb28fb84373c105a8 : SUCCESS in 28m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cd21702864f74f96920c5a71e50a3854 : SUCCESS in 52m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394b47bc04c14b1b83aad0bcdb1da0ce : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4da3a1b046c149eeb2f54636914e6a9c : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2987cb0e48848e2a66d9a14c98a0133 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee02f6ac69a4fc4b96eb8e35392a95a : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04153609fbbe412387913759f2e3a446 : SUCCESS in 1m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dbda6f3eab544102991e6e785fc653e8 : SUCCESS in 3m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0f083926241e45f5a98a7e4f3b798588 : SUCCESS in 19m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/056121df39bc42e8a0d5882abac6e98b : SUCCESS in 13m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8eb92891a96948cd8315ba35777e640c : SUCCESS in 1h 16m 14s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba2ed1108c6c42b9a7aaeaeaaf5050fe : SUCCESS in 14m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15b66c0b2d34ee680a99e76698564b5 : SUCCESS in 16m 01s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/dad6110763ce4e24bd9cec10b5276704 : SUCCESS in 1h 00m 21s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66081b8d19d14330bd4d8fb8f6bc7510 : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2ffc30355ad60e25b395bec26a8406202f8886b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 22:11:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"58cb8304ede2585f3d9c44d5009a3cb81d810eaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 22:12:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b809788e9daa47cd9b1985196d403a4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d83115dd66e4804b6753ac7d4a3b77f : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afbe85b5769f45c487a045917f038b26 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fe1364310becf189e3020cebe23342b7a3e2420","revisions":{"1fe1364310becf189e3020cebe23342b7a3e2420":{"kind":"REWORK","_number":1,"created":"2025-08-15 12:14:10.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/28/957528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/957528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/957528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/957528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/957528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/957528/1"}}},"commit":{"parents":[{"commit":"97350d833734da2cd8a8a6b28b6a0f2377bbdbf8","subject":"Merge \"Revert \"mariadb: pin to 10.11.13\"\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97350d833734da2cd8a8a6b28b6a0f2377bbdbf8"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-08-13 16:01:56.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2025-08-15 12:14:10.000000000","tz":0},"subject":"Bump libvirt exporter","message":"Bump libvirt exporter\n\nBumps libvirt exporter source from v1.6.0 to v2.2.0. This fixes some\nmetrics that were lost when the exporter was changed from the old\n\"tinkoff\" source to \"inovex\".\n\nSee also Icc8a3651d601ab45d448c9d252ca6155219d21cf\n\nCloses-Bug: #2120639\n\nChange-Id: Ib738325603173157b4c482f9de6d5d9cea0b758a\nSigned-off-by: Alex Welsh \u003calex@stackhpc.com\u003e\n(cherry picked from commit ed094e0af4caa2bcd5d574e6c9f4079be074347b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fe1364310becf189e3020cebe23342b7a3e2420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fe1364310becf189e3020cebe23342b7a3e2420"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
