)]}'
{"id":"openstack%2Fkolla~957645","triplet_id":"openstack%2Fkolla~master~Iba7e01d210b5a6947fc281d2a4d5c75e9b840eb1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-07 19:50:24.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-07 19:50:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iba7e01d210b5a6947fc281d2a4d5c75e9b840eb1","subject":"CI: Switch to configure-swap role","status":"ABANDONED","created":"2025-08-18 06:57:34.000000000","updated":"2026-06-07 19:50:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c26a10b00390d5cdc25fd02907292abf8b40121c","_number":957645,"virtual_id_number":957645,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-18 07:48:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 07:22:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-18 07:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16b79ff8c5a13decbc8b24fcdb91eef6517d02a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 06:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10041ed0d848bfb8cbaa85a2a263ff52fd1a9b78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 07:05:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"978ab57d154b2cd25ba73828dd8101a6613e6117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 07:10:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25220e66ee2856cd3f86f6b4657fbcbffd3251b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 07:48:12.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/054a84f39e064d8c9b7eb2bb039c6051\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/ca53c86cbe4a4ed8afb3322d5dcf840e : RETRY_LIMIT in 3m 04s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/4b2602f7a49c450984d989388602c52c : RETRY_LIMIT in 2m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cd910fe274184b13914171b13e78b22b : RETRY_LIMIT in 3m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c29dc85c56f34663bd59650a8fa96554 : SUCCESS in 17m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ccab3858124e4a23b5b791a25e3f61f3 : RETRY_LIMIT in 2m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1c999df7518449d390bf2b72a5d34da6 : RETRY_LIMIT in 1m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bd92c8b3aa453687e2adf5ab1dc64a : FAILURE in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/837d923b552f4537b08025b4eca3e4c1 : SUCCESS in 1m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/efec7f1eb7a747da9cec93c778a43bb8 : SUCCESS in 1m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c5ae20a3c224c9ca5a2f1ca2b01bf89 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb15fbae6af4443ea80f709154df9efb : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992c78f305c840c28ec629617711aa7b : SUCCESS in 3m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ca40118b5c4b4d3f9329c1a1785a3113 : SUCCESS in 2m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/268df52600eb4f8f8419c065268fb1c1 : RETRY_LIMIT in 3m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0fc1e6de6d184a82a416e4fc951321fe : SUCCESS in 32m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1ee5f9e180bb405bb3460193dbb0d51f : RETRY_LIMIT in 2m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dca63c98748e4d2cb814267d898c6555 : RETRY_LIMIT in 2m 38s","accounts_in_message":[],"_revision_number":3},{"id":"f6144442f50a00d62247dfe13a18622aa246013f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:53:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15682e86997048a4a83fc0d3ebfb8ee9\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c1ab7845a7414183ab8d70f4c21b02a7 : RETRY_LIMIT in 24m 54s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/b2c6fb99beb4421b94f7caaa986a8cd0 : RETRY_LIMIT in 1m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f9ae0f386284dce80f5bd396f3d1f00 : RETRY_LIMIT in 26m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d060e74554b4870a5dbf9199376501c : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28fc2eb4b0e44db8bf0e5226f6a05a4e : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2ffe9da113a4d9fa44b1905875af4e1 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2eb90baec434cb0badbed4427ba1680 : SUCCESS in 9m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d06e73ab8a98496a91caa0ecee90f9ef : RETRY_LIMIT in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c26a10b00390d5cdc25fd02907292abf8b40121c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-07 19:50:24.000000000","message":"Abandoned\n\nnot pursuing","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e73717d0a7ecead12690e33231977be671a236e7","revisions":{"84d716f92565a73aee75fed0496684d8c67765cc":{"kind":"REWORK","_number":1,"created":"2025-08-18 06:57:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/957645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/957645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/957645/1"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 06:55:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 06:57:31.000000000","tz":120},"subject":"CI: Switch to configure-swap role","message":"CI: Switch to configure-swap role\n\nMove /var/lib/docker to /opt/docker\n\nChange-Id: Iba7e01d210b5a6947fc281d2a4d5c75e9b840eb1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84d716f92565a73aee75fed0496684d8c67765cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84d716f92565a73aee75fed0496684d8c67765cc"}]},"branch":"refs/heads/master"},"c004fced3f00beba236dd00fe9b16278031b67b8":{"kind":"REWORK","_number":2,"created":"2025-08-18 07:05:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/957645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/957645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/957645/2"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 06:55:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 07:05:09.000000000","tz":120},"subject":"CI: Switch to configure-swap role","message":"CI: Switch to configure-swap role\n\nMove /var/lib/docker to /opt/docker\n\nChange-Id: Iba7e01d210b5a6947fc281d2a4d5c75e9b840eb1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c004fced3f00beba236dd00fe9b16278031b67b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c004fced3f00beba236dd00fe9b16278031b67b8"}]},"branch":"refs/heads/master"},"e73717d0a7ecead12690e33231977be671a236e7":{"kind":"REWORK","_number":3,"created":"2025-08-18 07:10:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/957645/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/957645/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/957645/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/957645/3"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 06:55:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-18 07:10:33.000000000","tz":120},"subject":"CI: Switch to configure-swap role","message":"CI: Switch to configure-swap role\n\nMove /var/lib/docker to /opt/docker\n\nChange-Id: Iba7e01d210b5a6947fc281d2a4d5c75e9b840eb1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e73717d0a7ecead12690e33231977be671a236e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e73717d0a7ecead12690e33231977be671a236e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
