)]}'
{"id":"openstack%2Fkolla~957672","triplet_id":"openstack%2Fkolla~master~I4c4e94405a0cc4872881a3290d505b9f5e20f469","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I4c4e94405a0cc4872881a3290d505b9f5e20f469","subject":"[DNM] test bifrost on arm","status":"NEW","created":"2025-08-18 11:12:56.000000000","updated":"2026-04-21 16:02:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"e738d57bcd6b1e4e06fcba32531e51350d9d2c46","_number":957672,"virtual_id_number":957672,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-21 14:29:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-08-18 12:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c23d1e49939317f68c5a1f20b4fa2df8b5a52fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:12:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54b547175e87021aef21f4d77a09cf33fbd8f89d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:13:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2a6bcec7fef2d7d669769269e58d4b09a6fad3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:42:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d067f60cb43e4fe88936342009d6e1c5\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/db2a38fb06654509aa8d126e0b847ff9 : SUCCESS in 15m 11s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/91c079ca72ca4eb7b18f3ffa4a165356 : SUCCESS in 38m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/caedacef0c144cdc883575168e354045 : SUCCESS in 10m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4c4c97554fd748c6bef8321ad1f7c0e5 : SUCCESS in 21m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7ece836cc6964c62a97fa9d7d779a200 : SUCCESS in 37m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d673a18fbe244292b8f6ac72deb76ff5 : SUCCESS in 1h 00m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98511082180b4e58bbba17200cb2e2ce : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc25592b127f494fa8a0fec82d663516 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a90774d06f844b058869b53405121e41 : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e48328db9614517a5b5c5d343292b1b : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8388901c1b5341fbb25e6e66786a9526 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf68e6fa8c2f435e89cfd7ac26a0c94f : SUCCESS in 3m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4144e50befbf4da592288e7fb780eaf1 : SUCCESS in 1m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/755edb3a28b84bf68ebb2d42b31532dc : SUCCESS in 24m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ff5db172e7fb47a08e48a4c47de0efb4 : SUCCESS in 25m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59fd7ad8325547bebe0af5efc53c191e : SUCCESS in 1h 03m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1094ff7ed00d4b4ab6bf45f0fa2c2369 : SUCCESS in 1h 27m 14s","accounts_in_message":[],"_revision_number":1},{"id":"adfbfac9557c1452e52b9688d6ff21e45adc4edf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:45:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d79f05add2c349e58160bc9c86f79be2\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/8dbe420325c54d6a94ebdbb8ecbd1f7d : FAILURE in 53m 01s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7af91c4ac282460a805a2f5cb67d10fd : SUCCESS in 2h 25m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/162df80b8bd74bf293a710aa6b7d92e4 : FAILURE in 47m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/81c8e91291d441e7a3c9d0d19a8daeaf : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9510667942c848da90d0cf71e659e0e2 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e5c971aeed9498dbc389f5c06a942ed : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47de7da552d34d2cb25c316522a90837 : SUCCESS in 7m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6f7a69184b1744b0ac1244d7f7e14658 : FAILURE in 1h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d70e8ae243188afc9800032bc9441e0078869c94","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-21 12:54:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/unbuildable.py\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"20ec4ffa9ea1e607fa97d7fc8d8c55be0c8014bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-21 12:56:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"058fa15bde3a97f22bd9bfd4f2fdbe9428ea85c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 14:29:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/014adb1d63f042ee95b6a75c00f7f843\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/aaffe49aa4004eb783f2bcfb83c28d75 : SUCCESS in 1h 19m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3e710fedb6ee4ac192647c12b7ef82a6 : SUCCESS in 1h 29m 59s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f9373ffea9f94780ab62c8502ef748a2 : SUCCESS in 22m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/53409a42e2944c7b9e1dbc7564dabc05 : SUCCESS in 1h 14m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e3792e0e779345eb9b505410569eaaac : SUCCESS in 37m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0960a0a0015f4540b8929d0b13183207 : SUCCESS in 1h 00m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1bb2ca72e5554af894465c25c4300fd1 : SUCCESS in 32m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9126f3ff9f634a48b9a60a5019fbaac0 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13989e5b246e4ad2a5aeb4709160124a : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa7d8b783f3441bab319a71492d27722 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbbcfe8e220b494eb6c3f212a20154bc : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/58d0a161bb95470f885c263ce9a58e4a : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819d127203ff448bb1ee3cdd641f34d7 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df998eb175f485880a0aa2ae706ec54 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e738d57bcd6b1e4e06fcba32531e51350d9d2c46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 16:02:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a3ccf59a4d46df86ccb4297091a750\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b5b38c194cf44853b52c8d319f6ed8fe : FAILURE in 1h 15m 28s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6a80c7a1a78947daa6d311eff764c166 : FAILURE in 2h 04m 01s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a1fc030644b1415693d00feff355d482 : FAILURE in 1h 34m 46s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7e33629ace6f4d98a3633cc41af29348 : FAILURE in 1h 25m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd14d86c88ac41bca53723fc22f28a25 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f1acea4ce4c481bb15bc94a15ec8bc6 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/38481065cdd146cbb402f815ba267eac : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46dc5f917451d957091c669344ce47e143551798","revisions":{"1e3d14bdb174a52f5b6c7001a3b2e835681420b9":{"kind":"REWORK","_number":1,"created":"2025-08-18 11:12:56.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/957672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/957672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/957672/1"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:12:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:12:45.000000000","tz":120},"subject":"[DNM] test bifrost on arm","message":"[DNM] test bifrost on arm\n\nChange-Id: I4c4e94405a0cc4872881a3290d505b9f5e20f469\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e3d14bdb174a52f5b6c7001a3b2e835681420b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e3d14bdb174a52f5b6c7001a3b2e835681420b9"}]},"branch":"refs/heads/master"},"1f1051cdfb187603efa6f57b8d08a22b158460d9":{"kind":"REWORK","_number":2,"created":"2026-04-21 12:54:51.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/957672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/957672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/957672/2"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:12:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-04-21 12:54:51.000000000","tz":0},"subject":"[DNM] test bifrost on arm","message":"[DNM] test bifrost on arm\n\nChange-Id: I4c4e94405a0cc4872881a3290d505b9f5e20f469\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f1051cdfb187603efa6f57b8d08a22b158460d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f1051cdfb187603efa6f57b8d08a22b158460d9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"1e3d14bdb174a52f5b6c7001a3b2e835681420b9","theirs":"a9bae9fe56dced112e71de51bd9f382f00d584ff","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"46dc5f917451d957091c669344ce47e143551798":{"kind":"REWORK","_number":3,"created":"2026-04-21 12:56:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/957672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/957672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/957672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/957672/3"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:12:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-04-21 12:56:01.000000000","tz":120},"subject":"[DNM] test bifrost on arm","message":"[DNM] test bifrost on arm\n\nChange-Id: I4c4e94405a0cc4872881a3290d505b9f5e20f469\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46dc5f917451d957091c669344ce47e143551798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46dc5f917451d957091c669344ce47e143551798"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
