)]}'
{"id":"openstack%2Fkolla~957675","triplet_id":"openstack%2Fkolla~master~Iabf606848a6160eaf9e2494a2564f37c6db09882","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-18 20:42:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-20 13:58:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabf606848a6160eaf9e2494a2564f37c6db09882","subject":"Enable Prometheus mtail build for ARM64","status":"MERGED","created":"2025-08-18 11:23:07.000000000","updated":"2025-08-20 13:59:39.000000000","submitted":"2025-08-20 13:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957675","meta_rev_id":"5eaacf6a7f0165193df4ca8276d9e0df90283f86","_number":957675,"virtual_id_number":957675,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 13:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-20 12:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-18 20:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-20 12:43:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-20 12:43:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 13:08:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 18:30:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-19 13:22:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-08-20 12:43:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9502cc096b0ea0c40184b6df9a96d22c29b931c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89249188c1c20965bad959763f9f28dd1de7760e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:23:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"6d543b16494fb24c02525794fccdc2bcf69f5ef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:08:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e0fbd529604eec970aedab328f3ff6\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/44fbb6f73518453e8c3263cecc06a1c2 : SUCCESS in 16m 16s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/0edc97565c724b7ba0a863412cf79c97 : SUCCESS in 39m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a2f9b15fb0d43c1970148e136c79c14 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/688d1c15f94d437ba965bd956a98d30a : SUCCESS in 31m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9bd23fe3307844dba1128430f613b27f : SUCCESS in 29m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8279808e911b4cbc9596a3c3fd468625 : SUCCESS in 1h 21m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e000a685894a1a97bdff483af56dc2 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87e967177fa04a0a8f3ec7fdb148ef33 : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/561ffbdde1e748fdbafa6283e78d0a7f : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5ed56bb039846e1a6c649ba918fcbbe : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0385984d5adf48819bbb750478c4fb92 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2105f63fb7a04a9b9ee651a549dc6663 : SUCCESS in 8m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6330250d2a3c4417b6767d61e408153c : SUCCESS in 2m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/548e3f75657241a9b58a7cd6dc1114ea : SUCCESS in 24m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/847064152e3b4c8ea21bb9d4297b2bf7 : SUCCESS in 28m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b4f5a57bbfc24e6e851f8fe216c1e0a7 : SUCCESS in 32m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bf5e6b5745004c19839cc36ee627a56c : SUCCESS in 1h 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5fcf2979a5314000b30cc230b43d48b292257f9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 14:51:42.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b17b19afce1af009bb334fcfc37c914d43c05ef2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 14:51:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1391462062c9ac0ccf086d3dd93cd96dc085d477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:38:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84c4d7b37b5b49c7bc3ed591752b27f0\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7ecdecd49215418a86b1934a3269b5e5 : SUCCESS in 15m 54s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/70ba4d84e9c34f7c9371bbcc9a23ab79 : SUCCESS in 1h 12m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e4ee61743cd84cdaaf556e64f30b72f4 : SUCCESS in 11m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4fa806305e444a5c8d7b5bc3e8cba515 : SUCCESS in 15m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef08e69962c64eb6b923b60a41fbc327 : SUCCESS in 36m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/66a739d6854c4c87af9e7e7b0a00a43b : FAILURE in 1h 35m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7f21e16ee64ea180e1502369edb1cb : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f32d22cefac1475fba677f29ca4a5c90 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cb7b286730d44a4b8f1e4770279f622 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b815c73f85c044029aa9ae82a59e2a27 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f67a32c286e14af0872c088824ffa9b5 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a8bc77d8d84e899c6a9b237b376636 : SUCCESS in 3m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/567c2f1f5de4446aa7896aa285a7f1b1 : SUCCESS in 1m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3379ba8835e148c4b5ec01d5497888ed : SUCCESS in 19m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/df918a58b14f4d9b9ef4f856d02a8d1e : SUCCESS in 16m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e54bd7708fd41d0913964387c6da0b7 : SUCCESS in 38m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2d8087c69c9049c789b387019d8373ee : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e7caf535b9f0ccfce9cca059c4f111ac5eb683b0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-18 20:42:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d1d8a469617941632c1352c3c71db2a0ae8c8215","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 22:29:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c17f97c275fd4cc3a8cf4b96abaff752\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/81ef307a29d7435ea2ad309d4f1d9ac0 : RETRY_LIMIT in 2m 05s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d942cf4147af4e809ed192669aae2feb : SUCCESS in 2h 06m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3410aa1c066b4adc8fa4412717a52cb8 : SUCCESS in 37m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14d54aa528524c0cb09d1b026fac5abd : SUCCESS in 1h 39m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1019e03aa7ae459e9105e117ac76961d : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d4657fbbecf42248ed3becf58cedf95 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42f22cacaf9c401480bf1781fe9a5b98 : SUCCESS in 8m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/391b24492e8d476298f93c70ce538db8 : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a03d0ba00ffa3fce83fb02bda8539884e7f14519","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-19 13:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"200fb6dab092fc1bafd9d934204131e2d428c5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:07:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34fd174e0d8e4095992093584b7c5d6e\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/e5c8fffa0de44e7ea219f3f7fea15d41 : SUCCESS in 12m 04s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/60a5f082109440e5a8ccccdd7a3bd6fa : SUCCESS in 46m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e0fb382e9fd14e9bafefb5301173ef4b : SUCCESS in 17m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d8c9d3520040473d8b923bf43c57cf36 : SUCCESS in 26m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b3d6fd2534d84786915027056e30aa41 : SUCCESS in 30m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f12eeb205d754f95ba482177aa61433d : FAILURE in 1h 38m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2150cfcb4545f3be8f86d6f7ef0680 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d62249a390054d68857db1a41a675839 : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98430b9cf41a43e3aceeeedfa9d0a2e1 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e18247246cff4329bbbf83b8c4c8ee70 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f164ab7034394ecc8d1774f08fabb4f8 : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b14deb71c84ee8b1d63aa585017672 : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/760fce9b81b846baa55759dfc1f08863 : SUCCESS in 1m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e49ae3118f874c3d81ad7da87ecddd79 : SUCCESS in 11m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b14b52465edc48049561a1a1db549727 : SUCCESS in 24m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa7ef8e99abd4fee89746df92d7f89a5 : SUCCESS in 32m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/193a31fe4a9a43f4b272610be8e4d4d5 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":2},{"id":"429c8c02c9edc40e997c610f9f922b1b646249d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:23:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8451189d93dd412a987f232e67be1f47\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5b0f3a39cdb64bdb8ce12fdb33528983 : SUCCESS in 1h 08m 52s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/ac8a6e9c393141459e0542a3131cc7c0 : FAILURE in 14m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f890fe89af646acabeae56c34b47eeb : SUCCESS in 1h 08m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/21acf54f6c074cacb2ee9860f08b0ab0 : SUCCESS in 1h 20m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/51a906faae6c4d5a97cd13bc95f52c33 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a1b3ddac4375474bb62b46f2ed54cd00 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bee4c54ba6cf4287aa3bd074670c6d1f : SUCCESS in 9m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d885457c60524abbadb53b5ca9b7cd41 : SUCCESS in 57m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b452331f791648b6629866cb1ddb9426640f4d47","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-20 12:43:26.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"340917963b5c346d4a831b7dbf1cb23c23847813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 12:43:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9312d7a163a1b32d35522906f2ae8dd12e4810fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:58:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947d455576f647a3916193491a2e965d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/830eb97dd1974015be93f5d07cfcefb4 : SUCCESS in 10m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d6f385cb18244989db579e6ce2f3809 : SUCCESS in 27m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/28660cf50cab41ff815bb3744130626a : SUCCESS in 1h 07m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6066686540ed430fb0daaefed7fd27f1 : SUCCESS in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/033913869af344ff8f9da7c9e708f091 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/620efb6314c44949b3415f761dba35b0 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0368241f6944f59d53c55fcab571f1 : SUCCESS in 4m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/49c95902e281465eacc95f4199ffa519 : SUCCESS in 5m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7053ffb549234a149cbd77acd497c4ec : SUCCESS in 21m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4b3094e57db34562b319544bd15db446 : SUCCESS in 14m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/805b48d221134eefbe0d733e7d36975b : SUCCESS in 1h 11m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a8124fc9a2134cc5add88306d6833bae : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":2},{"id":"25f4e46539977d418bb6c9944e5f8066f3b8c0a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:58:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5eaacf6a7f0165193df4ca8276d9e0df90283f86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:59:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1bb4b6b2e2c40d7918f17445ca4a690\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5237fb9bcb2e4136ab9be179c95bc6a3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7ada218ccd6730f6a0529f88d656dc466461b04","revisions":{"c0ce796c77c35691284a7dd92f964812eb6be43c":{"kind":"REWORK","_number":1,"created":"2025-08-18 11:23:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/75/957675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/957675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/957675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/957675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/957675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/957675/1"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:22:41.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:23:06.000000000","tz":120},"subject":"[DNM] test mtail build on arm","message":"[DNM] test mtail build on arm\n\nChange-Id: Iabf606848a6160eaf9e2494a2564f37c6db09882\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0ce796c77c35691284a7dd92f964812eb6be43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0ce796c77c35691284a7dd92f964812eb6be43c"}]},"branch":"refs/heads/master"},"a7ada218ccd6730f6a0529f88d656dc466461b04":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-18 14:51:42.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/75/957675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/957675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/957675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/957675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/957675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/957675/2"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:22:41.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 14:48:41.000000000","tz":120},"subject":"Enable Prometheus mtail build for ARM64","message":"Enable Prometheus mtail build for ARM64\n\narm64 binaries are available currently.\n\nChange-Id: Iabf606848a6160eaf9e2494a2564f37c6db09882\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ada218ccd6730f6a0529f88d656dc466461b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ada218ccd6730f6a0529f88d656dc466461b04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
