)]}'
{"id":"openstack%2Fkolla~957677","triplet_id":"openstack%2Fkolla~master~I400e9a9a3bf03053589835f933023eedee504f08","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-19 20:13:12.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I400e9a9a3bf03053589835f933023eedee504f08","subject":"Enable Barbican build on Ubuntu ARM64","status":"MERGED","created":"2025-08-18 11:27:09.000000000","updated":"2025-08-20 00:14:13.000000000","submitted":"2025-08-19 23:23:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957677","meta_rev_id":"ae260c00a1ce43e30d5630102edf4b91b69fc353","_number":957677,"virtual_id_number":957677,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 23:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-19 13:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-18 18:19:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-19 13:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-19 13:23:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 12:29:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 18:19:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-19 13:23:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"515d24ac99000948e82571908b4cfeabce8f42ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:27:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1128eba6b153f83ca256d23224f1976d0127cef6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:27:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"53c47d6ba709b355a3709d36bae9eef28b873a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:29:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6923efb06f3c4062bfc22a3f4d98069f\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/25eb2917315b457e8b8096006cf0ce80 : SUCCESS in 16m 49s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/892420c145e6438d817100ff4ae0bf4c : SUCCESS in 40m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/027426b1449f4c798f1d57d47bbcf1f4 : SUCCESS in 18m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6a515a8c89214e9fab5cebcfb269c1f8 : SUCCESS in 27m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e91a170c92a246d3be4c89b9a4dcc277 : SUCCESS in 1h 00m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0b2dbc3283704710a419fe0a688bd2ce : SUCCESS in 48m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84233b5785e8470db8126b3d9fdc4f4f : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a789bbde8964b64b51184c55e2c9a4a : SUCCESS in 1m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c59517bf7f4c4f6ea0a7adaf739562b6 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b94389c61d64c21964c8cb61157dda0 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4057a3c735f430d87d9b27d639358c6 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7412f53ebc94733b74948bde6c20fa5 : SUCCESS in 8m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d1f8a68bd45546b2b051fca98b06ac12 : SUCCESS in 1m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f1f20fde904444a38d71b4cde214cfcb : SUCCESS in 12m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f34a3ec779ff4fa9985af33e68b1aa03 : SUCCESS in 20m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4233e14af11c4d429ebdc20183ac69a0 : SUCCESS in 37m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9a03cb9c58f6439ca203f64862f88144 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a9f39b5c4e758aaa060191c0aa27af9408fad04f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:53:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/453fad0da48b4e42bbdc7f231c6b2c55\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c1d879268c6e41da9c201b9484f6c074 : SUCCESS in 45m 09s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/0b9cb41900314ffe9482928021e72691 : SUCCESS in 2h 21m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0cc60c8b7b5e4974b8e420b163b99621 : SUCCESS in 36m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a637c219e71d447db2129f0288280092 : SUCCESS in 1h 59m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef509cc1aa894679bbce049421b54624 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6951b61abb5743beb5af8eb94c59093f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/975fd697fc05412db95e9e1f2634c2b2 : SUCCESS in 6m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6bc39f94ba764c83a30cc0252cebba50 : SUCCESS in 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b72e07141782c4c154fb8c9f5ea9c09d8d898f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 18:05:37.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a2e41983d3b613b2b1bfea6892230c5e53d706d7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 18:05:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"ff8c9b47073d064f816a637e9519277bccd0746c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 18:06:20.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"95ff330dd6d02fc612d4996b03d378fa6ed7b8c1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-18 18:19:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"82de1861bde92563a7ae068b983c701b8808cf71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:58:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f56a64d0330479a94b9aeada1887a2c\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/a274d4621d794c118cbaca368404ad3c : SUCCESS in 15m 42s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/3b9727a9305c441e88c106c81aac7592 : SUCCESS in 1h 18m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77db35dcbd1f4a498a021a898107cdf8 : SUCCESS in 23m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/03efc368904e436e964f45784bc1aa42 : SUCCESS in 32m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/787661bb14024260995d5c07e51ed6af : SUCCESS in 1h 09m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dd32623beedc408db3081d33639bb805 : FAILURE in 1h 47m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ccecee8db994075aa623ec77c62010c : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecc580a69a4e4563b281ccd57c865bb9 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73783550c8424baaa86ccc77472f478e : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3db23c559bf4f9a92e9e8b9bbc3f7f2 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f1a7eac910f4290a32ea56703a759f3 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9c40b5599c48e99758dce430202c4c : SUCCESS in 7m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3d0519783aa44ca8960b192795b68394 : SUCCESS in 4m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5d359f10c11492ea3ada01b965f2718 : SUCCESS in 24m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6718efaa5d854fecaa2c377d000f92c9 : SUCCESS in 29m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/39832924d3ec439999225c7ac9b1be83 : SUCCESS in 1h 15m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c2ec310840494b07856f2bab23affe5d : SUCCESS in 1h 47m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8590837ede636ebf4070d321078dc70b5855a83e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 23:46:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f328769167542439b038febdae7bace\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a2f4857c33504b0bb1670638b243905f : SUCCESS in 40m 25s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/33f00f88f25a4130a3370a9097f25d00 : SUCCESS in 2h 24m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2bbae6d5b519403d8ea914283f521231 : SUCCESS in 45m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8ab82aaee46a4199978e93f981bbbdba : SUCCESS in 1h 48m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b5e82a7d73748efafbe3f63427460c9 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16b6b06586a34432b0fd470443bc7721 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4896f9e137c24bc1b17e586211b2ab1a : SUCCESS in 9m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8bac5ef153614df69299c9544e09fcaf : SUCCESS in 52m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"38a64c787d85ee5aaf07b2180da53df8e1821f0d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-19 13:23:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"43460a929ce15010ffab9ddb9a18dfbacbf0717b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-19 13:23:16.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"48d36965b88e327949eff9ef10e4c4b0379d2d82","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-19 13:23:19.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"de19d929686383d77954439f627bcf1c621f133e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-19 13:23:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6468acb4fb8be125e76a7d471ddefb6e3b2b147a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:23:22.000000000","message":"Patch Set 3: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":3},{"id":"d468e21487b0e0ffc0d6adffea88bf9509a78cd0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-19 13:26:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"debeec5aa875977e5df8472593373666a08cb24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:16:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27220b0c9cb74035878797d7aeda756d\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/ea4b7c1ed9e640e7b39507d8f8e0f18a : SUCCESS in 12m 28s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/2ae7fc3e1c7f48a28767d76380d958ce : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f85e45a58b1a401190d81c8dcb8c817a : SUCCESS in 18m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/47857e80a2504140a6686e52d0f61e52 : SUCCESS in 15m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1c36565ba81f43309b9dcc80605c42f3 : SUCCESS in 41m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a22d53fcd94546259a99a65265b2d160 : FAILURE in 1h 31m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f05ac1e6265453595d43044459fe606 : SUCCESS in 2m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d6358d95bad4dbcb85738e63208ebb2 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c4d1e6b169447f5bc8b93b614d5a703 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a31599fe4f64b2b8548689feb79cacc : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d9e66f0859542a8a5cc7b8690f8c6bc : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e5eff5223224246914eedb09c2bd7e8 : SUCCESS in 6m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/66520bba8e48469e9fe4e445b9b25695 : SUCCESS in 1m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bdea979b47174f769f46e8bead413b92 : SUCCESS in 11m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/45cf8b2b393b4a288c103c081a6f5eb2 : SUCCESS in 13m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/509273ebc4c446c189e3fc9fca20ece8 : SUCCESS in 1h 00m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0bcfa3e1a95447618bfb74fb3bb08543 : SUCCESS in 1h 50m 19s","accounts_in_message":[],"_revision_number":3},{"id":"357817089ba06a39c3ce8b4b3723f319b2d5a4d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:16:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7153e9825ff23de9dfb923790575f0e09dcb3ea4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 17:12:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a4ba29134a49beb1ab753c38e7c11f\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/d4c6edd1d5ee4c3eb4daac89473414aa : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/330731b9e3a44d84a583fee356b8b746 : SUCCESS in 1h 29m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32d85b8f055946beb38c4e8a6c47296b : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c811e85939b24ce79ba6e6de5570689d : SUCCESS in 1h 54m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d7467df4fa54955b692e77a216efcfb : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/462a427329e742d6a5fb6a7877717c64 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9eef3839404474cabfa0b7e7591f025 : SUCCESS in 9m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/06f05daf6d274140a2c429aab756f96b : SUCCESS in 55m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e18c4efad889741cd1244434c52ad02832125fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:00:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98d2a31a41424a11b873518f62261fec\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/74cc56e53cc44b5d9e5d8436540aedda : SUCCESS in 20m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9a6b7b5834924cf4b0f5b12c5016526e : SUCCESS in 32m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/88437250870942b8b19c664acd90fad2 : SUCCESS in 1h 06m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093baaaef4da4ceeaec229d8956c5ee0 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c71e93a03d024dddb4e7e29e200dc11d : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff26b414d8ee46eda06f29f22a015c36 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c007826ddf6a43c99ebc3a8839a4f5bc : SUCCESS in 4m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5859d3474ad4493ca004f29e55f36f26 : SUCCESS in 2m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b586d85ef95b4e7e96b64980e0dd34d8 : SUCCESS in 11m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e0b8de3b19904314acafb7251871e3d0 : SUCCESS in 14m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/882e07042d4047aabdf6b554418dfad1 : SUCCESS in 36m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/eaa5c81eb5844ff2ac4427d170cbf542 : FAILURE in 1h 44m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3b2f7a9f840982a3c6d8629bf6b9eb025b4e8271","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-19 20:13:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e561ff4596aacff98468482fc9b90b758fab4195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 22:08:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/673a52dda75349df90852000372a4359\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/cfc6874799074ed49aaaacd5cfda3f55 : SUCCESS in 13m 52s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/8dcad1dd080c43d2a032dc2cd37b70a1 : SUCCESS in 44m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/14ab0108e72d4115abfa9fe6089c6d8b : SUCCESS in 12m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aa773157300647048dd68a40666d3c4a : SUCCESS in 32m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7e6de5f33a9d4a04a1f6ca72ea996ea4 : SUCCESS in 1h 03m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3d0141ca5a4d4c34b990d776e2e3261f : FAILURE in 51m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eaa342d8fa845fcaba85cc941b40b60 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b68266dbaaa4c38af5f6093a42c2522 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88b91ca9e41f4e4ab06832c39ee5b850 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffd47f6caac544dcb7b927c5d6a78aba : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5686b87e3da4710915c2366cd3b78b2 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0387bb02caca4297b2ccbd61edabbe98 : SUCCESS in 3m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8d037ecca5884d03a230c204f9b794ab : SUCCESS in 4m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f9cbb3bebf0649b59eec4e86f717cb01 : SUCCESS in 14m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ce829fb090ab4ce2b7d57410d35b7921 : SUCCESS in 16m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b720d8d510c248f1ac70b5cb1e335ca8 : SUCCESS in 1h 13m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/61dab7078dfe4b7881439bf89d516f5a : SUCCESS in 1h 43m 07s","accounts_in_message":[],"_revision_number":3},{"id":"0e650dad49406d520f8d2f0c295e90f048386680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 22:09:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f28c4b6d1666700f452d73089b38e6d6a75e01a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 23:23:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b14b970e7db94f6b9fbb2a4c0798af24\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d98912195c20476dafca67965d7ee8f1 : SUCCESS in 11m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/748744da92d742799e3d425f5a6a8549 : SUCCESS in 32m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6138ec81618f478da05a3c44aae79b91 : SUCCESS in 1h 07m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0183d74e607f446c9428852cc22b4606 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df884d5a084e41639777f08d1baca523 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97dae566fd66448593f05d7e00b34d9e : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4453a6400da47baa07488a7fa593857 : SUCCESS in 3m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/32c3ed166c5b473abf4e2fd5b75124b0 : SUCCESS in 2m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b37508e2efc645a7a970a1a9bd8f6c6c : SUCCESS in 22m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c1553fbccb9649f29287f62597332489 : SUCCESS in 18m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a628c8fa881241d19306e6e052994350 : SUCCESS in 1h 12m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/35063322fd8e492e9022d322f40873b2 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":3},{"id":"1b873b90359da2419253087e3ac7620ba6f051aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 23:23:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"78e07e45e2a0a6ab798350c64a70163539024d62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 23:23:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a23086268443a48de9bf848f0685d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c046181f74460b83a55bab651a4173 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"ae260c00a1ce43e30d5630102edf4b91b69fc353","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 00:14:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fd2311114e2455f8172c1b933d994dc\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/85a543077a90464d9b5c990366b33f35 : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a6630e9cb5c74c8eb30ee74460e3594b : SUCCESS in 2h 02m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8487a7d0049d4534b038332c0c919f88 : SUCCESS in 1h 10m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eeea836228944155aab0dc114d739002 : SUCCESS in 1h 52m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38f854d8f52d41e89a3f8b2f9a85e2d2 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ddc7a315106462e8af9a42b46a818a5 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a18ac19d09849348557a3bfb451c24c : SUCCESS in 9m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7c6e1790e2954363ba9e84d4336b7e91 : SUCCESS in 1h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"85878fe6a89f1a881ac276ec10dbfa0c093fadb2","revisions":{"143debc9c79fe953026b84d0245a7db5a8b213ad":{"kind":"REWORK","_number":1,"created":"2025-08-18 11:27:09.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/957677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/957677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/957677/1"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:26:48.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:26:48.000000000","tz":120},"subject":"[DNM] test barbican build on ubuntu arm","message":"[DNM] test barbican build on ubuntu arm\n\nChange-Id: I400e9a9a3bf03053589835f933023eedee504f08\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/143debc9c79fe953026b84d0245a7db5a8b213ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/143debc9c79fe953026b84d0245a7db5a8b213ad"}]},"branch":"refs/heads/master"},"7a8e68acff3044e4bca56997db0b0d1cc1b5a95c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-18 18:05:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/957677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/957677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/957677/2"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:26:48.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 18:05:15.000000000","tz":120},"subject":"Enable Barbican build on Ubuntu ARM64","message":"Enable Barbican build on Ubuntu ARM64\n\nRe-enable the Ubuntu arm64 build. The previous exclusion was due to\nuWSGI issue [1], which is no longer reproducible on current Ubuntu\n\nChange-Id: I400e9a9a3bf03053589835f933023eedee504f08\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a8e68acff3044e4bca56997db0b0d1cc1b5a95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a8e68acff3044e4bca56997db0b0d1cc1b5a95c"}]},"branch":"refs/heads/master"},"85878fe6a89f1a881ac276ec10dbfa0c093fadb2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-18 18:06:20.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/957677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/957677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/957677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/957677/3"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:26:48.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 18:05:57.000000000","tz":120},"subject":"Enable Barbican build on Ubuntu ARM64","message":"Enable Barbican build on Ubuntu ARM64\n\nRe-enable the Ubuntu arm64 build. The previous exclusion was due to\nuWSGI issue [1], which is no longer reproducible on Ubuntu Noble.\n\nChange-Id: I400e9a9a3bf03053589835f933023eedee504f08\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85878fe6a89f1a881ac276ec10dbfa0c093fadb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85878fe6a89f1a881ac276ec10dbfa0c093fadb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
