)]}'
{"id":"openstack%2Fkolla~957680","triplet_id":"openstack%2Fkolla~master~Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-20 13:49:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7","subject":"Influxdb: use stable RHEL repo","status":"MERGED","created":"2025-08-18 11:57:13.000000000","updated":"2025-08-20 13:50:08.000000000","submitted":"2025-08-20 13:49:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957680","meta_rev_id":"4088837a652600b208731089dc5ba581265d8a0a","_number":957680,"virtual_id_number":957680,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 13:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-08-20 12:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-18 18:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-20 12:43:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-08-20 12:43:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 13:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 18:21:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-08-19 13:24:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-08-20 12:43:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"71b192742d1a57bd7f5f2325c4a8c72c65039921","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:57:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d4c84c5e8a082a459b6727a902c4d6426ffab06","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 11:57:24.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"918ee70bd024cfa7f6bbdfce43a60590aa37ea86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b423a2a84f4142ea8af849ec32f3207a\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/9e842993f0104f0c8443a45c68b0f761 : SUCCESS in 13m 55s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ab577efa42a746948b5c5af582e65d85 : SUCCESS in 43m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/32c2d8803798483fb06c9609d6030472 : SUCCESS in 25m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/697e35126d4b4f56ab4b486028715cb1 : SUCCESS in 26m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/377694a3266b4979adff84ae73ae5324 : SUCCESS in 55m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1519098331d04eeabcde7af0f50d7995 : SUCCESS in 58m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545cc3b1609e4d9badef96172d01e59c : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0cdf38c91f14674abccf12f56fb5671 : SUCCESS in 1m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e18b5557369b47989cdfafd00fe01b14 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eed33db7ff7c45beb9148235c77eca03 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8b569731a3347d48ad3e14e530d3b2c : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7fcaa1660045df90015d37e2380f58 : SUCCESS in 3m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d5c4cea71aa84624856bf73fc4714195 : SUCCESS in 2m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0de81b509b244a9d8a4186baa0f12eac : SUCCESS in 10m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d779386a25c14a20a9a95d4383e677f8 : SUCCESS in 13m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6621d9ab075642ca9721c400082fb2b7 : SUCCESS in 59m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/533115288b9b41ac95f83e08646bc38c : SUCCESS in 1h 29m 47s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/704729994edb461892555e59206ed945 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/71af538283c2432c97edde4e5bac7fd3 : SUCCESS in 1h 50m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/70672b98994c45c6a1cff9e9a0813098 : SUCCESS in 46m 51s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b3f9b1953b384c2eb8e2e608062f3779 : SUCCESS in 15m 15s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/78bb1f4d5fdd4aa2924d223723ee7e60 : SUCCESS in 28m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4e876bc4931248ba9c9c960db3c61452 : FAILURE in 1h 06m 27s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0eff9ddd1fb548c8a3f8efc9725f9671 : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/da03c3cb7ded4bf7acf5487aba3889e8 : SUCCESS in 26m 55s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/c6192ecb8b334ad682c3d1fbc164907d : SUCCESS in 1h 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a803226f9a07a5038444775776103e02b5838ac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 18:08:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678fcb7c491a47cc8524ef6579a3f084\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3252e97e10eb48e7804a4dec7d9c6f73 : SUCCESS in 47m 43s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3235c9933a7440e9b0b7803482d9b338 : FAILURE in 2h 24m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3e1d00d1cc4847f784dae71dcf97f6f4 : SUCCESS in 50m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a01547e0953e44ebbbbd5702bce64083 : SUCCESS in 1h 41m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/215929162ccb4055a307ad6d8b87445c : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e1a43d039434ec0aa19d2832b1ba54e : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebf29d78ffe24b8095305f1d311d9e2d : SUCCESS in 8m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aa4bf09bf52d412ba70421e80a8c1c54 : SUCCESS in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dbe65c8eef14fbc7934087783c03a26ccb38685","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 18:14:37.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1f48d60ee71c847c02d7c83f13961b482d96d0be","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 18:14:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"bdc8aadc18f797ea11067634a0a51fc8030eb06a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 18:15:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b470f4292ad1e46988011cde422db210a1960b12","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-18 18:21:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0188462f6cac97e63901ea27749d33a3a8963209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:48:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00874322b5f942869075a968db33dbb1\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/db0b75743b0d4ea8ad4a448af5be312d : SUCCESS in 14m 35s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/a9007ecaf57b459c81c3f2f5afb53ed7 : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a6ed0459dd77426a8a1fb527b556f2c4 : SUCCESS in 22m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8e2c4c3a9e0142c2a9b20c49c95d2e14 : SUCCESS in 31m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0af8fd14e9eb486f832f43ed3388b952 : SUCCESS in 1h 05m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c5499817378d4fe6847479a1eaf876c2 : FAILURE in 1h 36m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8870700aa7a41df8839b04bb2eefd91 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f86d1f570a94e9f9a6b2734822a147f : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d31bafb956af4b12a91c580234578912 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2b77f28f5184e03b821111e6d5e6548 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/064a7039b0934dbba4a133a4f9eea2a9 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d43d1c7cd13475da9d314f492aeb1f8 : SUCCESS in 7m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/36b7ca86b9bd4e87bf55572ec85b9681 : SUCCESS in 4m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/04f612f1c5d64f4d910640a82028a7ac : SUCCESS in 19m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9d64e615e8ca495192d178636bdff2af : SUCCESS in 20m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a1e8f3bb6ad74631bb624e73cbbdfb3b : SUCCESS in 1h 11m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/28182161be3a4dd587876a53a3f957a3 : FAILURE in 1h 46m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c4bf50db6f884676bf6756a8a4113ce4 : FAILURE in 1h 27m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3e80efe113644fde9046eb31e377141d : FAILURE in 2h 24m 57s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/43e5c4fe6b6c4bc591b3e9a309055eb6 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f12f3598705d4ceda609f81c46cfd5dd : SUCCESS in 23m 07s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/cece46c560914439a45220e785341749 : FAILURE in 52m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8d87c90d79624e998e264338ef5ab374 : FAILURE in 49m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5ab0eacc43834d00a2e10bf9bdc58600 : FAILURE in 1h 06m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7d61692899d74f2fae1b42c52e013c0a : SUCCESS in 24m 07s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/6cd8729079244584a2e084073a853eb8 : SUCCESS in 1h 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91fcc44bdab930566d97d77eefea2f1ccc0e6308","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 21:15:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16941c212a22785eab9e7fee24364154a93ea071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 23:45:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e95c65df2f3a402aa8c0beba6acdeba8\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/0b886f53ce2f426f8fbaca68be04918f : SUCCESS in 17m 53s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/422bebc34cc24ff0be299eb2706a1712 : SUCCESS in 33m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9226d3ede90d4cdf80d4331024fab466 : SUCCESS in 10m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f1be747ff25640f8b281272f7d356927 : SUCCESS in 15m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4e842d2d0d4845f4ac02cd9911391f1b : SUCCESS in 1h 06m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5b5332d5302b4a76b7b33a66a5d22bd3 : FAILURE in 50m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b55117af5c54eee9b4574e2aae492a1 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3b21e45b4f14003a768ec1e8e179c4d : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e528f29032074aff8c376275987efc54 : SUCCESS in 1m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a179d8efad0244efa45e9e2107d7e867 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3488bdfa917d47d393c529b6967106ea : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf86d38f6949400fa3a31aa3e43852e8 : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5ec1b5f7e166449ca151959d4703cea2 : SUCCESS in 1m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/48ec4cddc3994b57880058a0092fc591 : SUCCESS in 13m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a3da5871020a4f9a82fbdfd04f7ce11a : SUCCESS in 16m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae0f413f83864477a93f6e9fda270fdc : SUCCESS in 57m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c7c1349a5ff742129bdc986cafcf845f : SUCCESS in 48m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6205f4b5b58447589df8ff42a235a55e : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3d19647c29f54670a588efa7b9dcdf48 : FAILURE in 2h 27m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9c469dc2533749ea97dd8ca8eb6b6e50 : SUCCESS in 44m 42s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/eba2e3698b9540029a9c1c68e4540982 : SUCCESS in 17m 53s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8c9c8893832b412f9cf800acfcc4d64d : FAILURE in 38m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ac9dbcf46c16450d93e893955f70f298 : FAILURE in 26m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6a9f300ceb0b42dd84068eb23ce5c73d : FAILURE in 1h 01m 10s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/441da7ec49be4943a09dbcc953236771 : SUCCESS in 28m 06s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/bbc3f8247c97442a8fbb01301b0f5f35 : SUCCESS in 1h 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cfccb08771abed9aab4834f345ab51f84657f34f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 00:36:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7315427cc87246ad89596d4ef6c185b5\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/94c9e61c0574448281525d8758dfbb03 : RETRY_LIMIT in 2m 21s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/ade7f6e7607c405d9f093928524fdf38 : SUCCESS in 1h 55m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/da4e1dcda8124727b870dc2ef5b43634 : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5b738eb3bae74ce7971a904809b6bd79 : SUCCESS in 2h 04m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/53b5a4a8fd8f49838692a80fb147b3a9 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cad5014c15144f4e9a57d08ffb3963d7 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b695d6d2314e45588069a85061cc7e20 : SUCCESS in 8m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8de3c256f41c442e95271c181aff0893 : SUCCESS in 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3afe5afec606e6bd189228a76d2bcefb483dd182","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-19 13:24:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"447c1ce62ff4dfd7a319d4af009ceee69b3d6cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:20:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b41909fb1654beebe2f65f71583273e\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5fd288691ee2469fba831811c04df33a : SUCCESS in 12m 24s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/8187e8129f5f43f68f32ae98e5f02ae4 : SUCCESS in 46m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9806ef30e537402ca74cf3a885047319 : SUCCESS in 18m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/14b00e4fef6c4ccbbad40914381ebeb1 : SUCCESS in 15m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/78aa310e24714e05aa0c42ff672e04b9 : SUCCESS in 29m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dfd95b4d394344e49cd2dd561689016f : FAILURE in 1h 06m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a915862bac1b460f82277888c698f272 : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4c2ae4c7dca44dab2359342916ea321 : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a66d19ba0aaf4259bad89426e3861a47 : SUCCESS in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2d3c667966b46a88fd4c2f5724a85fb : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc173a7035c840a78985809994a22583 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4cae28fa25f4009b1a039ce54bc578a : SUCCESS in 3m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/38f3c2c36ec042539e3d31e6b5153bfe : SUCCESS in 2m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f7678c25fd914163a8e849c26f5afe64 : SUCCESS in 10m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8b27b676ecb0496e9745fe5c4a784061 : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b37b6b4b23be4559951704f345dcf865 : SUCCESS in 31m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d2c462bbf4be4705b49014d9bbf63e60 : SUCCESS in 1h 30m 44s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9280c95616664232aeaff68195fc37b9 : FAILURE in 47m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e3c62570c8f64cb590f645a4639d6b19 : FAILURE in 1h 37m 14s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/46f6b82c723f4f248ec708dd70735fbd : SUCCESS in 50m 36s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/171b37e8935e4372bb0fc8ee02722e2e : SUCCESS in 14m 53s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/da1e8e0d77e6450680a9dbea1dbab0f2 : FAILURE in 27m 31s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/eebce8a9e7e44a94807ba3abdac271ef : FAILURE in 21m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/106b527921a24447851ceded5dee6cad : FAILURE in 49m 32s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/81913ebfc4894ca2aeb5348c120e8444 : SUCCESS in 13m 22s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7c7b7f6a09c04aae83194d766411118a : SUCCESS in 1h 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a574cb049e0adc93d629892e080446ab4c324669","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 16:52:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50eed3efbf0740ccab619940c7454625\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6cac2caff30f4c6286d45e7efb54142e : SUCCESS in 1h 16m 37s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/0bc135a635474375b1776bd9820d9d8a : SUCCESS in 2h 17m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ecaa565939a4020afe4e436e2fcd300 : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c6017a0423124c00a3a640b7e93deeea : SUCCESS in 1h 57m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e16ef17ae6d4954b363a90aa0125b1d : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70f71031723a423badd591ebe1ebf161 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d67c7ffe81841bc8d3818c6c7459504 : SUCCESS in 6m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d724e5216788466cae6a9a79cf45c5ce : SUCCESS in 57m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9309b07ff9148f509276e57ee86039a92fd160b3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-20 12:43:12.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b7fb90f12cd27f37248dbd402c6c228eae21d56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 12:43:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"222ba37adadc663fdabc841954844fca5e18c259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:49:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1832a27ec1f14cc8abac3a7f16a66457\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad16adc871e044b5a094e3e2bc076987 : SUCCESS in 29m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/975a64f28a284fb1806e994cda03bafb : SUCCESS in 14m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/02e3c23c037142d194477caabd1fb2e5 : SUCCESS in 55m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ddaec9d17a460689b33fb170d9aa93 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fc2df09d317485a8a88746503b21552 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/482a189a39f6438ab2fa18de967d1a11 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4a2cc5d27a4a03b2033a32ecb5e527 : SUCCESS in 7m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/92ca0f2301804c33ba9bf9914744246b : SUCCESS in 2m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/357b221367fc4984a6f514b9d15570f3 : SUCCESS in 16m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/808d7ffb300a42d790ec812fb3f85fa9 : SUCCESS in 20m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2af6bdc81de24a55961edbad24cd8376 : SUCCESS in 42m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/57b5fe6be334481b8bd1c999f0c478ad : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ebb550871458c8eca38c1fdf42cb3779d4fded6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:49:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4088837a652600b208731089dc5ba581265d8a0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:50:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/506126e896da4523a550b3be29ddcecb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201ae2950ae14a809db37212eaec6410 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"444a81b62bb5ccb571f82e34f4ce0c48d8f639bf","revisions":{"36bc9cab955a6a5b10fbbe2a3f84263e5f326f1a":{"kind":"REWORK","_number":1,"created":"2025-08-18 11:57:13.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/957680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/957680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/957680/1"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:56:54.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:57:11.000000000","tz":120},"subject":"[DNM] test telegraf on arm","message":"[DNM] test telegraf on arm\n\nChange-Id: Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36bc9cab955a6a5b10fbbe2a3f84263e5f326f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36bc9cab955a6a5b10fbbe2a3f84263e5f326f1a"}]},"branch":"refs/heads/master"},"3c641c5b907eda259a35617c6e9b4c01bc8c8df2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-18 18:14:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/957680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/957680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/957680/2"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:56:54.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 18:08:03.000000000","tz":120},"subject":"Influxdb: use stable RHEL repo","message":"Influxdb: use stable RHEL repo\n\nSwitch the RHEL repo to the stable channel as per [1]. Use new GPG.\nARM64 binaries are available, so Telegraf arm64 builds are enabled again.\n\n[1] https://repos.influxdata.com/stable/x86_64/\n\nChange-Id: Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c641c5b907eda259a35617c6e9b4c01bc8c8df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c641c5b907eda259a35617c6e9b4c01bc8c8df2"}]},"branch":"refs/heads/master"},"444a81b62bb5ccb571f82e34f4ce0c48d8f639bf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-18 18:15:39.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/80/957680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/957680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/957680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/957680/3"}}},"commit":{"parents":[{"commit":"ed094e0af4caa2bcd5d574e6c9f4079be074347b","subject":"Bump libvirt exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed094e0af4caa2bcd5d574e6c9f4079be074347b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 11:56:54.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-08-18 18:15:13.000000000","tz":120},"subject":"Influxdb: use stable RHEL repo","message":"Influxdb: use stable RHEL repo\n\nSwitch the RHEL repo to the stable channel as per [1]. Use new GPG.\nTelegraf arm64 builds are enabled again, as binaries exists.\n\n[1] https://repos.influxdata.com/stable/x86_64/\n\nChange-Id: Ie4f6f15404f6965df51f3214506ae9f8bb9b61e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/444a81b62bb5ccb571f82e34f4ce0c48d8f639bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/444a81b62bb5ccb571f82e34f4ce0c48d8f639bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
